)]}'
{"id":"openstack%2Fkolla~687253","triplet_id":"openstack%2Fkolla~master~I4f95c6515f73c9e952d0e66c1ed44b1271c798c0","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I4f95c6515f73c9e952d0e66c1ed44b1271c798c0","subject":"Remove designate-pool-manager container","status":"MERGED","created":"2019-10-08 09:09:46.000000000","updated":"2019-10-08 14:18:10.000000000","submitted":"2019-10-08 14:16:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687253-1570544211632-25c6a380","meta_rev_id":"dcbcc21bd611fa346022e58557b92d906554c3cd","_number":687253,"virtual_id_number":687253,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 14:16:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-08 09:53:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-08 12:36:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-08 09:16:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-08 12:36:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 09:16:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-08 09:53:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-08 12:36:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-08 14:16:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"146cf4a6133550aa1573e6187b2024e7b7e9d5c8","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-10-08 09:09:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3b9acc341877613ff81a3e5d0edc4acd01f6faf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-08 09:16:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"649d84274ca7e482229dc3e33a35127ffbc03a43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-08 09:53:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/53/687253/1/openstack-check/tripleo-build-containers-rhel-8-master/b3e9c77/ : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8aa71f2328687236bb0e980a506934faf7af426c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 10:48:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a9f90a2adb6b4ebdb062204f1a076daa : SUCCESS in 47m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f247063040af4a1eabffee6171bc7f82 : SUCCESS in 1h 00m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/98deaa4fd6a347d585221370b3897d87 : SUCCESS in 1h 27m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3daa21777f9a47f8a20a0e9a4c3a41cb : SUCCESS in 1h 31m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1b76474ffda740859a53628b1c7e7421 : SUCCESS in 1h 30m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fcd16ac13a924973afdf99109023461e : SUCCESS in 54m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f2e6cc4e6dbc48e58b3a60250be7bd43 : SUCCESS in 27m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1e275e65e344670a91b32c2e3b2a461 : FAILURE in 18m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/403c1c3e27b5466ea71f392c4df159ec : SUCCESS in 1h 10m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5c36e14986b40b7afc70eba3b21910b : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3948257da664dcb83b92b5954904703 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64c5aebc280148cd9640239a7cd0d87d : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6772e4ceb2014a1c90d9aca8a052f309 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a1ba9fab54a4f12b02ee99ea1d4c4a3 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c781da677a4a3b8c04030f9e9ebfc1 : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6edeed88896c489493590b7a81762df7 : SUCCESS in 27m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8720b7e298ed4f4ab6ed6d1889108490 : SUCCESS in 32m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2df128375f0246d89334d790af025230 : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ded8d6e4917a406390f8037f1167908d : SUCCESS in 55m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/292197ed03f5444a8fe9f1da0a32f0d2 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e1e319f628acee50661321ec5824974761dcb25f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-08 12:36:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d45ef0c92110ef7a6e4a3e862aa70013ecaa5dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:36:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40d24c3e832053d9d1f326912f68fefb2fcfcf66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:16:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"419fd7e5736885b5218f3a898b4860e0b7f371c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:16:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/68c6e9afa107473183182a87ba19f568 : SUCCESS in 54m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/87f928dbe47e465e99f95f364b4a3eb0 : SUCCESS in 1h 26m 14s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7eced1df857b4fa7b163e7fe223acf27 : SUCCESS in 1h 12m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/41b2ae92948146408343512036b254d5 : SUCCESS in 54m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/682461e9e7224e6db810303231867822 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d604229419c94221b2d01dd6a08f9834 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9068d8efcae84d7f9f5b5c4383fb5664 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f21e8b65ca420fa51a590b75c91965 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acbb51c14d89410aad131fbb0453a6d8 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfccad085ec5484d8fc9c7b784b06f96 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d1c4d8a1cbd4bf68be46a3f8cf66b47 : SUCCESS in 31m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bfbd19b42ed425393655871a382e850 : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afe3124166fe467395285c38007d62bc : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dcbcc21bd611fa346022e58557b92d906554c3cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:18:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51a1f27ceab43ce8ec2075c9120fc27 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14be9ebeee67ab79fbd9ff2a22e77eaf6473e7e3","revisions":{"14be9ebeee67ab79fbd9ff2a22e77eaf6473e7e3":{"kind":"REWORK","_number":1,"created":"2019-10-08 09:09:46.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/53/687253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/687253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/687253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/687253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/687253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/687253/1"}}},"commit":{"parents":[{"commit":"138afba32efb631cec5f2f4b4ee39b695427b5e9","subject":"Merge \"Add Debian/source kolla-ansible job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/138afba32efb631cec5f2f4b4ee39b695427b5e9"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-10-08 09:06:14.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-10-08 09:09:43.000000000","tz":120},"subject":"Remove designate-pool-manager container","message":"Remove designate-pool-manager container\n\ndesignate-pool-manager service has been removed in Train cycle [1] so let\u0027s\nremove it.\n\nThis container was already retired in kolla-ansible [2].\n\n[1] https://review.opendev.org/#/c/665476/\n[2] https://review.opendev.org/#/c/682253/\n\nChange-Id: I4f95c6515f73c9e952d0e66c1ed44b1271c798c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14be9ebeee67ab79fbd9ff2a22e77eaf6473e7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14be9ebeee67ab79fbd9ff2a22e77eaf6473e7e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
