)]}'
{"id":"openstack%2Fkolla~687615","triplet_id":"openstack%2Fkolla~master~Ib95106c96127a81dd757dbd839080a47bff725db","project":"openstack/kolla","branch":"master","topic":"train_rdo_rc","hashtags":[],"change_id":"Ib95106c96127a81dd757dbd839080a47bff725db","subject":"DNM: Train RDO RC testing","status":"ABANDONED","created":"2019-10-09 16:24:14.000000000","updated":"2019-10-21 08:27:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"daab5c7d4c402993b020f9954e41e02ca49071c7","_number":687615,"virtual_id_number":687615,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-17 10:03:08.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-17 12:34:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-17 23:41:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 07:18:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-10-17 12:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-17 23:41:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00abfb61f42a5edd0738b7cda9053815be457a1e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-09 16:24:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d7f008b1a6db0ff895bcc1355392ade7059eae9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-09 17:06:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/15/687615/1/openstack-check/tripleo-build-containers-rhel-8-master/31af819/ : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"86d4253f43ec62145433f366928cbea25721d43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:10:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4e9270db06d04255aacf76862298c7a4 : FAILURE in 5m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/34859085f03045acb61152383be31dbc : FAILURE in 5m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca298614805044289c9a01e14b52c686 : FAILURE in 8m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/124dbf6e6c67436f8db2c22b0b038959 : FAILURE in 10m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/693e6546d901401b9da28db8ea3b03ac : FAILURE in 10m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/728f2b835656432e99bb508537c06648 : FAILURE in 19m 07s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/4eda794acbe24da39e2d395963a0bf67 : FAILURE in 11m 20s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4b71bc3315d84b74b1154ad2450a7fe2 : FAILURE in 10m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f2f400a459c6438e84fff9cb592eea29 : FAILURE in 13m 14s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/78607a1952934269bfd9a0e6f20d3df9 : FAILURE in 8m 25s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/156346948adc4acf8e705523dab690c4 : FAILURE in 11m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/e11745ea46824d9b8e07683fd5ac9907 : FAILURE in 9m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d0cab87d2f2f4a57947bc542c28e8a10 : SUCCESS in 28m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/97e39b96a49648a68d0becf20112b23d : SUCCESS in 23m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0dc599c4710c4e42ac1f1f5de81f262b : SUCCESS in 55m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/307313187a4d4dc0965fdd7049f08b41 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3304a25ca39241efbf064e47632e78e1 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8f1e19f73d44e73800ad6cd91173604 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9068ac9c0f548e7a1d63ba9838b3d0f : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82b389ddb96f4ed28b1aa9571a6a0fbf : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9525422b33942ee805e3e9a57356da8 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b28e86a1c8bb43c58cb6f8bf935f9eea : SUCCESS in 24m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aeade25d2f8340eea3372ec06256faf3 : SUCCESS in 28m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e2fde9013704e7f8c3869bb0fe0c87d : SUCCESS in 59m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/344be3ab6f40413787771d895ecfc29f : SUCCESS in 53m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf43a8bf375749b2a51f2c6f6e2a6a7f : SUCCESS in 1h 04m 02s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/955e8db0e6664e61a1f5dd60f4245e0e : SUCCESS in 1h 16m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/8725a921d91a4375b1d7dd2b832956f8 : SUCCESS in 1h 33m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/02b616559cbf4ae2b2b8a4025a8b3357 : FAILURE in 5m 27s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/fa08c0b1d7434ef5b09c2929b99e8030 : SUCCESS in 1h 12m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/39a42d05994d44a5ac0f1384861c2a47 : SUCCESS in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58f4b75ff86c37e8907263b1af4c09f20ace7cee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-10 06:19:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a52cf6944588a74f1e8f3a09fddd9b6e66872e0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-10 06:34:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"55a81f9535dc6aaf92f7399aef23714ab8feb227","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-10 07:13:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/15/687615/3/openstack-check/tripleo-build-containers-rhel-8-master/31f46ca/ : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":3},{"id":"96230ffb2ec60e5531efc7cbb591907dd5d7e56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:13:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/13cf4363d1284000909201d50c652d5f : FAILURE in 41m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/58301081db68477cbd4ff146846d5e1d : FAILURE in 57m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2cd1d4ea87624506a243a16651785152 : SUCCESS in 1h 30m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/eb74842d47af467f8d261b6128749fb3 : FAILURE in 55m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b1afb252546c4fb68f762f7fdc2d0d82 : SUCCESS in 1h 56m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a1489c9d70bb490e80344806c7aefcb2 : SUCCESS in 1h 14m 56s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2c58e5eb54d74ab39d0c43c59f0257ee : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/cceeabd81d124faca6f3e53e47a854f6 : TIMED_OUT in 2h 32m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6db196967ae14b68b2b6927e577f8b75 : SUCCESS in 2h 03m 43s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/67f022cc02e443f9bea88394be0d7c64 : SUCCESS in 1h 36m 15s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/37a7c660fa1d46d1ac77e12fb9980579 : FAILURE in 1h 58m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5d0dcf25d78046978334dfd8957f209a : SUCCESS in 1h 35m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8b0a179bc9ff45c08a4cdb82c1e4fba3 : SUCCESS in 34m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0fe39f5a0e3344c0b455d70466b3733a : SUCCESS in 34m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c50369048bb438f83deb8529b9e1509 : SUCCESS in 1h 09m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/172122ce150646239ae0f15cc62c5ec7 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2ac27a5e3c4f00b346083a7bb48f46 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f258c64658a04785a97476a4c5a8e0c3 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3413a2fbdb0642af850d401f91e0b3c6 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fc53e9be27f4c7aa56ca716eb71fb9b : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca86f28590534f4791518cb4e61a056f : SUCCESS in 12m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49e9f6d204ea4449a3c78715e493604a : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b08e1254daf42029aafb67cf87f7a05 : RETRY_LIMIT in 2m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87625a7c95764704a538d4825c6dad25 : SUCCESS in 1h 30m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77225b30406a46d6bbc9f75afaa09b74 : SUCCESS in 1h 24m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dc0ae791f9d84d2f84ecfc4bfd0fa375 : SUCCESS in 1h 11m 00s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/457e3ca7f9f94acfa6f2197d8786a587 : SUCCESS in 1h 32m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/14bf77f3088d4018a24789bbe6a60b8f : SUCCESS in 1h 47m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/25b5dac24bad4a0996e38a2e66d512a0 : FAILURE in 1h 42m 00s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d6fd70cd3372411ebe2fe1522076f2b0 : SUCCESS in 1h 25m 12s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6266a560ae39428f8d7f8033df3cf70f : SUCCESS in 1h 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ecbfc7357ac3628c76c8d2bd96df9a0973910e6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-11 07:18:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"86ecf4f474a9957bd987a16568fae41a848bc1d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 07:55:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/15/687615/3/openstack-check/tripleo-build-containers-rhel-8-master/860553c/ : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1c54dc9bc192c1a28c0b5bce1b7b7e316bd17be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 10:06:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f22f89d4dcc4156a5294d0764bc0a91 : FAILURE in 37m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b2fc1725c4e14cbf8bf2c4dd041e4538 : FAILURE in 48m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4caa3dbcf6da4108be700433ad5ec2dc : SUCCESS in 1h 15m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/16dab45fa7ca4303ad866ef990c192b0 : SUCCESS in 1h 12m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1e4c7dd1e1964d009512fd4c12d598df : SUCCESS in 1h 37m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cfb97c51254c4f37bc1ca30db6f31b88 : SUCCESS in 56m 39s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d7083a8d2a584ec087f567d6355e364e : SUCCESS in 38m 55s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4239d361bce54c0988ec59b70d8e9d8c : SUCCESS in 2h 07m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/95db89724b1d485dbc26f83ae85934e9 : SUCCESS in 2h 32m 41s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/fc116748305d4d39935a7f75258a24f5 : SUCCESS in 1h 53m 52s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ce414314dc744c479c51c244873d38c0 : FAILURE in 1h 40m 05s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3f58326ed535476caaaf75dc18100e2a : SUCCESS in 1h 41m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee127fff42454904a14eeffa00cd5301 : SUCCESS in 41m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1974e0a87000475d939c73f2aa6c5a68 : SUCCESS in 19m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b4dfb4e8b8934f7b8015d0b5959502d2 : SUCCESS in 1h 12m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cee65140bd6648759ff60b2fadb4df1f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd75a2c35c1f45e8a8c224fb41b7d8f9 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b15dbe60f15b4dd3a18d6a24c92cfd3c : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee7137ee24d4413fa6f60ab75d7a3b3d : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbee614738d24513b0cd403760de9c2a : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469f9a92dec1448ab7623e4ad0fcfef9 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f9e548b40194e95a586371a42822e6e : FAILURE in 25m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d59371b307e4425fbe0dc85ca558e6a4 : SUCCESS in 28m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb5e5044cdd044f9af734243ba943408 : SUCCESS in 51m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f57eda419404200a0cea38584256769 : SUCCESS in 58m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0d86075c385460b93e14db6a4c75559 : SUCCESS in 59m 46s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8076c2ac1abe43a3b7fae4e3da6f9102 : SUCCESS in 1h 17m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/4ddf637e37c3484e88b37340bca1e6c9 : FAILURE in 38m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/08f6467524574aafaf7b93f3cf48ab4f : SUCCESS in 1h 39m 07s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/5eb79b73abf9497396f3374f4bd6ee71 : SUCCESS in 1h 16m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/91f5a8b9199e4c4aa26e6aa8076445a3 : SUCCESS in 1h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0009db016c5da208ece16593ac53291877c70e96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-14 07:41:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e09f7d7f58d311a81f858bf7be8178f3a864fb70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-14 08:20:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/15/687615/4/openstack-check/tripleo-build-containers-rhel-8-master/fb36a51/ : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":4},{"id":"2fc3ecc0235e14865aa1c091b2b3896cf4d3786b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 10:33:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1a2c345b53384d43af80d61b346265b8 : FAILURE in 49m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/30962400a0384c1b8a9bf697f191de49 : SUCCESS in 43m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b3e97fb77b9f402c904b91abccd2f125 : SUCCESS in 1h 18m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/442cad2192d14865a7d41be4251a89cb : SUCCESS in 1h 12m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6c7dbaa8266248509f295cb122c9df68 : SUCCESS in 1h 29m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cabc2394c3484a608ccc7a259aff8186 : SUCCESS in 50m 44s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/ca542b6b314c44798cd7f4431ed1336d : SUCCESS in 48m 36s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/8424ff19d4bb49f088519c77b35dc188 : SUCCESS in 1h 57m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9fd18b04fc3c4f128893a2715f0b0bd7 : SUCCESS in 2h 15m 45s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f06f2b7d43d94ae292a42f8e52030d15 : TIMED_OUT in 2h 02m 29s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5ac2a5ac50604f6793ace774ec92cd59 : FAILURE in 1h 59m 11s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3589aca596674cbb97df113da657d32e : SUCCESS in 1h 43m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a5c3b649866144a9b67574b25aac2044 : SUCCESS in 24m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/75b78656716743c69b33499a00426a00 : SUCCESS in 19m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c5a41c165e184360bd02d4586271ec08 : SUCCESS in 55m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e71abc462dc54821987a4ec639da7eec : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d4e6fdffbc4c1cb23e9ad0db19780e : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/119a7182a18847b689795427dec34a71 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77923402d4f8441097ecca4307f98855 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d6262bfca694b8cbef9c7f8b0642462 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26030d52e303464f832c226da4675eeb : SUCCESS in 7m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6b597e6b6ac4caea92c1b52ee8c0d00 : FAILURE in 33m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/542d5ed0eb1c4b7c97ec1209a89c1a72 : SUCCESS in 36m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3619981e52cc45f68bbfd5885a9ae4ea : RETRY_LIMIT in 2m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/906f94c434e84cf2bb14150441a94810 : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac817168d83846a1a879eb3a69eeeffd : SUCCESS in 1h 08m 28s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1958454393c54e3b8593296d47706f25 : FAILURE in 1h 07m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/44a69eaccbca49e5b000d492878cc268 : FAILURE in 38m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8fcd444c868445c3947d3096845234b5 : FAILURE in 1h 17m 54s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/bfc382cbf88e4f20b85f348b83062430 : SUCCESS in 1h 26m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6b742759aba54e00a7a09a05dc803296 : SUCCESS in 1h 05m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ee6d713ae1cc2ab29391499c96dde02ece83ba2f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-17 10:03:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a8c8407621e0f246f47f2d0444a575657d4f124f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:34:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ab9d5ccb950d4a739765ce1e6fce9879 : FAILURE in 47m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f536164076ad486ba5900efacf0aba44 : SUCCESS in 1h 01m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b1918ee271784186b36ce73e1e248f00 : SUCCESS in 1h 23m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5a9d4684c16c4a8b93834588170f158c : SUCCESS in 1h 25m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/91b60f03434949eca6ceacb780438fa4 : SUCCESS in 1h 42m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/32c8f4e88fb443bbbf3936a863cc8b6a : SUCCESS in 1h 01m 33s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0fc55fa2a73843d1a6e4e152f386bdee : SUCCESS in 1h 03m 02s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/2811f4efd24a44f99e8263f6edefdb06 : SUCCESS in 2h 10m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/02d6dc78840642b3a15beaf96d1ad6bf : SUCCESS in 2h 26m 26s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0d95394a099046d5b616ae059b8ed9ae : SUCCESS in 1h 57m 55s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d5d7c0f2c34c474492a084c730cc6cf4 : TIMED_OUT in 2h 02m 39s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/d8cf0bf73c894d69b2a9182a1dcb0fef : SUCCESS in 1h 35m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f4a6e5068e3746e08100508dec0c2ec5 : SUCCESS in 26m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6eae46877854449db2da606b4bb9bf47 : SUCCESS in 18m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d4439d15f4346c1b827cd6aa091ad4d : FAILURE in 19m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b965f513d4f48be88442d298f18a52c : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bae289b106d4ef79b8fafde55c76232 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fa20573fbae451ca968f077651ce7a4 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fad9177b54d041568cbd23518da5e897 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c810d2382554e55b072756027694614 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae328c1ebf5436cb3d75b0e8a612478 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dcb376c7afc4454b17d40585dc9c44d : FAILURE in 45m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4612016accdb40e984e5773afdadc36a : SUCCESS in 34m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a357a6c027e431fbf8c15b868fdc23d : SUCCESS in 53m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3774165e0a18430a8e59a40639f99119 : SUCCESS in 59m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff40d33c097049509d24e855874296a7 : SUCCESS in 59m 08s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d0b9da460f874c88bdb7ce367241ba82 : SUCCESS in 1h 15m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/227b26de9d954ed7b0114365eecdf265 : FAILURE in 34m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a2bb1e36ef48444ebb1c625960649489 : SUCCESS in 2h 07m 37s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/591215b3602b47a3b04312a3e7d4d3d4 : SUCCESS in 1h 25m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8712bc5a38254e7ab831d2560884b1b2 : SUCCESS in 1h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"020964e796968c357dba39758040d8c76d068f7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-17 23:41:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/15/687615/4/openstack-check/tripleo-build-containers-rhel-8-master/eaae005/ : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":4},{"id":"1586cbd1b6263c28b15a13b9cc2e740241e836f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-21 08:27:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4f650204af83ac1482a008bd9c59c216e161d615","revisions":{"51e32daf207a686d206d28c00a5adc8153456b58":{"kind":"REWORK","_number":1,"created":"2019-10-09 16:24:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/687615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/687615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/687615/1"}}},"commit":{"parents":[{"commit":"a55af958a319bc31abe58311bab3d19b344a4f0b","subject":"Merge \"Remove designate-pool-manager container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a55af958a319bc31abe58311bab3d19b344a4f0b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 16:23:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 16:24:11.000000000","tz":120},"subject":"DNM: Train RDO RC testing","message":"DNM: Train RDO RC testing\n\nChange-Id: Ib95106c96127a81dd757dbd839080a47bff725db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51e32daf207a686d206d28c00a5adc8153456b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51e32daf207a686d206d28c00a5adc8153456b58"}]},"branch":"refs/heads/master"},"6e8ac4802637bdc414f9e45cd57d9b8101b1ea6f":{"kind":"REWORK","_number":2,"created":"2019-10-10 06:19:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/687615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/687615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/687615/2"}}},"commit":{"parents":[{"commit":"a55af958a319bc31abe58311bab3d19b344a4f0b","subject":"Merge \"Remove designate-pool-manager container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a55af958a319bc31abe58311bab3d19b344a4f0b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 16:23:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-10 06:19:40.000000000","tz":120},"subject":"DNM: Train RDO RC testing","message":"DNM: Train RDO RC testing\n\nChange-Id: Ib95106c96127a81dd757dbd839080a47bff725db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e8ac4802637bdc414f9e45cd57d9b8101b1ea6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e8ac4802637bdc414f9e45cd57d9b8101b1ea6f"}]},"branch":"refs/heads/master"},"1648db13877c432ce73f920796c9a50821576749":{"kind":"REWORK","_number":3,"created":"2019-10-10 06:34:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/687615/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/687615/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/687615/3"}}},"commit":{"parents":[{"commit":"a55af958a319bc31abe58311bab3d19b344a4f0b","subject":"Merge \"Remove designate-pool-manager container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a55af958a319bc31abe58311bab3d19b344a4f0b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 16:23:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-10 06:34:25.000000000","tz":120},"subject":"DNM: Train RDO RC testing","message":"DNM: Train RDO RC testing\n\nChange-Id: Ib95106c96127a81dd757dbd839080a47bff725db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1648db13877c432ce73f920796c9a50821576749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1648db13877c432ce73f920796c9a50821576749"}]},"branch":"refs/heads/master"},"4f650204af83ac1482a008bd9c59c216e161d615":{"kind":"REWORK","_number":4,"created":"2019-10-14 07:41:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/687615/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/687615/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/687615/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/687615/4"}}},"commit":{"parents":[{"commit":"a55af958a319bc31abe58311bab3d19b344a4f0b","subject":"Merge \"Remove designate-pool-manager container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a55af958a319bc31abe58311bab3d19b344a4f0b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 16:23:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-14 07:40:59.000000000","tz":120},"subject":"DNM: Train RDO RC testing","message":"DNM: Train RDO RC testing\n\nChange-Id: Ib95106c96127a81dd757dbd839080a47bff725db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f650204af83ac1482a008bd9c59c216e161d615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f650204af83ac1482a008bd9c59c216e161d615"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
