)]}'
{"id":"openstack%2Fkolla~688390","triplet_id":"openstack%2Fkolla~master~Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196","project":"openstack/kolla","branch":"master","topic":"686963","hashtags":[],"change_id":"Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196","subject":"handle push error properly","status":"MERGED","created":"2019-10-14 11:46:42.000000000","updated":"2019-10-16 12:33:15.000000000","submitted":"2019-10-16 10:27:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688390-1571221654658-fe29816b","meta_rev_id":"7350adc73aba5d2ea59fb1a7ce245771b621d5e0","_number":688390,"virtual_id_number":688390,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-10-16 07:29:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 10:27:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-16 05:40:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-16 07:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-16 07:36:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-10-16 07:43:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 13:23:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-16 05:40:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-16 07:36:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-16 07:43:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-16 10:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc3137d6f731e280335f003693b6f18c1eb00c04","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 11:46:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b54cbea16b6d1476ba324185f4b2dce0f385b418","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 11:47:45.000000000","message":"Patch Set 1:\n\nNote: it also reverts https://review.opendev.org/683200 patch as no longer needed.","accounts_in_message":[],"_revision_number":1},{"id":"beb320cc3333a0d12d5f8850057606f413ee0734","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 11:48:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a3b48ff35fa0e43ecd515755f2d4e16bab59772","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-14 11:55:51.000000000","message":"Patch Set 1:\n\ncommit message garbled","accounts_in_message":[],"_revision_number":1},{"id":"6bd50123ef60e788f5df1fc5a080f9e0ff8d6f4d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-14 11:56:55.000000000","message":"Patch Set 1:\n\nmention the revert - and may we have a test for it pretty please?","accounts_in_message":[],"_revision_number":1},{"id":"db7b9346a62359f47f38c50eed0d8ba595b39473","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 12:02:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a0035ee2de42b899bfd9c4af6659149d1b3d99b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-14 12:24:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nPlease don\u0027t revert my patch, or you\u0027ll break retries again. Please test a retry that eventually succeeds.","accounts_in_message":[],"_revision_number":2},{"id":"baaf776fc37a753ecbafd8188819f44a2ccc44ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-14 12:25:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2dceedd8aeeb024a2134312025b272a2651bd59","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-14 12:42:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/688390/2/openstack-check/tripleo-build-containers-rhel-8-master/d7e8974/ : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fd379c274316973e283c1cd6019ac79d5fa74a13","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 12:53:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"15288c39fc59afc74a63b397092355a22bc00af7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 12:53:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b139fd81f1d769b6ec24394902009f36b1c2c691","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-14 12:57:22.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"df9de942f35fd308483a805da4f8f05a91599b8a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 13:27:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4b5190709fa662e43555c7e7da239898e805b414","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-14 14:13:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/688390/4/openstack-check/tripleo-build-containers-rhel-8-master/dbd31bb/ : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":4},{"id":"ed7bb228a6fc87563c11887099b1a9b4ae02157c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 15:11:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/36ca2263cafa49ff918545e6b9d1d0bb : SUCCESS in 59m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7e87d75b3ebf4d4fad63c098e4f5bba5 : SUCCESS in 1h 13m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c30b1156590444eb9d650e4ddbcda3e3 : SUCCESS in 1h 31m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b99cf11f125c454a89436c54230eee55 : SUCCESS in 1h 28m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7a314ae1179f4bf3916384f311d119ea : SUCCESS in 1h 35m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ce6d73f8c6fa46f88c0b84d63eee14fb : SUCCESS in 58m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/64964a3219de45febe96e95ac32b9cf0 : SUCCESS in 28m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6be870c4c43249f2a4ad7efcb7f89423 : SUCCESS in 18m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9236a49facf749fda42bbcf5dc8f192b : SUCCESS in 55m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb49e0de46404af99d369483190b4907 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118dfe0b92414edea27ea590ce3056bf : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45aa1fc45d3b49b59c54a5aaed440d2a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1ad2595046c49e4b03f95115c00bbb5 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32b6a6dbf955402ebe1e5bd439e5cee7 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d141cb7f60f040f89b9fa93d346007f5 : SUCCESS in 7m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c325bbe056cb482dbb38e54cec1edd56 : FAILURE in 31m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/416da771c364465b8e7edec2f71b3a04 : SUCCESS in 32m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d0f997359b64e05a1d13cf3285944f4 : SUCCESS in 53m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5a8636a728a4ff38807126bcb767818 : SUCCESS in 51m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b296f2a944342d19aaa10499eee452a : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":4},{"id":"551d43d4d271edef2c48b57ee90fa6bf41b18c1b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-14 16:49:16.000000000","message":"Patch Set 4: Code-Review+1\n\nwould be wonderful to get unit test for that (we are maturing)","accounts_in_message":[],"_revision_number":4},{"id":"85e6f0b6f3796a49cb179abca16354705da77d7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 10:12:24.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e would be wonderful to get unit test for that (we are maturing)\n\n+1 for unit test if possible. A test would have prevented me from breaking this.","accounts_in_message":[],"_revision_number":4},{"id":"e26a369f47765b4ef45593a69b8c5e1e17241584","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 10:45:31.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)\n\ntest in progress","accounts_in_message":[],"_revision_number":4},{"id":"1dc570862b302e6f42a8d6dfa98041bc01f7bdb1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 11:04:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"347c9a7b20b62b1531583887483ac1788fe18def","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 11:05:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"096d0180321fbffb153faa3118b83ccc8d44b23e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 11:10:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bdaba9db481e1a0ad6e3446cc562d5fdfa6b08fb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 11:29:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b0e495e0ee4f84e870b0d50339f37fc7714121e3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 12:09:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/688390/6/openstack-check/tripleo-build-containers-rhel-8-master/1f70529/ : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":6},{"id":"13606f08902470bb2806cc35ee1964fa30e4a834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 12:29:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0ddcd30c43334335af5607e29eaddac6 : SUCCESS in 36m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/960115171c06436daca872b3b2180cba : SUCCESS in 55m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/79b47edc5e9b439db14f0b765c08a077 : FAILURE in 3m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/34e894d1598d468c8691bc4acb2f1ceb : FAILURE in 3m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9485f8c2fa164403a1496a2664e43cc8 : FAILURE in 3m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/926663325a544420bcac36c03b4cedbc : SUCCESS in 43m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c4ff671a2a645a2b49575258df984be : RETRY_LIMIT in 1m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/df5509ca992a4deda8175ddf5862ccf6 : RETRY_LIMIT in 1m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b6f099bdaa55446e8887fc1f84eed2ce : RETRY_LIMIT in 1m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdea57d701f74322b00eb4a6a9fbeaf9 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7bf4562a45e4e49b3e2e19a4d527571 : FAILURE in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7736c755fac24a7f9fb9244bcb88d249 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9a834ae468c4863985aeb9d0080c629 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20bd4078d12b48b8b78579f9547f1b9d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6c97def6034b3c90cc57436f4aff6d : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be3db1c26399458a84a32e875646acaa : FAILURE in 24m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c939bda274b4010a666e4b1612c0b96 : SUCCESS in 31m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39fae14bfb6f49cc9a5913ddd3198137 : FAILURE in 3m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba9e106005a841f098edca0e3abe5b2f : FAILURE in 3m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/003e7e6d1cb842019b266e34518f0423 : FAILURE in 2m 49s","accounts_in_message":[],"_revision_number":6},{"id":"4701e7cb5c4694bb734ae8d400ef984471f96650","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 12:58:17.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b9719d7b6030b07939f56acd8e4187e991ed89fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 12:58:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"40d15d6b0c2ce739898216275de2b98757d5d56a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 13:23:49.000000000","message":"Patch Set 7: Code-Review+1\n\nTest looks better, thanks","accounts_in_message":[],"_revision_number":7},{"id":"49bb7861b9c5d77277991e629dfea97005e6806f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 13:33:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"55822ad53ec365298f64422d4d86ad27e3601ac8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 14:13:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/688390/8/openstack-check/tripleo-build-containers-rhel-8-master/87fa500/ : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":8},{"id":"2b5b4ef4c5b3f19a991a9a34834b9a944c4d45cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-15 14:49:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e8c1b34843c6a3fa8faf26c6c5a7f3c89e6cacab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 15:10:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c9a322c829f24121897642f6eb58f8d0 : SUCCESS in 44m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/302fa52531b34e3ead7263d06a4a49f9 : SUCCESS in 49m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f0538a005e3b44aea7750399979a3ae4 : SUCCESS in 1h 28m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d54ecfad766d4b28b9bf0da11ac66318 : SUCCESS in 1h 20m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/486bb3b0c39347a194535490617293a7 : FAILURE in 1h 28m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/915d4fbaba3e4883aba18e8b0c3b89e3 : SUCCESS in 52m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/986da363bc9448eca3c2d0ac421b499f : RETRY_LIMIT in 1m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fddf6300fdd14b18b3c7e8c5b0967a16 : RETRY_LIMIT in 2m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c48b88532e1b417f80fcc39a7c754edb : RETRY_LIMIT in 1m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c68d73ecc1a49fbb61838214f19670d : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ab0da5041c4997b6e565892e248ff0 : FAILURE in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be5965bcb6b34491b12b621e1f57f39d : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c9c800a7a854360b990ee03660db05d : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/811f64edc7b44079912e933c40c46bd9 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26762c4496d648479b3cfe3482fe8bf4 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa078102210f4994b5ed44c0cb8ca45b : FAILURE in 26m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b9a2750e0ee4d38b396a342f9ce7f1b : SUCCESS in 31m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64c44a93475745a0ba2d2eb15954a195 : SUCCESS in 50m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f0492f9550d14fd892640ed81c193415 : SUCCESS in 52m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11f04b186c83478c885cdf66ce37c1e9 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":8},{"id":"3e2ed0e58b282303d0bfc469d222e80087eba1d4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 16:00:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fb09901b3faa0af3798ceb919960e48b845f71cb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 16:00:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"60e7615efc6252cc110d3a0f269cbde2433038e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 16:18:52.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"5e1a54efda5d6be6923d573d987a3cc34c93f4f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 17:10:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/688390/10/openstack-check/tripleo-build-containers-rhel-8-master/5cff72d/ : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":10},{"id":"48db8277efb83475d4c6877ec366ca60afa21b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:03:07.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/030de24c1dc4460c8d00550ac76797d8 : SUCCESS in 52m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8f5fb5b8507544da8ec7c55d47f1a143 : SUCCESS in 58m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/732cdf3639224795834cebbccf62a36f : SUCCESS in 1h 36m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ad1cb72c7158412cb15d70c4c9f9536a : SUCCESS in 1h 21m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7b310198560d41aa9416c19457584123 : FAILURE in 42m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/46b5763f134d410ab753d84cc8e97c6f : SUCCESS in 1h 04m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/174d7f8e9b7f44cba6a2617326d26b25 : RETRY_LIMIT in 1m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8dc7a7a9fd194f4cad481da20b8399f1 : RETRY_LIMIT in 1m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36a4c673e89140ebafa909ee9ac9f030 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5353bc9bf1c4ebe849fd1d88ef00ddc : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bec2db022e453c9ac9b3ddeba09c9d : FAILURE in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e8d793b57ce48108b3b87cf685d4318 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f16ba5424e446b8735e37515f4ad0d : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45f16529a5ed49b4a561f423d69870e3 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a074e4685b2d4660bcb563a8b0280c6b : SUCCESS in 6m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe30eb57a91d4cdf82fa7d9c40b4f54f : FAILURE in 33m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6593da1d8aa14bc381e19b999679c5bb : SUCCESS in 39m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea5516c50b9545008d319c92236f8551 : SUCCESS in 1h 15m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa2b0fdbff4245e19836b4783f5e5c4c : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ce502e7a247d45faa89fbe49f5032b0a : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":10},{"id":"e6afcf0c347d7baf4e590cabbe2b93aaf0809640","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 05:02:16.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"96ced266b5abeece9b56da59c3ae1e7f8e96f9c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 05:40:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/688390/11/openstack-check/tripleo-build-containers-rhel-8-master/428ee8f/ : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":11},{"id":"d8de7c9b3a7a5a6be621454e796516aa2d65a4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 06:46:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/353a4e393d71437daa57db185372d553 : SUCCESS in 38m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/262088356b444fc1b01240e00cbecdc2 : SUCCESS in 58m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9763d9c62392433894719cebb0647177 : SUCCESS in 1h 36m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2ea6dff6f2934136aed17fff7103b16b : SUCCESS in 1h 26m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/34c0095021b247dbabba731a53ff23e8 : SUCCESS in 1h 37m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4b72d36cf6e74c8290b3afc5a0df16c6 : SUCCESS in 51m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ce2002d24333440283c112274c788e82 : SUCCESS in 25m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/96c51e8f56f741759c3659b072f058d8 : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ea48366416ff45c3a29d834e3feef619 : SUCCESS in 50m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/307d6fbb5731443c88c5f7f2ef3faa5d : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446d25ae11404992b936599105bce770 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f08d80eb9fbc4a05995bb19ea6e38eb6 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75537e249be24d649ddc993ff152bdbb : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6c7ad21cab04bba87fed23958f087ca : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b7f57963b948edb2078396f4b15a8d : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4fff8f193fd24bd08d2cd9a39704420a : FAILURE in 44m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84cb0a2ce48947b0b1e7589302b299d3 : SUCCESS in 30m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2dec8d1766a4749802a4dd74ff64afb : SUCCESS in 51m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8efe5e6128ef42819fabd7e0f4fd9304 : FAILURE in 1h 00m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a8ba63f7043944a7ac22ddbf26ff752a : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":11},{"id":"d74ac56033c403302d0da5e781a03ad196852efc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 06:56:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0e9a05b3a62e7740b497641947dc70dafbad2cba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 07:21:40.000000000","message":"Patch Set 11: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7333366bdd7a041a622c7b62e5ee5507d9e65c85","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 07:22:41.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c4ef7e81c31a41573eb2a0adc44d59a39659326e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 07:29:25.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6b49323324b12f97ed0025ba10e2f1083deab84c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 07:35:06.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a19c5f805fe317e131902ecab24f4f273ea19dc6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 07:36:04.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6975aeb8795174192735e8661caadebb3ab2de78","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-16 07:43:23.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"f80375e32ce4cbf81b5220fab311c4a48d117186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:43:38.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0587fe4c69403e350f1e0b5ddf1f5e1bcb9664df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:27:34.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7f9f45d3faca418abc51f1638051d11c : SUCCESS in 58m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/64968ae9ae6c470880a71b953e71272f : SUCCESS in 1h 15m 46s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/76fc0882b77d4c7cbeb9cdcb0ae77c01 : SUCCESS in 1h 29m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/375c5a43a39249e7a39552e6093b4d6d : SUCCESS in 59m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1da02baf4b4f49aca81f82256634ebf6 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dcaf08c7bec44a1a512f9d26384c7c2 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47cf0a2e0d134226a47db1494325d665 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf3b6573c9104a858dadbc3000de9197 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/569a4333d5db4ab684f8b71faba2e82d : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cabfeb953f3435388594e7358f91efd : SUCCESS in 5m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e34f80bf6ce4a399a25c1556da871cc : SUCCESS in 35m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/81307ced3beb45099d95811753daef42 : SUCCESS in 1h 05m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ac6a7fe35bd4656b3c671226a8c86b6 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":11},{"id":"f5c570ee711e2f0bb15aa6659652889031b4535e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:27:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"3b33974b41a7a7d0dcd77da231e188f83228fdf6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:29:22.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e009835f978947b68c56de630f7bcb3e : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":11},{"id":"cec10f309e78cd665ae7a258e5d7a092d8417441","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 10:38:14.000000000","message":"Patch Set 11: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d6f46d544fd757fdc811d738422f85e64a6dc09b","accounts_in_message":[],"_revision_number":11},{"id":"0369b73ea3094715a2491e74b0a19a191aadf6e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 12:33:15.000000000","message":"Patch Set 11:\n\nQueens too?","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"52cac09d3d0b8dcdbc8d68a0bcc8092008220309","revisions":{"69cd3d16480c1a5dc5556e1ec1e8021250a0f251":{"kind":"REWORK","_number":1,"created":"2019-10-14 11:46:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/1"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-14 11:46:26.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69cd3d16480c1a5dc5556e1ec1e8021250a0f251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69cd3d16480c1a5dc5556e1ec1e8021250a0f251"}]},"branch":"refs/heads/master"},"efc3344dd3105becde84b8a12efb247812703d41":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-14 12:02:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/2"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-14 12:01:57.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc3344dd3105becde84b8a12efb247812703d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc3344dd3105becde84b8a12efb247812703d41"}]},"branch":"refs/heads/master"},"107f27f777761bd9397c801099bd592c66b04107":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-14 12:53:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/3"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 12:53:26.000000000","tz":0},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/107f27f777761bd9397c801099bd592c66b04107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/107f27f777761bd9397c801099bd592c66b04107"}]},"branch":"refs/heads/master"},"0e4fc9070ec575ed4f43221b72687a28129c10eb":{"kind":"REWORK","_number":4,"created":"2019-10-14 13:27:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/4"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-14 13:26:52.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e4fc9070ec575ed4f43221b72687a28129c10eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e4fc9070ec575ed4f43221b72687a28129c10eb"}]},"branch":"refs/heads/master"},"4971fd8b1b5d85519ad18129d895c1cf36a1cf80":{"kind":"REWORK","_number":5,"created":"2019-10-15 11:04:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/5"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-15 11:03:49.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4971fd8b1b5d85519ad18129d895c1cf36a1cf80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4971fd8b1b5d85519ad18129d895c1cf36a1cf80"}]},"branch":"refs/heads/master"},"6a3c1eb70b40b2d620de370c8a7530cb75b91cfe":{"kind":"REWORK","_number":6,"created":"2019-10-15 11:29:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/6"}}},"commit":{"parents":[{"commit":"97c78790897598890a51ee5677e2d39b37b147a3","subject":"Merge \"setup.cfg: Cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97c78790897598890a51ee5677e2d39b37b147a3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-15 11:29:39.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a3c1eb70b40b2d620de370c8a7530cb75b91cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a3c1eb70b40b2d620de370c8a7530cb75b91cfe"}]},"branch":"refs/heads/master"},"7cae6535901e7bf32fdde795813a6ea7c395658a":{"kind":"REWORK","_number":7,"created":"2019-10-15 12:58:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/7"}}},"commit":{"parents":[{"commit":"97c78790897598890a51ee5677e2d39b37b147a3","subject":"Merge \"setup.cfg: Cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97c78790897598890a51ee5677e2d39b37b147a3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-15 12:57:44.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cae6535901e7bf32fdde795813a6ea7c395658a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cae6535901e7bf32fdde795813a6ea7c395658a"}]},"branch":"refs/heads/master"},"b520a28a574e580c57487d45d6f052d72f33ec47":{"kind":"REWORK","_number":8,"created":"2019-10-15 13:33:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/8"}}},"commit":{"parents":[{"commit":"97c78790897598890a51ee5677e2d39b37b147a3","subject":"Merge \"setup.cfg: Cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97c78790897598890a51ee5677e2d39b37b147a3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-15 13:33:01.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b520a28a574e580c57487d45d6f052d72f33ec47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b520a28a574e580c57487d45d6f052d72f33ec47"}]},"branch":"refs/heads/master"},"8a590e609222d70d136eb7723b410cd89c0e55ff":{"kind":"REWORK","_number":9,"created":"2019-10-15 16:00:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/9"}}},"commit":{"parents":[{"commit":"3da7f07dd65701bcde5fb09cc4e8b20e00894362","subject":"Merge \"Use C.UTF-8 locale\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3da7f07dd65701bcde5fb09cc4e8b20e00894362"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-15 15:40:29.000000000","tz":-420},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a590e609222d70d136eb7723b410cd89c0e55ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a590e609222d70d136eb7723b410cd89c0e55ff"}]},"branch":"refs/heads/master"},"dd57559ff5c425c583fc9ff72bf557ec5f5fd5cf":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-10-15 16:18:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/10"}}},"commit":{"parents":[{"commit":"97c78790897598890a51ee5677e2d39b37b147a3","subject":"Merge \"setup.cfg: Cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97c78790897598890a51ee5677e2d39b37b147a3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-15 16:18:52.000000000","tz":0},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd57559ff5c425c583fc9ff72bf557ec5f5fd5cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd57559ff5c425c583fc9ff72bf557ec5f5fd5cf"}]},"branch":"refs/heads/master"},"52cac09d3d0b8dcdbc8d68a0bcc8092008220309":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-10-16 05:02:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/688390/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/688390/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/688390/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/688390/11"}}},"commit":{"parents":[{"commit":"88d26e52d350683f4407191ccb196a203167e756","subject":"CI: pep8: Fix yamllint error on .yamllint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d26e52d350683f4407191ccb196a203167e756"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-16 05:02:16.000000000","tz":0},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52cac09d3d0b8dcdbc8d68a0bcc8092008220309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52cac09d3d0b8dcdbc8d68a0bcc8092008220309"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
