)]}'
{"id":"openstack%2Fkolla~688430","triplet_id":"openstack%2Fkolla~stable%2Fqueens~Id49f320513674dfd0b7f7072edf4b9d3f18a53d1","project":"openstack/kolla","branch":"stable/queens","hashtags":[],"change_id":"Id49f320513674dfd0b7f7072edf4b9d3f18a53d1","subject":"Bump queens octavia to 2.1.2 (CVE-2019-17134)","status":"MERGED","created":"2019-10-14 15:42:10.000000000","updated":"2019-10-14 21:12:05.000000000","submitted":"2019-10-14 21:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688430-1571087525060-0f5a3271","meta_rev_id":"926a92e0faed15753500d9b8e32dc2359ab90391","_number":688430,"virtual_id_number":688430,"owner":{"_account_id":30740,"name":"Joseph M","email":"josephm@inmotionhosting.com","username":"noxoid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 21:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-14 16:06:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-14 16:54:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-14 16:54:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 16:06:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-14 16:54:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-14 21:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5530f372876e18365a4aa2202085202528781bc1","author":{"_account_id":30740,"name":"Joseph M","email":"josephm@inmotionhosting.com","username":"noxoid"},"date":"2019-10-14 15:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de6ac3f9c443cf413c39dead4f8f745d62526796","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-14 16:06:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"704d4af4d4179bd64070114887149063e973ea4d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-14 16:54:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"906084e678264bcb8305364c1ebc812a31180282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:09:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9c7591980b6e48698dfac0e1114ee89a : SUCCESS in 40m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/28bb1b9de6eb426eb2538b669223fdc1 : SUCCESS in 46m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a1092f9bbdfa4bd6b06dd7db7384060b : SUCCESS in 1h 20m 50s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1df6360f30dc42d8b8ae0813db73505b : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/e4cc5bbf4fa740c190dab63a4db4a84d : SUCCESS in 42m 05s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/71e74501b53c4c72970e3c6340f60abf : FAILURE in 38m 21s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/fba791f6162a4fda808e7471bcca6bc2 : SUCCESS in 1h 24m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/afcfd5e9020440e0a7f181c8586a2d48 : SUCCESS in 1h 13m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb41aa51cc0c4a2cbef2eff9fd89fa19 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1abacf1228b64d05bc238e21c47971df : SUCCESS in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/54b6914a63d648fe95ccd29e7259f8c4 : SUCCESS in 4m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b393f6a5cc754ff1a0e0347330976508 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ce82059cb3c4cce99ea35c971b445b8 : SUCCESS in 26m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b2f6112d95e40f590446d62822da1d6 : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0277c99d5f064f42934425121bd611e4 : SUCCESS in 53m 27s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ec8fe2506ac4b7a93c29efc325da585 : SUCCESS in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"994d3b2779050a598a829fa84f86846b89baa380","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-14 18:18:01.000000000","message":"Patch Set 1:\n\nrecheck\n\nrandom error","accounts_in_message":[],"_revision_number":1},{"id":"3cd578de13fb0eaeaeeba13e25d895d54fc21197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:16:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3d98fbbda3314adf98666c025758e42a : SUCCESS in 32m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4efd5424a6254586a97d9dc25108d00e : SUCCESS in 38m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ab07ef9da24040d78bbd30233bd0833a : SUCCESS in 1h 22m 41s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/77a0981450764599b8fe555921e5f80c : SUCCESS in 1h 51m 31s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5f4b5c6b315f4f8a98ffc426e01f7c3b : SUCCESS in 51m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cb9d41e6e2f14a53a571518b2a77fe44 : SUCCESS in 41m 45s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f7273882f84e4e1e88ecf8eab249c4a1 : SUCCESS in 1h 05m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fd3317fbe4d7441fac13df542983c86f : SUCCESS in 1h 12m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5551e6223a846f8b866d42397b12bf5 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2634f5d7ee0b43eab47c627f03ba9224 : SUCCESS in 3m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d11b3a429fe7479eae40028d656a47bb : SUCCESS in 4m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7fc4e78b458546b5a52fcd4e25a7ac1c : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c795dc85af84510a304420e0e2f2a1e : SUCCESS in 31m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84f3f14428e24d23acd51c9e505c21b8 : SUCCESS in 31m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9122a5931f9140dea18faa71e4fcaa70 : SUCCESS in 53m 44s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a4dfd7ff1294729938bd824f9e1be3c : SUCCESS in 58m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2739b5819ea606b095b3a44d5e4b46491a57f652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:17:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6ebf57a1b235d10a22a15a2138f87ef5a7aebb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:12:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2bb2bcc6f5db456f8e26fdefb1cc1de2 : SUCCESS in 46m 42s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2f3db134d40b40d0ad613e790bc3cae5 : SUCCESS in 46m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25dddc4f38142f485339a4c25359c85 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd612fb403bf44a1a7123ff99951443a : SUCCESS in 3m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/20d360abb2924ab1b68009104cbe43f9 : SUCCESS in 3m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/edd486e36a6443f19e628f8517d388e8 : SUCCESS in 4m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/484acd095bee4ceaae2ff0f065f428cc : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"926a92e0faed15753500d9b8e32dc2359ab90391","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:12:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a36add61a48e664fbcb79c490cdff254f0dfdf96","revisions":{"a36add61a48e664fbcb79c490cdff254f0dfdf96":{"kind":"REWORK","_number":1,"created":"2019-10-14 15:42:10.000000000","uploader":{"_account_id":30740,"name":"Joseph M","email":"josephm@inmotionhosting.com","username":"noxoid"},"ref":"refs/changes/30/688430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/688430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/688430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/688430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/688430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/688430/1"}}},"commit":{"parents":[{"commit":"27a1000215399a28f3cccee1953bae26f928966c","subject":"Merge \"Install rsync in swift-base containers on Debian/Ubuntu\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27a1000215399a28f3cccee1953bae26f928966c"}]}],"author":{"name":"Joseph M","email":"josephm@inmotionhosting.com","date":"2019-10-14 15:42:02.000000000","tz":-240},"committer":{"name":"Joseph M","email":"josephm@inmotionhosting.com","date":"2019-10-14 15:42:02.000000000","tz":-240},"subject":"Bump queens octavia to 2.1.2 (CVE-2019-17134)","message":"Bump queens octavia to 2.1.2 (CVE-2019-17134)\n\nChange-Id: Id49f320513674dfd0b7f7072edf4b9d3f18a53d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a36add61a48e664fbcb79c490cdff254f0dfdf96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a36add61a48e664fbcb79c490cdff254f0dfdf96"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
