)]}'
{"id":"openstack%2Fkolla~688735","triplet_id":"openstack%2Fkolla~master~I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486","subject":"CI: pep8: Fix yamllint error on .yamllint","status":"MERGED","created":"2019-10-15 15:30:12.000000000","updated":"2019-10-16 08:06:12.000000000","submitted":"2019-10-15 22:01:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688735-1571176867307-70f2cce9","meta_rev_id":"05c9c5b87d8c0c42bec510c06dd054ab00fe51cb","_number":688735,"virtual_id_number":688735,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 22:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-15 15:40:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-15 15:30:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-10-15 15:40:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2019-10-15 15:30:49.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-15 15:30:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 15:30:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-15 15:30:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-15 15:40:54.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-15 22:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b9de3a1020be75a9c1c933a4cba927b136d930b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-15 15:30:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a884328fb3f434e5508a6af78976b3b6a9e28a51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-15 15:30:37.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"d05b22e3b686e0c13caae7bb1e8415164c3c90f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 15:30:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ecf8b61a3cdc9a7ddc5e2696d6bfa6b7cac9b62","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 15:40:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96de55a4a8ab513c7794a10463da9c056592a5b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/75ba47a7a5414224970e993ace860534 : SUCCESS in 56m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a4c7658e53f347ffb765753f074fa13f : SUCCESS in 50m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8f610858d5794cbcbf562b11a7fc31d4 : SUCCESS in 1h 37m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8955e0d4d487423eb2d5018de0c89213 : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/585a91a8d4d547d590cfa1ec64d61950 : SUCCESS in 1h 34m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23489a3c3cc648e9a71258458a512016 : RETRY_LIMIT in 2m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f0296e45801c4ed091ba1daaad7a1c22 : RETRY_LIMIT in 1m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3203a2fcdddf47978a348a337a0635d8 : RETRY_LIMIT in 1m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3254a90dba59485086e0beff04123f0a : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945174f0f4e64579b97e63197d8a9686 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a1403aa9e8e4f31a72d95e7d307d4b2 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30eb6244550546388060e4a5233d1ad3 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6fb5414c234c4aa8a2378691d5f5900f : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab08e9519102436ca3ed88317cff31a1 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/754203adaa934bcaa3f1eb4bddc1b5eb : FAILURE in 28m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1473705cbe2b4314b31df056486f18a8 : SUCCESS in 30m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e309d8b65154071b41dec9b2dab6698 : SUCCESS in 54m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5d08ba2005a419aa20630fd2daedeb8 : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2aee8eceb99148cfb1c85b7d20f0a054 : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":1},{"id":"663751d7e7fcb994cea8369ccfa583c199cb30f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:14:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00fe8d43ae3ca28595dabbd57f48107ec2936508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:35:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b16ca193eb7a43909b3570e0c22f3cf1 : FAILURE in 7m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/88a3ff3b8a5b4356ae985838c3380ebc : FAILURE in 7m 47s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/eb3474f630bd468e94be114463c85424 : FAILURE in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63001460bb2f421ca2b8096f0ead73cd : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/471baa7d884541cd8c8de7d386fdeaa3 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcb50fc5502d4bd1bb17adb890e38c42 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7e63dc88c4e4709b6dc13d489bdc273 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea5e494eaf24490580ad2e4584e14dd0 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733e844618344eaaaeacf21cd27c5c8d : SUCCESS in 6m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af8258a8c19442f3ac7dc7315a235c67 : FAILURE in 5m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f70c1a02d5148048f7beef28b3dc11e : FAILURE in 7m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac7ca0bb9a99450fbfb92cb4242af459 : FAILURE in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3af9a05d6bfd8f440e397fe6f8c271a6a12c1b90","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-15 18:49:44.000000000","message":"Patch Set 1:\n\nrecheck\n\nducking registry","accounts_in_message":[],"_revision_number":1},{"id":"52f791b36e65dc247b55aea776c457c15f3bf6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:18:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d5e7bd13ef194cb5a2b3768bd4310b06 : SUCCESS in 40m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1d68ef56f8b74d2ba57dd9c8dda95294 : SUCCESS in 55m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/539f2bfc76bb4f9cad7845d6e5e17802 : SUCCESS in 1h 16m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/065b856a0360449bb46292a41c1a8d93 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c65d546c464e4b1cb832e781bb757230 : SUCCESS in 1h 23m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d7c139de4ec6472fb17522606db6972d : SUCCESS in 34m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8732c31ee90646bb98cd2371d1fa38d7 : SUCCESS in 26m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a2db311e7f2144eb9c3abefa76b300a3 : SUCCESS in 47m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7584196ac434595aa4ebe01ee793d1c : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a498e0f9caad4eb282a705eb8f7b7e61 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/daa4dc31fb22419db7f998731d8b36b1 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/152be571a07247c48673540ba5b53762 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bb6e28e112c44c6a4f40184d06c694d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eee182d3f254eefbc536cb7c3213f5d : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc4f21d1fbb3475a995ec66a047a4faf : FAILURE in 26m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e20046c8565345d3967e3c8a9cd5eb52 : SUCCESS in 34m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0834ac27111e4821926146255ba83820 : SUCCESS in 51m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7800e3d4983a48f0ab89bcdcfdd1b5df : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85c8125989684a86b5f835bfe16e3fd9 : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ff984f6dbc5010c8fcc9e868dfa3dcd967559d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:18:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15e6dd73723810fcded856a955d13e44cb6fa171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:01:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/620888f1e4b540b1b8645579bbc80e36 : SUCCESS in 1h 00m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d21afc1caf234b33afa092bb7ea1afe1 : SUCCESS in 1h 40m 31s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a296fbcd46c54c93b1fac88a9efd76db : SUCCESS in 1h 27m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59adf43bbd6a47fca82ca9fec1a44cc4 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3ee53879d44653a0c07c43cc164c20 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1695bd880c104bd5a277b5f9b358d060 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334437545d664fb4bf240901a9aab4df : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6374c62f5ce4fcd9a5d56880bafd1bd : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7581d0b81e1d493494d95f678319c7b8 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7277e96685e843f3bae172d3835adfc2 : SUCCESS in 32m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52fa4f011dac4764a3581c9f383f8d68 : SUCCESS in 59m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/845ac3502bbd4b1397b46ef37c2e6518 : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a7378b663960e9eaa7fd8858fcc2e95086986ead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:01:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3aa8e90d9dda965cdeee84aabb6440349a464e17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:02:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb73d18ac36b4464a11c6d686c0116e9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1},{"id":"e443120c1aaff3f0c5196a79d264a271129d0e57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 08:05:57.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit dabf34c9901b7e213621e3e7adf9296ed38f9142","accounts_in_message":[],"_revision_number":1},{"id":"05c9c5b87d8c0c42bec510c06dd054ab00fe51cb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 08:06:12.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 350873a00c03599aab4010dbaeef5b52db9ba632","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88d26e52d350683f4407191ccb196a203167e756","revisions":{"88d26e52d350683f4407191ccb196a203167e756":{"kind":"REWORK","_number":1,"created":"2019-10-15 15:30:12.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/35/688735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/688735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/688735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/688735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/688735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/688735/1"}}},"commit":{"parents":[{"commit":"97c78790897598890a51ee5677e2d39b37b147a3","subject":"Merge \"setup.cfg: Cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97c78790897598890a51ee5677e2d39b37b147a3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-15 15:29:13.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-15 15:29:13.000000000","tz":120},"subject":"CI: pep8: Fix yamllint error on .yamllint","message":"CI: pep8: Fix yamllint error on .yamllint\n\nChange-Id: I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d26e52d350683f4407191ccb196a203167e756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d26e52d350683f4407191ccb196a203167e756"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
