)]}'
{"id":"openstack%2Fkolla~688855","triplet_id":"openstack%2Fkolla~stable%2Fqueens~Ibd3bc04f3cedbcf3e029579877f60e90eb7081e1","project":"openstack/kolla","branch":"stable/queens","topic":"collectd_queens_backports","hashtags":[],"change_id":"Ibd3bc04f3cedbcf3e029579877f60e90eb7081e1","subject":"Service assurance plugins and write_prometheus","status":"MERGED","created":"2019-10-16 06:10:50.000000000","updated":"2019-10-16 10:27:36.000000000","submitted":"2019-10-16 10:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688855-1571221656125-25c6a380","meta_rev_id":"029d632afba00c29d4ce23e29f5700e64203a2e7","_number":688855,"virtual_id_number":688855,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-16 09:10:40.000000000","_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 10:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2019-10-16 08:38:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-16 09:02:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-16 09:02:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 06:10:50.000000000","updated_by":{"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"},"reviewer":{"_account_id":10061,"name":"Joe Talerico","email":"jtaleric@redhat.com","username":"rook"},"state":"REVIEWER"},{"updated":"2019-10-16 08:38:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-16 09:02:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-16 10:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4256335f95eaac403c8b029aa609b82cb89d310b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-10-16 06:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26bd2ee944805ae06db6f9080833106cfadda0ee","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-10-16 06:36:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dfa814332ace599c62bee9a29b82847a3797b952","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-10-16 06:36:34.000000000","message":"Patch Set 1:\n\nThis basically backports rocky to queens.","accounts_in_message":[],"_revision_number":1},{"id":"d918b748e4a2d090920ec13224c54185304c5498","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 06:39:41.000000000","message":"Patch Set 2:\n\nconsidering queens is old and nearing em (planned in 9 days) I would prefer to keep it the way it is; we will probably stop building new images when it reaches em so it can as well be patched locally","accounts_in_message":[],"_revision_number":2},{"id":"3104a076da00958ecae60929ae729abdbf8d80f0","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-10-16 07:25:51.000000000","message":"Patch Set 2:\n\nI understand. \n\nI would prefer to have this merged upstream before doing downstream backports.","accounts_in_message":[],"_revision_number":2},{"id":"8dd506476fe8e54db96ecc3b0d411da1c69714a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 08:16:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1b7880b50df44f04b0c40582394be3b2 : SUCCESS in 37m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/21185557493d4fa7a163a2efcb3d951e : SUCCESS in 34m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3372a60fb71f4fad9e72f2a3c312a633 : SUCCESS in 1h 13m 48s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ba54d5126a0645a285bdba1718fc6ce5 : SUCCESS in 1h 11m 31s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fc609cff13354914a38aec265ee6d2ea : SUCCESS in 49m 54s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/47fd4b6db43b40f7b57b1e324baa7d29 : SUCCESS in 42m 47s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4fb1413485d44f19b1c282df58d8705d : SUCCESS in 1h 25m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/004aac42417042c8b67e83c2f1b46a31 : SUCCESS in 1h 10m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8473aac3dfe04ef5bce6bed2133c3258 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c741159628649c888a5256ec5b1db48 : SUCCESS in 3m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6840e4192b15422fb8ea2e9eee4a6bbf : SUCCESS in 6m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4b20b643e267428a9f5e36cb187f7e78 : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b526ee3489da46d8b2362e2cc8e618c2 : SUCCESS in 41m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/93d82c2d48e34064a9f9f2b7c58862b9 : SUCCESS in 42m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b73a0b22b74c45519a3f659f454c3ebc : SUCCESS in 48m 12s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43e0c71dcf9d4d15afe8f26cd96dbc4f : SUCCESS in 54m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f63ba8af8f423f416f9a8a8ec5e24b1b4fe6ad2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 08:38:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"811ffffeaa46c140f0dbb1fc1698c6a55c924f0e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 09:02:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlast kiss of queens","accounts_in_message":[],"_revision_number":2},{"id":"caa1a21065c10c3c4b98673d4a9c155fd5605ecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:02:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe8fb155fb432580d2c79eb932678916cc0d5ba1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-10-16 09:10:40.000000000","message":"Patch Set 2:\n\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"14ad8eabfd13863ba25050aed32aed7a19c05615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:27:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4bdea59f024d4675a958f8ebce040d3d : SUCCESS in 49m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f074e669e1fe4de6864e52327be20a10 : SUCCESS in 44m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f30945acfb4687b3370b67d62b28b3 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e331aedcc7ab4fd7a36b610f3d68410e : SUCCESS in 3m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fce618d252934dfc84532d8ebf8a055f : SUCCESS in 4m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ee77874381b246d784977b284af8f993 : SUCCESS in 3m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/023ab8ce4e9e4c68a4004fda9b6e5f39 : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9470e65492c2be91dc6af9935d7ee242724b539f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:27:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87486e72f58def28805f9e9bbe125f77b7e76149","revisions":{"cc6fe496d3613956993975b92a5351888ed8bb40":{"kind":"REWORK","_number":1,"created":"2019-10-16 06:10:50.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/55/688855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/688855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/688855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/688855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/688855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/688855/1"}}},"commit":{"parents":[{"commit":"a36add61a48e664fbcb79c490cdff254f0dfdf96","subject":"Bump queens octavia to 2.1.2 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a36add61a48e664fbcb79c490cdff254f0dfdf96"}]}],"author":{"name":"Joe Talerico","email":"jtaleric@redhat.com","date":"2018-04-19 20:22:37.000000000","tz":-240},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2019-10-16 06:08:39.000000000","tz":120},"subject":"Service assurance plugins and write_prometheus","message":"Service assurance plugins and write_prometheus\n\nThis enables to send metrics and events to AMQP 1.0.\nwrite_prometheus provides an exporter to be scraped by\nprometheus.\nThis also moves turbostat to x86 plugins\n\nChange-Id: Ibd3bc04f3cedbcf3e029579877f60e90eb7081e1\n(cherry picked from commit 706b70d4bd25b93b0c6c0d6150124a2d7cf81055)\n(cherry picked from commit 9bde747dab92e04e4be37d1104446f2573a06ba5)\n(cherry picked from commit 31a7792cbb40cbf9fcaf0bb6fbe8829513370e8c)\n(cherry picked from commit 3f8db42f1fdd1cb392a021bfa3850300cefc41df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc6fe496d3613956993975b92a5351888ed8bb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc6fe496d3613956993975b92a5351888ed8bb40"}]},"branch":"refs/heads/stable/queens"},"87486e72f58def28805f9e9bbe125f77b7e76149":{"kind":"REWORK","_number":2,"created":"2019-10-16 06:36:20.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/55/688855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/688855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/688855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/688855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/688855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/688855/2"}}},"commit":{"parents":[{"commit":"a36add61a48e664fbcb79c490cdff254f0dfdf96","subject":"Bump queens octavia to 2.1.2 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a36add61a48e664fbcb79c490cdff254f0dfdf96"}]}],"author":{"name":"Joe Talerico","email":"jtaleric@redhat.com","date":"2018-04-19 20:22:37.000000000","tz":-240},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2019-10-16 06:36:07.000000000","tz":120},"subject":"Service assurance plugins and write_prometheus","message":"Service assurance plugins and write_prometheus\n\nThis enables to send metrics and events to AMQP 1.0.\nwrite_prometheus provides an exporter to be scraped by\nprometheus.\n\nThis also moves turbostat to x86 plugins\n\nChange-Id: Ibd3bc04f3cedbcf3e029579877f60e90eb7081e1\n(cherry picked from commit 706b70d4bd25b93b0c6c0d6150124a2d7cf81055)\n(cherry picked from commit 9bde747dab92e04e4be37d1104446f2573a06ba5)\n(cherry picked from commit 31a7792cbb40cbf9fcaf0bb6fbe8829513370e8c)\n(cherry picked from commit 3f8db42f1fdd1cb392a021bfa3850300cefc41df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87486e72f58def28805f9e9bbe125f77b7e76149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87486e72f58def28805f9e9bbe125f77b7e76149"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
