)]}'
{"id":"openstack%2Fkolla~688871","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486","subject":"CI: pep8: Fix yamllint error on .yamllint","status":"MERGED","created":"2019-10-16 08:05:57.000000000","updated":"2019-10-16 16:50:58.000000000","submitted":"2019-10-16 16:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688871-1571244544659-f62c7b6a","meta_rev_id":"18c33e5fe9f37136cb00400bd0e331ee50fb130a","_number":688871,"virtual_id_number":688871,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 16:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-16 15:43:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-16 08:51:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-16 15:43:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 08:51:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-16 15:43:43.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-16 16:49:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ede25418734337e32725939ccd4c43e732195c01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 08:05:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dc1643f6ea9bb1477ae40a02fa16cdbe63af9acf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 08:51:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc4534a735c38a5c57379e490f8c3ad982c59692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:49:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cb5d585b7b8f4d37895f78fe36c201ee : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c4f2f92a5ae84c1fa494ee56266750df : SUCCESS in 51m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f65dd0c493ca47dcb21298f5d88a72af : SUCCESS in 1h 24m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c90b838033e24cb19918ae6cea5837bd : SUCCESS in 1h 12m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1b84850deab94091afcd66fb140b94d6 : SUCCESS in 1h 36m 38s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f26065391a7b4dc4815f723445827948 : SUCCESS in 55m 06s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/46d0888be1a545f3a893bf14459628fe : SUCCESS in 55m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15aa65afae864db8b6b215cefe4d1d06 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee0c879d8b684d1c8cb745ec20476d6b : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0f3b68d9b1040b48dd5780227a43230 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a873bab4bd4c43bd71c3ef76bf7ebe : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b1639af95894ff888be24b70dc57d29 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0dec97792664820bbb15128b171cf12 : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0615d3e50344930ab8252f0699ebe86 : SUCCESS in 26m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49fe8120856a4fe19feaa3424893ace7 : SUCCESS in 29m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/88bc9fa357d34f818201efd239699a55 : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf0974e9ddc84a9eaedbae4f35d4c075 : SUCCESS in 52m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2dce6ec53d6f4004bb5223434cef1eca : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0e437e98735e3e0ed3c770b62a4fc476f25bd800","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 15:43:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2016d75a1f49a4b192b784556c6e51a15df05f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 15:44:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1eae2e84fe33ce27ebe571a499d6f8e69aaafc7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:49:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2f9f51863eaf48d49bafac9749e53675 : SUCCESS in 1h 00m 03s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/62d1c46d5aa24a0fab9c645001c990b3 : SUCCESS in 50m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6873bcf413f49dd8a82a3236744edf5 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74b72fac251c4883a8ab840d333399a1 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4021cc496d1645fa8b640941f4bccd80 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf7b7ec2ab704b25b21f99c47c3706d8 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9986797624df42139a5e0526874523c4 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca91290a4d2c4757a4f0b132734233f0 : SUCCESS in 3m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6a8e8f5e38f4bf9877b05def4af6889 : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":1},{"id":"327dd05f54a3f44110ff94000474959a18b51124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:49:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18c33e5fe9f37136cb00400bd0e331ee50fb130a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e69e11276b4093b089df466d570e61 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dabf34c9901b7e213621e3e7adf9296ed38f9142","revisions":{"dabf34c9901b7e213621e3e7adf9296ed38f9142":{"kind":"REWORK","_number":1,"created":"2019-10-16 08:05:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/688871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/688871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/688871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/688871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/688871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/688871/1"}}},"commit":{"parents":[{"commit":"7e31dac6a3e31ea5c54be274a9f99502c66f15b2","subject":"Bump stein octavia to 4.1.0 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e31dac6a3e31ea5c54be274a9f99502c66f15b2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-15 15:29:13.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-16 08:05:57.000000000","tz":0},"subject":"CI: pep8: Fix yamllint error on .yamllint","message":"CI: pep8: Fix yamllint error on .yamllint\n\nChange-Id: I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 88d26e52d350683f4407191ccb196a203167e756)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dabf34c9901b7e213621e3e7adf9296ed38f9142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dabf34c9901b7e213621e3e7adf9296ed38f9142"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
