)]}'
{"id":"openstack%2Fkolla~688872","triplet_id":"openstack%2Fkolla~stable%2Frocky~I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486","subject":"CI: pep8: Fix yamllint error on .yamllint","status":"MERGED","created":"2019-10-16 08:06:12.000000000","updated":"2019-10-16 16:49:05.000000000","submitted":"2019-10-16 16:49:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688872-1571244545734-9e252c24","meta_rev_id":"1d1d9c9f2a1d032985fc56e7be689a71048a85ee","_number":688872,"virtual_id_number":688872,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 16:49:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-16 15:43:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-16 08:51:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-16 15:43:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 08:51:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-16 15:43:54.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-16 16:49:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15fad734a9053f9906db1b091447bf7ddc9a1d74","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 08:06:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"67014ea90749566231211d9891115769ec484316","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 08:51:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2a571681ba16edcd3cdb4d2029356f9b1cb709c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:56:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c0af35b4a3a3498fae32d066f6c6e0c7 : SUCCESS in 35m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/43855fc6a55445efb7b79d5bed9ec5e3 : SUCCESS in 1h 03m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/38769503274a406aa1bd218ec09c8102 : SUCCESS in 1h 34m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5601ace308e04d71b0c6e9350fc55714 : SUCCESS in 1h 09m 40s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4d4c0ceda42e4cc884a968d67cd2fb45 : SUCCESS in 59m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/891ce040dec341e8a824fa1a2fae9937 : SUCCESS in 44m 07s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e12bb4bfcecd4ee6a2dacdad38bdaafd : SUCCESS in 1h 31m 43s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/925e20b0009a47dc8f70e758382de8fc : SUCCESS in 1h 26m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e991d1636354eaca689caa07ee7b21b : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0427dd4594dd45d89103ba82b62efba2 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8abd30e638243a6869afb610c938393 : SUCCESS in 3m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/71191b032e4f425ca7e8a0c9c7bcb8c1 : SUCCESS in 4m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4577838763a34ff282c6ef7a5f37e6e3 : SUCCESS in 3m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d808ace42fd4443e9c0a49b43d123fb7 : SUCCESS in 1h 24m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e18a2b6dc7a47429f7dffaa3dd75044 : SUCCESS in 36m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/944b5dd10bdc488282ed2cf2420aa17f : SUCCESS in 58m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98dfaddf06d647da93cf8626cfb219f2 : SUCCESS in 1h 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59b9359d9065d898e688a467bf7824f53d9ef635","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 15:43:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b01125282066d6b6f1e0d22fd72774870fca8b75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 15:44:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f7765c73671db44c9fb7a94c121039de33b3303","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:49:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d1d9c9f2a1d032985fc56e7be689a71048a85ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:49:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8d2145d0d1964edf96423e7610fe3426 : SUCCESS in 50m 20s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2651633adf0c4c5f9866b112ff75f575 : SUCCESS in 48m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeca58aad61c4451b0f8762256f6ff5c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994abf4c945146378be2d56e61f755b1 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b2e74e2b6c34929ba2cd4a2193f9b76 : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3071fc27f7ab4b9a80e4e1978cd1d172 : SUCCESS in 3m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d6d9e2fbd89840838af75988beb4acbd : SUCCESS in 4m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7332b6f84f0429e956ae68f6a61b584 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"350873a00c03599aab4010dbaeef5b52db9ba632","revisions":{"350873a00c03599aab4010dbaeef5b52db9ba632":{"kind":"REWORK","_number":1,"created":"2019-10-16 08:06:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/688872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/688872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/688872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/688872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/688872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/688872/1"}}},"commit":{"parents":[{"commit":"f788df3dfe8fa351641e007ed6b5d95d47eb858f","subject":"Bump rocky octavia to 3.2.0 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f788df3dfe8fa351641e007ed6b5d95d47eb858f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-15 15:29:13.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-16 08:06:12.000000000","tz":0},"subject":"CI: pep8: Fix yamllint error on .yamllint","message":"CI: pep8: Fix yamllint error on .yamllint\n\nChange-Id: I0ffa25a4bdc7b7966b986e1718b4bb8ce02d2486\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 88d26e52d350683f4407191ccb196a203167e756)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/350873a00c03599aab4010dbaeef5b52db9ba632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/350873a00c03599aab4010dbaeef5b52db9ba632"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
