)]}'
{"id":"openstack%2Fkolla~688897","triplet_id":"openstack%2Fkolla~stable%2Fstein~Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196","project":"openstack/kolla","branch":"stable/stein","topic":"686963-stable/stein","hashtags":[],"change_id":"Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196","subject":"handle push error properly","status":"MERGED","created":"2019-10-16 10:38:14.000000000","updated":"2019-10-21 12:49:17.000000000","submitted":"2019-10-21 12:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688897-1571662085101-9e252c24","meta_rev_id":"fe95e163b825dc8f55605695c66a0a71f7ce49c7","_number":688897,"virtual_id_number":688897,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-17 10:00:46.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 12:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-16 15:44:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-21 10:22:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-21 10:22:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 15:44:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-17 10:00:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-21 10:22:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-21 12:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67b5171468a4bc3e99f6766d98c43c67b17cbf1f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 10:38:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"430656b395bfc5bff5e5181aa8548858ab284a64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 10:38:22.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 40d4d12b219806dc8452c7d7dfcedb77c7d452c5","accounts_in_message":[],"_revision_number":1},{"id":"0e9ba6d6a26d590184524f97ee660544b9d71e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:23:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b956585a508d40f299f69f927bca02ec : SUCCESS in 44m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ce5e0b97ae674bd1bd8640dbfb7add3a : SUCCESS in 53m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bcb00e05f2474bb4ba176f0bbe9c7d08 : SUCCESS in 1h 14m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d8bf7099cdad4423afce7a9d48453664 : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5d55b60e55d6422eb3c2a3dff754f10b : SUCCESS in 1h 40m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a21b6ef0957e4249bba43ee3fb342a2c : SUCCESS in 1h 01m 43s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8e314ea5f1e94da1ba03ec942b1f467d : SUCCESS in 1h 00m 18s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/31322e02d5f04abab45a3287485db93d : SUCCESS in 44m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b712bd81e7d54df28389e19fe1a90633 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772958eacb674e9280934c15c111b7a1 : FAILURE in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5fdfc953d7d45478bc7cf5fc45d65db : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/520c36eadb4644b3bdf923ab257a400d : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d53c72ecb41a49aba946827c87c58433 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182fe2ce5f8f494fa7d4b277b5f2a530 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9b6f585da724bf3a370e50d505f4c32 : SUCCESS in 26m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3fe11f0b28f945098e9186f2c10781fa : SUCCESS in 35m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3640f37da7314b3fbb217c1295daef19 : SUCCESS in 58m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ce3f54c71b743cc9f4328d27d43ccc2 : SUCCESS in 58m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4896a47d0b1e4c89b8188e9c19aef7ad : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a563848d53b38229586756fd033f8e533865476c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 15:44:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f21f0e078acc77300e31db3501953603161987fb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-17 10:00:46.000000000","message":"Patch Set 1:\n\nrecheck (yamllint change merged)","accounts_in_message":[],"_revision_number":1},{"id":"91c908ccc8f1d44474f8755c7bede90ebd762dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 11:46:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/957b4b8bb04f45e9a8e019c242fcb9db : SUCCESS in 50m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a1d4e71630334679b540eb3f8434fe5e : SUCCESS in 56m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/31ecaf7ad4a442e3b783f6315e187030 : SUCCESS in 1h 18m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/80a24a4c2d044466bbc10df7065b5c74 : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/33c8dac57e8b470faf899880d8ff61ca : SUCCESS in 1h 40m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b764b74bfa624f3c819331749bc235f0 : SUCCESS in 1h 02m 56s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5379c5c0e3474710a4bce17373e5c991 : SUCCESS in 44m 45s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d9c0e57d19fa4631a9a466ad857bf8db : SUCCESS in 52m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8c6fe6b61c04aebb01657cb7e236483 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a79bec48e33449380ca4309bae7aec0 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8f2f77fcd884af8a848a2e78d707152 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53f219bb525b46b3bcca20d7bd2852bf : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04057249b77c402db0989d8f0e9e3bc2 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23237dafd7184422ba34191be681a523 : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d62fc22071b34ae0b4d4b49251092881 : SUCCESS in 23m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/330f8547edca420abcef2e8b3b112d95 : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/565526e009944f83bbef3812d0753e65 : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65d6319d20bd4c7ab50aace1735cbb5f : SUCCESS in 47m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4c9ab16467314ee6b05956160915e713 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"46824c467aef94a479dac44eaf856d4771e79b67","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 10:22:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bb3f49055d5e8ea77f6d79d952c49333f1605de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:24:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c38f7f0c20bbb8a07a4dc7888db81562f9a65978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:48:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1fe51adeea9e4ab8bb7d5b8b91984f8c : SUCCESS in 51m 03s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/59d16928ede7470fa6603bd9cbbbbb97 : SUCCESS in 55m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24789c796d8f42288642682cda015b79 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206992c28a624273899cc6effc1df967 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdafe8f291c840df9f796e93ef349528 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8d2a844f9094d628283bb40f4646c45 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/393f26dff4ff42d7a913be34fd8d3642 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48750cdc60534af3804bb63b080c9f17 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0deec8027a404d3ebe67ce0723b93824 : SUCCESS in 26m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dd9ee4a9d477d055bb0e098c441f668d5c274841","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:48:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a81db45b1b30ccbd255a6b3730e775faef01a624","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:49:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a3820d77074774a1f87e84f892330e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6f46d544fd757fdc811d738422f85e64a6dc09b","revisions":{"d6f46d544fd757fdc811d738422f85e64a6dc09b":{"kind":"REWORK","_number":1,"created":"2019-10-16 10:38:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/97/688897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/688897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/688897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/688897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/688897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/688897/1"}}},"commit":{"parents":[{"commit":"7e31dac6a3e31ea5c54be274a9f99502c66f15b2","subject":"Bump stein octavia to 4.1.0 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e31dac6a3e31ea5c54be274a9f99502c66f15b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-14 11:36:48.000000000","tz":-420},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-16 10:38:14.000000000","tz":0},"subject":"handle push error properly","message":"handle push error properly\n\nIf there was some error during pushing then Kolla was greeting with \"all went\nfine\" message anyway:\n\nINFO:kolla.common.utils.aodh-api:Trying to push the image\nERROR:kolla.common.utils.aodh-api:Get http://10.101.16.1:5000/v2/: dial tcp 10.101.16.1:5000: connect: no route to host\nINFO:kolla.common.utils.aodh-api:Pushed successfully\n\nThis patch changes that. Now if there is an error during push then proper\nexception is raised to PushTask and image is marked with PUSH_ERROR status.\n\nThis way at the end of build it is easy to spot which images did not got\npushed to registry:\n\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nINFO:kolla.common.utils:Images that failed to build\nINFO:kolla.common.utils:\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nERROR:kolla.common.utils:base Failed with status: push_error\nERROR:kolla.common.utils:nova-api Failed with status: push_error\nERROR:kolla.common.utils:nova-base Failed with status: push_error\nERROR:kolla.common.utils:nova-compute Failed with status: push_error\nERROR:kolla.common.utils:nova-compute-ironic Failed with status: push_error\nERROR:kolla.common.utils:nova-conductor Failed with status: push_error\n\nCloses-Bug: #1848019\nChange-Id: Id2ab97bf4c0dc3423268a0ea435b56f4a65f7196\n(cherry picked from commit 52cac09d3d0b8dcdbc8d68a0bcc8092008220309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6f46d544fd757fdc811d738422f85e64a6dc09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6f46d544fd757fdc811d738422f85e64a6dc09b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
