)]}'
{"id":"openstack%2Fkolla~688954","triplet_id":"openstack%2Fkolla~stable%2Fstein~Id9312830fb9961927ac8d317c432f52a4ad7af24","project":"openstack/kolla","branch":"stable/stein","topic":"686134-stable/stein","hashtags":[],"change_id":"Id9312830fb9961927ac8d317c432f52a4ad7af24","subject":"CI: increase both limits to 3h","status":"MERGED","created":"2019-10-16 15:08:02.000000000","updated":"2019-10-17 08:58:49.000000000","submitted":"2019-10-17 08:56:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688954-1571302598627-4f81ad49","meta_rev_id":"add7340d6820a9c1f298c01e83403d125ad5bed6","_number":688954,"virtual_id_number":688954,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 08:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-16 15:44:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-16 15:32:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-16 15:44:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 15:32:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-16 15:44:29.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-17 08:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da2a74cd37140bf32574543f7b069e3e17eae71c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 15:08:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"44e23176d537fa6313d7fefa852d36f7132eb453","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 15:32:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"efec6bcaee5c120d1a20d2fd56413193d2785c69","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 15:44:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"785398cd05f2d0114b40100b59a1b2ae155b8f6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:54:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ab9a95f17e0849e6b93bf3790d6d2366 : SUCCESS in 47m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3e7a014aa6464fe98de740d52cadaa8c : SUCCESS in 46m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1bd282eb361d4358801504e64cf27ade : SUCCESS in 1h 15m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/146e8069655d4efe89a621c876a5da8b : SUCCESS in 1h 29m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/eff2702804d8485f8d68d95bb0155234 : SUCCESS in 1h 41m 25s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2b6717c3a3ec47089bfdcbd769aad431 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1093f5d9d5dc42828a01b4d545d24db3 : SUCCESS in 58m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60adaea6e366475faf6ac0b348d7048c : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19af8ca2a9474710ae5ecf8db1f36f34 : FAILURE in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98563ffd24874a2c8d43281a3cc6afcd : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec383e34f33440bcb1356f54da21b385 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2a7b437033c4681a563e623d74e9241 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71890fa782244e1895b0e0e8e3dce5d7 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e73ad4a409c54283822d3c6165cdb670 : SUCCESS in 24m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e551735148543368b577315ce6e51a2 : SUCCESS in 38m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/603ca61d5cc443db84cd9197c15258a4 : SUCCESS in 1h 04m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31865d7486ba4b6fb3eb058e08f318e0 : SUCCESS in 51m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6c662f8177684e11b0c8fd78a3241a7b : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"04d9c71bf2dc5c5c8d4a802c67e652c6f8a8cd7e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-17 06:23:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"38cc9bb22a418997cfabbb8ded8811eaba0c8d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:10:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7af395edba0c4253a269f5be0f231468 : SUCCESS in 52m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/62b2a472440d4595b32587865cc9f4ef : SUCCESS in 52m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fffdcb134cb048cc8055a7345a02cc54 : SUCCESS in 1h 26m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a79d4d3c24f04b22a56f8b00606a2c2a : SUCCESS in 1h 09m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/871a2631fc6145ca9094cb3694ad2ad2 : SUCCESS in 1h 41m 08s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bf209b284657484ca52f936dcbc5633c : SUCCESS in 1h 01m 33s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/75d25648441d45ca9a7d87239038b317 : SUCCESS in 1h 06m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f1b167521de4963bc380e8769e7fde3 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20062b218f45421d820856298bf30715 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57b96b723a464c98820e1fe6838358d9 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00ab59f399e14848be792bc64869a0a1 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e173a298e10404ea43c575013b73e95 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f97c25f1a348e185d9ac19f39087ca : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f54bb48b5bea46ef94baee63406035de : SUCCESS in 41m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/95890a87a04b4d02a8bc13e3e2299f98 : SUCCESS in 39m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ae1cfecd179420a841788330bee096a : SUCCESS in 57m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3de8de95c8d41d0900c27402e3b3580 : FAILURE in 37m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/672182b461894d60962bacb108c7053c : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3d31287f22c44170b15433cbb70adac79b2d3b76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ff670e33011771a92e9b4e4f9a81fc746eb2551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:56:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e840a91a401645fd82e867a69cf287a0 : SUCCESS in 44m 48s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/373ed5717b0a4cdeb3f7bd58f796673d : SUCCESS in 43m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/264740ddb5564b49a0fdf08545de973a : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f614eee05964e8581529b72c043f34e : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/705d80f00605441e972acfadee12f9ce : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e36343faec14aa895697d8bcab5ba4b : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0994b219aca944a3b766e60e662ab3d3 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3784c74d93734ce09f8808d5fe257235 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe42d17c3a7d49a4bc2e103cb8dd13d0 : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":1},{"id":"04aa20c50aca92f5d5dfa3b6a35fd8e5bf139e81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:56:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb95a778b2d673831892a79f137001c9577119f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:58:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a7be90f3c94563b916a42e4cbd04d4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8cddefbe7f9e89cda08530aa4ce772e7b28abcc","revisions":{"c8cddefbe7f9e89cda08530aa4ce772e7b28abcc":{"kind":"REWORK","_number":1,"created":"2019-10-16 15:08:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/688954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/688954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/688954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/688954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/688954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/688954/1"}}},"commit":{"parents":[{"commit":"7e31dac6a3e31ea5c54be274a9f99502c66f15b2","subject":"Bump stein octavia to 4.1.0 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e31dac6a3e31ea5c54be274a9f99502c66f15b2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-04 09:41:03.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-16 15:08:02.000000000","tz":0},"subject":"CI: increase both limits to 3h","message":"CI: increase both limits to 3h\n\nOur publishing job is barely managing to fit in 2h time so let add\nanother one.\n\nChange-Id: Id9312830fb9961927ac8d317c432f52a4ad7af24\n(cherry picked from commit 7e1d3eba928ec54899f90368b0235b6737fda3b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8cddefbe7f9e89cda08530aa4ce772e7b28abcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8cddefbe7f9e89cda08530aa4ce772e7b28abcc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
