)]}'
{"id":"openstack%2Fkolla~688956","triplet_id":"openstack%2Fkolla~stable%2Frocky~Id9312830fb9961927ac8d317c432f52a4ad7af24","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"Id9312830fb9961927ac8d317c432f52a4ad7af24","subject":"CI: increase both limits to 3h","status":"MERGED","created":"2019-10-16 15:11:41.000000000","updated":"2019-10-17 09:19:58.000000000","submitted":"2019-10-17 09:19:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688956-1571303998232-d25b5e01","meta_rev_id":"d9c814ae3a08eac52cac2d30303366fcb3634088","_number":688956,"virtual_id_number":688956,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 09:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-16 15:44:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-16 15:32:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-16 15:44:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 15:32:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-16 15:44:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-17 09:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ba56ab01d6c0b0e1855d7fff9be6549669d356a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 15:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daa8072af42f35a00c9c599792dd768f773ab710","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 15:32:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03b99a81edd319e912a1d931d3f541a54091e963","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 15:44:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b529673fa7854cbfc00bd41af6eaf3cd2d73740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:25:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4875c09c0256404992db5c265f97663a : SUCCESS in 47m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/73dd3a66eafc4b71a3b30d1c565a8a10 : SUCCESS in 46m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7e138a15293f437e950861aa05cb3960 : SUCCESS in 1h 36m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f083b8bf552d4e85a2b6f0f7be3b76f5 : SUCCESS in 1h 38m 48s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/dc76411274cd443e98afe304bb2d7c6e : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6566f16b0f844e539613782faf194386 : SUCCESS in 49m 57s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3be7182b5912425dbaae0b00ae4fdc7d : SUCCESS in 1h 29m 06s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/e13cfac5961a480886d28aef6e1b2ed2 : SUCCESS in 1h 15m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c235a507e8444bd097858a111f0570ba : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf6cb7d641d4818a7a65ef5bf277e0e : FAILURE in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b46d1b01ebbc45548d5230f643c9f471 : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/347926ba756749fea469884be44ae5d9 : SUCCESS in 4m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9a73685462b849a8a7c469ab5b96bb54 : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8e0d800e9d0472aaf3dc731802d5d16 : SUCCESS in 1h 37m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3d12b12bfe7475481fd33d0a268c2c7 : SUCCESS in 34m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca0d8d3e0262490f8c13de8df38ba9f5 : SUCCESS in 1h 12m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a99351deb09f4128ad69a657424f9b74 : TIMED_OUT in 2h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"773885b84adb643c35c58dc4a748b35db76be9a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-17 06:23:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1af02d81c9c12796531d33b374ff38e6eeb51012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a5fcce8257a84cc4a566048ca3e4d53e : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d9d63b8e69274d4aa06564e92d3a9a5f : SUCCESS in 51m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9ae478c4c57b47bcae23d590841196c8 : SUCCESS in 1h 37m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ef74f16f51934c5081d46b3deade2b1a : POST_FAILURE in 6m 22s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5b2f28bb39284303a4e5530705beae23 : SUCCESS in 42m 10s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e107dfe991184f77a1ffe5819a773821 : SUCCESS in 56m 16s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ab843472dcf94d4b8db43f7503ef8212 : SUCCESS in 1h 29m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8ea2c5ff8dfb4c38a455acdce23eeeee : SUCCESS in 1h 08m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04da4abdf0de44a4a929289c13652da2 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c561efeb8ed414f9776fbad02bc3f6a : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77ba3b68f968448db98878f26cba8cee : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f9af476ce42648b4b218787ea10e8daa : SUCCESS in 3m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19d4dea3a0c74acd991f8d179abf5df7 : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed35403768b44476a3ce6980067ff9ba : SUCCESS in 1h 23m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ef72b184c8741988a8e7a74bb6e87a6 : SUCCESS in 42m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d9814853aaa43048d0067281ce6627f : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d70b8431f84c4abeb8cf394f887ec4b1 : SUCCESS in 1h 54m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8691f210d0fc5614d19b7647ad089478269c4067","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:26:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8262eff678cbc87460d77d794540bf84977f2af6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:19:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d524f859d3fa4d72985b5be897b03872 : SUCCESS in 42m 32s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/20ac32d394584c13a2d2448de1111dbc : SUCCESS in 50m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44eee64854444059974fd9e8291f976a : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51cfc58edaec437b904c04c8ead7e29c : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68f7d439d7cb412e9b445a46ddbad95d : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/85aa044456e94005b2046251238510ca : SUCCESS in 4m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4653dacf25574bd6875556a6738be124 : SUCCESS in 3m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f59b6d682e345bd84cf2d17825972f2 : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d9c814ae3a08eac52cac2d30303366fcb3634088","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:19:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4436837c66058d763deda9164ce626cd70a0da9f","revisions":{"4436837c66058d763deda9164ce626cd70a0da9f":{"kind":"REWORK","_number":1,"created":"2019-10-16 15:11:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/56/688956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/688956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/688956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/688956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/688956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/688956/1"}}},"commit":{"parents":[{"commit":"f788df3dfe8fa351641e007ed6b5d95d47eb858f","subject":"Bump rocky octavia to 3.2.0 (CVE-2019-17134)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f788df3dfe8fa351641e007ed6b5d95d47eb858f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-04 09:41:03.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-16 15:11:26.000000000","tz":60},"subject":"CI: increase both limits to 3h","message":"CI: increase both limits to 3h\n\nOur publishing job is barely managing to fit in 2h time so let add\nanother one.\n\nChange-Id: Id9312830fb9961927ac8d317c432f52a4ad7af24\n(cherry picked from commit 7e1d3eba928ec54899f90368b0235b6737fda3b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4436837c66058d763deda9164ce626cd70a0da9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4436837c66058d763deda9164ce626cd70a0da9f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
