)]}'
{"id":"openstack%2Fkolla~689244","triplet_id":"openstack%2Fkolla~master~Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6","project":"openstack/kolla","branch":"master","topic":"bug/1848572","hashtags":[],"change_id":"Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6","subject":"Neutron: apply db migration for BGP dynamic routing agent","status":"MERGED","created":"2019-10-17 18:59:31.000000000","updated":"2019-11-25 10:54:00.000000000","submitted":"2019-11-25 10:30:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689244-1574677850242-1353a74c","meta_rev_id":"6d9352df74ed9c047810830fee2ad65c37e62e1a","_number":689244,"virtual_id_number":689244,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-11-22 17:06:50.000000000","_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 10:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-22 17:44:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2019-11-23 09:22:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-11-25 08:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-22 19:17:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-11-25 08:36:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 16:25:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-18 16:27:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-22 17:44:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-22 19:17:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-23 09:22:24.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-11-25 08:36:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-25 10:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61649feaed8e3de7402097a911695283485d30cf","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-17 18:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c610fe0aaa4310617988c200fc11182164bef00","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-17 19:38:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bf325afafa7d224c984911d4d67f6bb0c81df59a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-17 20:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/2/openstack-check/tripleo-build-containers-rhel-8-master/3fa579e/ : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":2},{"id":"48e3d53c3729fc415ec07a1a09d6a2d82aa2f32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 21:22:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7152164dfb7c4dd7910e171584ff9844 : SUCCESS in 49m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/07c1954481754c3e943d555ee3db00cc : SUCCESS in 48m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d6daa897d0e14cbba2cb2635b8dffd89 : SUCCESS in 1h 30m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/76391f666517400dad33b8957b1304c3 : SUCCESS in 1h 22m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/193df6e9953e47818ad5e0e90201f24d : SUCCESS in 1h 38m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9f1630857ece47cead3277b7eeb1712d : SUCCESS in 59m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/50bf5470ef8f4399baa7e23920f22c17 : SUCCESS in 21m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d360eef1ee2e4fa2be94f6fa4051011d : SUCCESS in 18m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2eea89f1308d4b4fb6a0e67d0f33d660 : FAILURE in 18m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81fc55404442461a8cd7db59e5de3e06 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e65b37fa8241168cc08f9594a82938 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8603d946f35943f78ff752d04c3c3fee : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fae1d9ca12ec452f877c4fbc743684fd : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/644911a34b2342849257af19b26bc72a : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3733902182a401ca1c16d8da143a56a : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe5169477fef494b9cd259aed7937d05 : FAILURE in 23m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ab980a8bf55491b84daebbe74ad3a74 : SUCCESS in 28m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4754a38ae5b249a3a079015378476131 : SUCCESS in 54m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3c463c7171640bebf5ac164fb5f09e9 : SUCCESS in 57m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62f3718ef3964df5befc2c8b56647c03 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"00726db9e7dc1cd17fc2891a1f4486e929f91258","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-18 16:27:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e76ab26fd9394a5001aca22ca1ad4f5db8f38dc8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-21 08:53:35.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"332320a4168ae70b8e2c3d805407573b0cf91f22","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 09:32:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/2/openstack-check/tripleo-build-containers-rhel-8-master/ee664d8/ : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":2},{"id":"36095c0ba59a7a38cffa97239bac46867efb4a4c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 10:03:30.000000000","message":"Patch Set 2: Code-Review+1\n\nthis needs support in kolla-ansible","accounts_in_message":[],"_revision_number":2},{"id":"c47286dfe9ba4bd244230e3f1ba8a54180e0416a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 10:04:31.000000000","message":"Patch Set 2:\n\ncurrently neutron-server handles all db ops - maybe it should still","accounts_in_message":[],"_revision_number":2},{"id":"6eabde83b7dd4c631c7a73786e2745f3147dd497","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-21 10:05:31.000000000","message":"Removed Code-Review+2 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"7f7f6765897f7fecf7d09e055c00a02d7c0acf0b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-21 10:05:59.000000000","message":"Patch Set 2:\n\n\u003e currently neutron-server handles all db ops - maybe it should still\n\nMaybe that\u0027s a good idea, but we would need to do it in k-a and discover for which subproject we need to run the db migrations.","accounts_in_message":[],"_revision_number":2},{"id":"6ead9803a3bd7a3ced1b1cbaa334fdce0fcb6662","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 10:06:01.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"6cd4c244a06fff0ef1f8bb97a5d4f2f63b76ed07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 10:06:27.000000000","message":"Patch Set 2:\n\nWe already do something for SFC.","accounts_in_message":[],"_revision_number":2},{"id":"a612c846b7f455d424129a03722a397fafdc2d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 11:01:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3d766885075e495fa44d0f5b68a70df1 : SUCCESS in 50m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d2ef304ccb2146ae9410e5c06e235a22 : SUCCESS in 46m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9d049b233e0c4659baec48588fc74af3 : SUCCESS in 1h 38m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1f8c5f9b8f54477393c6d92656bbeec7 : SUCCESS in 1h 30m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b571729ad42d40b191e07bf5d817e29d : SUCCESS in 1h 31m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6527ccfca8614acca9d0f9e7da5e3ff3 : SUCCESS in 53m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e82ca1fada564d598ef79cb41e56e943 : SUCCESS in 28m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/33d2484b78464cbb80672107ab034940 : SUCCESS in 23m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2fbf4c225cd440a78805ee333d028036 : SUCCESS in 49m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e979e3168bf442f3a57509c3287735c8 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650aabececad4ee5a1a9b857a10f9bbe : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5951bc24f922465683dfd5b7cf2bd15c : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4be281b005f24920a2fb4f4d2459fb30 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cc262461c684851b5c498176b55ad0d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f8f0de44fd4c82bd59de078279070d : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c1f70eac8614ee3a982bbd58894825b : FAILURE in 27m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c55a79af60e435fa237f655de980e24 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d322c605e1c14877ad0485950b177819 : SUCCESS in 58m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a391319f6754282b10f3e88d9d673a5 : SUCCESS in 53m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d54f2aef8d6b40f589b32edc497e8639 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a15a53c3a5671a83a9b2cac70462b602e0caa4f0","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-30 16:41:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"499043bd2bcc40889888c6fa471e8a1df6b4f22a","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-30 16:42:25.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"317a84968b8ce08ff2fd7c928053344ac5938683","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-30 17:27:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/4/openstack-check/tripleo-build-containers-rhel-8-master/36c302f/ : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":4},{"id":"d3b5144146a0ac6539fa518d533b8154efd1f55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 18:14:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5d8b7bdf2e3f465ba27ba5c03152099e : SUCCESS in 51m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f90477ca29094770b2e71138831cda4d : SUCCESS in 45m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7ecb6b19b58645358cfd6534f65e21cf : SUCCESS in 1h 29m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/83d0cd02bbfa4bbfaf1a038211496b22 : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b8a4fab5d0bb4d6eabd13ccecaab6c1b : FAILURE in 49m 32s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f22f433254d942369439b311d00b5c49 : SUCCESS in 49m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2aec648b3bff49bdb30ec134938e1961 : SUCCESS in 31m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a3b0362744c94b8086b8059910f90fac : SUCCESS in 27m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4492ea4ac7c1404187617dcdcc866a23 : FAILURE in 23m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9c56f3ae1204d23988ba19aaf8137bb : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ff5febd718474fa1e0ebfb942b87c5 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2793584e677c42a7ad673307153cfba9 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e45e7905aca43f68955b52c248359f0 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a283f66b06024235bdd32edb86ffd24e : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab4930427108499fa5149e66634b7b4e : SUCCESS in 6m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/75f1fbe973174a9e9bef07072f62a5f2 : SUCCESS in 32m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7540ada25d0b4647a01ed24e111da086 : SUCCESS in 40m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a00db5fb69346468893be4b1a27ddf8 : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f2cdfcf7d2b41a3b7c11417b2fe84ec : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b627e4dd68b4ca5826832b1bf109fd9 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":4},{"id":"9cbb0d8a80aaff8ed446e030d415640198340e56","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-31 14:29:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1e27ca37e8c60d4ad4f8bbec60cc3cc3c80a3b37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-31 15:08:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/4/openstack-check/tripleo-build-containers-rhel-8-master/6bd3f8c/ : FAILURE in 38m 21s","accounts_in_message":[],"_revision_number":4},{"id":"492fefbcd47cd9702e122377ed5ba6ccc5c8d468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 16:17:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b4a3c3ffd021464691bb84d3ab6a078c : SUCCESS in 51m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/64c4bd7655604845a98bfe4cf6b2c259 : SUCCESS in 1h 00m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd8f07ff280748d983f7e4f2f7c0e262 : SUCCESS in 1h 23m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/af272a76b6814ff488e8852f41251345 : SUCCESS in 1h 26m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/586c2745941140f58d38b6e296fcb864 : SUCCESS in 1h 36m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/89d9c5062a7c4ee999a9aed39b9af3cc : SUCCESS in 42m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b604b1d6da5147e98ce67e7b39bd5d4d : SUCCESS in 35m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6fc83a652235417893a695194923c929 : SUCCESS in 25m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/372f5739492c48bbb1d974da0b061c18 : SUCCESS in 59m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e51172a4c0543d5a1e6884afa0e1479 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64523c2d383b40338e912db7035a820b : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d06033ff0bd0455d911d3f9ed9e3afc0 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddaa7d2fe14841958d6319e79391b585 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/500c7ae1367d412aa8e55ee3468487b9 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0941a8cc70214c409859802401831bfc : SUCCESS in 7m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d17a2ea73aa0490f8f408304c2bffb17 : SUCCESS in 34m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/519b4667f2ec4ecca502fef6d86fb2d6 : SUCCESS in 26m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e46f16061f44746a462866d96c13913 : SUCCESS in 1h 05m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d84ea917b084331b9047df82c68f6b2 : SUCCESS in 1h 00m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a41a7faaabd452eafafe0feb8e72129 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":4},{"id":"a785d520d2a664dc1fb3a6f49d397281f6b9f652","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-19 21:51:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f08ba46152e7c67f6f113f31296f049cffc8884e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-19 22:23:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/5/openstack-check/tripleo-build-containers-rhel-8-master/5b3409c/ : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":5},{"id":"58475e55f86ffa613dd2c3a7936ec985903cb3bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 23:24:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6618965dd3e041819e820c56a5cfe287 : SUCCESS in 48m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b804241e8e76414b853e2e523e07eb91 : SUCCESS in 57m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0f711f1649114159a5ab925a41afe906 : SUCCESS in 1h 28m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3f2d061055c749668d6778bc5d20d608 : SUCCESS in 1h 27m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/674dda66faa54936bb5b6ee5117fa275 : FAILURE in 55m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a8ae01f9c6f84472ac7ccd3322ed6e1e : SUCCESS in 55m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f7f2f790b1834f9082bc4dfb5e0b2f30 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/df8a393d6f86463685a5fa129ae5b1b4 : SUCCESS in 15m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49171bab27794116b6d9aa2633dde81b : SUCCESS in 50m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a011c4ffea74a5480627577ad575aff : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183e2a99a48240be9abc0cf70e2f64ee : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75e8fe270ff6462b96b55aaa00a30d96 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65639256095c470f87d74cf4446b6eed : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd0b217ea6e4d50a34a0f5afaa054a3 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4006261938594248852d9ba7415c7b5c : FAILURE in 24m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c3fc6fa26d24deaa2f76d8e670452f6 : FAILURE in 22m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6b64c6dacde4bfd9b20e30fb999350d : FAILURE in 12m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/901981e1bf8547f293063fe24d0709ce : FAILURE in 19m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b820bd43fc994aedac019faf4ffe1a25 : FAILURE in 13m 16s","accounts_in_message":[],"_revision_number":5},{"id":"07a26c17a0082e4de3f97fccc9c5174bd269a8fa","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-20 00:14:59.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d62c0f984a230213265fae2e03d2e40a5f41eba1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-20 00:46:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/5/openstack-check/tripleo-build-containers-rhel-8-master/8d16136/ : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":5},{"id":"170c3cf1b4d06aebb005f85ad443a797160fe195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 01:59:58.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/35d9a03fc0fd40f89260a7f65e894e3f : SUCCESS in 36m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4150a9039451452690d148048bb671d4 : SUCCESS in 56m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d082a0509327439ba420869984996725 : SUCCESS in 1h 20m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9e6d5440eb4e4644b539e56c12a7f94a : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/053c04265a964abf906a1fb9aedb02c8 : SUCCESS in 1h 40m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c8703ba1e0d4cad9e085996bb2fe8fc : SUCCESS in 50m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ae59e3ba2f3f42b594089dd9e0ba2025 : SUCCESS in 32m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c03b61a87dec4ca4aefbe4b64c0131ce : SUCCESS in 17m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cecaec35fb164a69b5f8ce1a5b123a86 : SUCCESS in 50m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0da6c5557c3c46e1979dafcbbde5cc56 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8898cbc6203b49e0a0118573561c03a0 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8c18caad7f42ebb2cfc4c639159643 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2813ccbd761146f6a841b3049bd91747 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f632807d75494c108637dec36d0fe2e8 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28bd0cde3e25496cb382f4289b80358b : FAILURE in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/54bba7be2e794d498affd46f2a9bd904 : FAILURE in 19m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3a84a32cc93427288b58b65a6f9a390 : FAILURE in 11m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19d339e366e445ecb0254f29de56995b : FAILURE in 18m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fa93dee9ff6437fbe5dbe219bdf0663 : FAILURE in 12m 34s","accounts_in_message":[],"_revision_number":5},{"id":"37cae86c9950e176f6c9049e4434f8275eacb44d","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-20 11:35:21.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d7e29f6f9f31f2e41960b60c56d77efb5ae2f240","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-20 12:07:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/5/openstack-check/tripleo-build-containers-rhel-8-master/9c38400/ : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f0700076acaa7da99ce21ddcbc2fe6f413b9bf1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:20:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ab83291b05144c789d6e856afac908fb : SUCCESS in 45m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eddf3f3206184de8a4c08f0cf188bf07 : SUCCESS in 1h 02m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ecb06baf097e480aa64ada68dcbd038a : SUCCESS in 1h 16m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ff052b11198645bbb166d06b9f44e66b : SUCCESS in 1h 18m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f237ac56fc9a470b9f0fb15ffcf53039 : SUCCESS in 1h 31m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d38d733d2d21481eb23f8b3c16280c0c : SUCCESS in 40m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/02c0bb27dcb84fca8176dc400fec5682 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1f17166cb07a4f349bccd45a7d070806 : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8dde167c1b154b798f6e30053fa76b53 : SUCCESS in 1h 06m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12e2aec74bea405dbcaa4a6b746d8955 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f552afe6e8c438fac33d745add62de8 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb8676b0e66f47fdb4d9b5f342089c2b : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fac0271963b9454c90c5eabc2578fdca : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0084f69971547309647dc00ef0e1655 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aac14e04d63a41c2ac216dda78faa43c : FAILURE in 31m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/085a729c517d46dd8edf918c7ec3cabd : FAILURE in 22m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4df3fa7d2f7469e982c2fcbe7007abc : FAILURE in 14m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ca41cb1dd7f42c08015bbff021cd9de : FAILURE in 19m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9fe77ab9bbc14a1fa2958ca4fa206826 : FAILURE in 11m 55s","accounts_in_message":[],"_revision_number":5},{"id":"8c503f50c8ab2e51a9bfaf0e48a1e86594d8c696","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-22 17:06:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d879dab433246a98551e32376ff23c3a0f1c41f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 17:44:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689244/5/openstack-check/tripleo-build-containers-rhel-8-master/44f20df/ : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":5},{"id":"74430ec5e4e9d83a910f8d2916fe60afc8617df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 18:39:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f07d18550cc24a59a59c032f0cda5464 : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3ed9bd58c28741839bc884404f9008c5 : SUCCESS in 1h 12m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/03bf525127114209b80f32ed6d6525b6 : SUCCESS in 1h 15m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a382a4d30e75405fadaddf4201ca7880 : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7fe21ea7fd104fdbb435390fca7c469b : SUCCESS in 1h 30m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ccd6b06275d249dfba5d96b98818a154 : SUCCESS in 1h 04m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10955617842f43ff9483e78afa293bd0 : SUCCESS in 26m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a9ed7aea20af4aeebebc3575f6e0963f : SUCCESS in 16m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f64bd8b50af7484092a23ede0c651106 : SUCCESS in 57m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a4fc4373cb4525b88affa2d81f8f8a : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6b58c9eb21493b8a83afe83a70aecb : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2ce94126f64117973e543814d5839f : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/697ac42c484844fdb91437fa4017c55b : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d23767d9db437dbfcc6d7357bf47f4 : SUCCESS in 7m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5debc027b20f453eb56cbba968dc19a3 : SUCCESS in 26m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbbd95a3e53945299689030454376630 : SUCCESS in 34m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c187d193ba7149c9a102dcfcd4ed5f70 : SUCCESS in 1h 00m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/930db1932eaa4e16955084e338af1a56 : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d5916bf098154adb9fbfe04b6637a71f : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":5},{"id":"35dfee3e6378ad6bcd27cee567bb664fa14dfcfa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-22 19:17:42.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a3398a4cf2d8f32756fd071e468d682b08b294b4","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-11-23 09:22:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"783d05c1852a1cc9b6a4b79e0dda603c52d40906","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-25 08:36:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"cccc18f4f8f5db50e1ee2198f90909eddf36762f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:36:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"293d4716f4ffedb2b5ea99ea2746b8fda94b08da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 10:30:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fc449c05f4a04477b257edc6533f1241 : SUCCESS in 1h 10m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/379b26341f6c4c49a3cd037fa975cfd1 : SUCCESS in 1h 24m 46s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c253a73b4d66483784c95a9990744ac4 : SUCCESS in 1h 36m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c9246435b8f54fa585a6622e872f4639 : SUCCESS in 50m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d930d6bf8efc42478b1b925bdab4caa6 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6285fbe2d9944a0bbbad0118d15a904f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d25d6d6b39ea4652a201d04a356c7095 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dc988fdbd3741558c891d5984ef3c15 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24c92a63edc48b19cf96e1b061b4f7c : SUCCESS in 5m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1843939cbc84ddb8e0eb54545ce802f : SUCCESS in 32m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bf648c88f924ef0abd0d0b0e542118c : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d3f2ef8d68a4b69b934dce3e8f40bf1 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":5},{"id":"785a084834279fd005e614a9dc5ba79e6c8a9174","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 10:30:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3a91fdd5916f35c05ea56e81e3b18a336c081ac2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 10:32:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2a721f2ba347ed8b56a643bc9b75ac : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":5},{"id":"de9034689ac978f1459fef7205e13d7f767da028","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-25 10:54:00.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b421830c763317d33ff1454701ac7d68bb4321dd","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"89372f0c5f312c222fe769a2aab8d1fce445dd84","revisions":{"22129f1f40b17c7283289aae230f288202dd6a98":{"kind":"REWORK","_number":1,"created":"2019-10-17 18:59:31.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/44/689244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/689244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/689244/1"}}},"commit":{"parents":[{"commit":"fd5bff9cf1d9eac371cf481d2888771666eba359","subject":"Fix source link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd5bff9cf1d9eac371cf481d2888771666eba359"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 13:18:14.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 18:55:45.000000000","tz":120},"subject":"Neutron: apply db migration for BGP dynamic routing agent","message":"Neutron: apply db migration for BGP dynamic routing agent\n\nChange-Id: Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22129f1f40b17c7283289aae230f288202dd6a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22129f1f40b17c7283289aae230f288202dd6a98"}]},"branch":"refs/heads/master"},"c2867dcf125fd44d8e7ba94acb448a0c96ce1ca7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-17 19:38:45.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/44/689244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/689244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/689244/2"}}},"commit":{"parents":[{"commit":"fd5bff9cf1d9eac371cf481d2888771666eba359","subject":"Fix source link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd5bff9cf1d9eac371cf481d2888771666eba359"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 13:18:14.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 19:36:10.000000000","tz":120},"subject":"Neutron: apply db migration for BGP dynamic routing agent","message":"Neutron: apply db migration for BGP dynamic routing agent\n\nChange-Id: Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6\nCloses-Bug: 1848572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2867dcf125fd44d8e7ba94acb448a0c96ce1ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2867dcf125fd44d8e7ba94acb448a0c96ce1ca7"}]},"branch":"refs/heads/master"},"a682123ddf0223e16247a25d4f2a164fe828a737":{"kind":"REWORK","_number":3,"created":"2019-10-30 16:41:59.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/44/689244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/689244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/689244/3"}}},"commit":{"parents":[{"commit":"fd5bff9cf1d9eac371cf481d2888771666eba359","subject":"Fix source link.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd5bff9cf1d9eac371cf481d2888771666eba359"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 13:18:14.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-30 16:41:30.000000000","tz":60},"subject":"Neutron: apply db migration for BGP dynamic routing agent","message":"Neutron: apply db migration for BGP dynamic routing agent\n\nChange-Id: Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6\nCloses-Bug: 1848572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a682123ddf0223e16247a25d4f2a164fe828a737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a682123ddf0223e16247a25d4f2a164fe828a737"}]},"branch":"refs/heads/master"},"e9073abaabaac4df148c1e8dd571f1e6a01ca728":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-30 16:42:25.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/44/689244/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/689244/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/689244/4"}}},"commit":{"parents":[{"commit":"ff40bc6e05082524057188c9757058ced5c21a37","subject":"Merge \"base: use en_US.UTF-8 locale on all distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff40bc6e05082524057188c9757058ced5c21a37"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 13:18:14.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-30 16:42:09.000000000","tz":60},"subject":"Neutron: apply db migration for BGP dynamic routing agent","message":"Neutron: apply db migration for BGP dynamic routing agent\n\nChange-Id: Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6\nCloses-Bug: 1848572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9073abaabaac4df148c1e8dd571f1e6a01ca728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9073abaabaac4df148c1e8dd571f1e6a01ca728"}]},"branch":"refs/heads/master"},"89372f0c5f312c222fe769a2aab8d1fce445dd84":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-19 21:51:57.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/44/689244/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/689244/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689244/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/689244/5"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 13:18:14.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-19 21:51:42.000000000","tz":60},"subject":"Neutron: apply db migration for BGP dynamic routing agent","message":"Neutron: apply db migration for BGP dynamic routing agent\n\nChange-Id: Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6\nCloses-Bug: 1848572\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89372f0c5f312c222fe769a2aab8d1fce445dd84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89372f0c5f312c222fe769a2aab8d1fce445dd84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
