)]}'
{"id":"openstack%2Fkolla~689437","triplet_id":"openstack%2Fkolla~master~Iccaf4c0bb79e7e33072e37aa83711b37938d6719","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iccaf4c0bb79e7e33072e37aa83711b37938d6719","subject":"Bump Openstack Exporter version to 0.6.0","status":"MERGED","created":"2019-10-18 12:21:24.000000000","updated":"2019-10-21 06:54:19.000000000","submitted":"2019-10-21 06:53:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689437-1571640783776-f62c7b6a","meta_rev_id":"2ede1f77dfb79a8c2d0e96ae1e3f001f9881c592","_number":689437,"virtual_id_number":689437,"owner":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 06:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-18 13:03:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":1,"date":"2019-10-18 12:39:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":1,"date":"2019-10-21 05:10:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-10-21 05:29:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-19 17:49:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-10-21 05:29:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 12:39:55.000000000","updated_by":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"reviewer":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"state":"REVIEWER"},{"updated":"2019-10-18 13:03:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-19 17:49:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-21 05:10:15.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-21 05:29:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-21 06:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5685b2d6fbbb549bde786231ce021f8eb24399c","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-10-18 12:21:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce1b571d9eee82d42f9841dfb910b1baddb693ad","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-10-18 12:39:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bfa52444eef439a6ddef4b68d727692ea7c567be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-18 13:03:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/37/689437/1/openstack-check/tripleo-build-containers-rhel-8-master/90c45a0/ : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5cd81a00009486c4303ead1773dd55b7b3eea15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 14:07:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/097d330d09f14d4eaa5aeebc9f765275 : SUCCESS in 45m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7dae27a05d0448808ce749df018bd795 : SUCCESS in 53m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/29175d2e118d445b8c4c93e389089cb3 : SUCCESS in 1h 32m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/13fbe385119446a9963e5d8260b359c3 : SUCCESS in 1h 25m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8a739195302744388cfcbaa8a036172a : SUCCESS in 1h 39m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e38c5b30e09140f7a4e9d010d89b1579 : SUCCESS in 59m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3d1fb13704f34ceda2708f8122fbc4b0 : SUCCESS in 35m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/94a8e2b1dc6d4a37b53f9f5cd5c25ba5 : SUCCESS in 17m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e84bbbbc62f345329ac5ce4a2ca7a5a0 : FAILURE in 19m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84992194f7a0430aa029997bccb54e1c : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3989cf3bb64941cfbb194d7a525bd7a8 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4cd2ffb198c4ae7b5ffdb83c7379df0 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86049b036f6a4fbd816779668b4e92f8 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79a845c53fb543b2b85e19fc9ec5b67d : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85487330335441db9248cb2058837cd1 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e4333e373684dc88a6a4b731a7aec9c : FAILURE in 27m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcac46800995459abc8c0b0b60ccdaed : SUCCESS in 31m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76ad91e394044e6e895eeabd69b8348c : SUCCESS in 58m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/96129668a68b4e16b1c975fb6c3d12a3 : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9cd4ad7e71e54fabb13b75335bc34b80 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f8ca6a1a817a8cf0f13294875eb46193e7fc6fe9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-19 17:49:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"404d3902b9af1a2e1a950e7e2d50b5e3faddfbb3","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-10-20 20:59:41.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8b66618a32a9347e0ad434a92d8e6e62307bc761","accounts_in_message":[],"_revision_number":1},{"id":"04f2c405e54fb7db55e7779a6af9b31a2b95c3a0","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-21 05:10:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"993a353627f5d94b602ac5a119b2cffbeac848d7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-21 05:29:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51fff07ef74703b67c823f6d9da5ed72797dc7f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 05:29:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da5f67fd7a515e958229054a239c62003f94d89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 06:53:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8b0deb7562b4797af35f35e428c79db : SUCCESS in 48m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5b6d841360ca4b9f9a820344948cb1d8 : SUCCESS in 1h 08m 43s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b3215a5a211e4d799c3afcda8c96070d : SUCCESS in 1h 12m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b90467392c0b416c940dcdf3f59751d7 : SUCCESS in 41m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c7f770db0484f2f847bb5aca3c75a29 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1160db8b7ef487cbeb25ab591fdeb60 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dfdb4bad0d84727890b228fcee05f46 : SUCCESS in 2m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a272a22cbd8e4204bbbf9c60cb29c7f1 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f9cd4420f4d404b973ba90155e8b643 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13a5b62110a42b78597aa89d28678f0 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a7f8945a06d4e3a934291619f46e47d : SUCCESS in 26m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bc3d5f09e2f4c87bef9c9c0537da3c3 : SUCCESS in 58m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8d91d1aa7d948e39de721332d86407b : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5ac3b83bd8d15c40304b239215252191a125c213","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 06:53:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2ede1f77dfb79a8c2d0e96ae1e3f001f9881c592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 06:54:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7febd5b8adc4ace9e693504da060e24 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd7f78e8f87b194633b60bcce89852f701a77c90","revisions":{"cd7f78e8f87b194633b60bcce89852f701a77c90":{"kind":"REWORK","_number":1,"created":"2019-10-18 12:21:24.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/37/689437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/689437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/689437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/689437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/689437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/689437/1"}}},"commit":{"parents":[{"commit":"eee5c08484e42105070cfca73b5647673d6a2092","subject":"Merge \"Tidy up release notes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eee5c08484e42105070cfca73b5647673d6a2092"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-18 12:19:39.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-18 12:19:39.000000000","tz":-240},"subject":"Bump Openstack Exporter version to 0.6.0","message":"Bump Openstack Exporter version to 0.6.0\n\nThere appears to be a bug in the prometheus openstack exporter versions\n\u003c\u003d 0.2.1 causing cinder data to stop being collected after about 24\nhours.\n\nBumping to latest stable version 0.6.0.\n\nChange-Id: Iccaf4c0bb79e7e33072e37aa83711b37938d6719\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd7f78e8f87b194633b60bcce89852f701a77c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd7f78e8f87b194633b60bcce89852f701a77c90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
