)]}'
{"id":"openstack%2Fkolla~689659","triplet_id":"openstack%2Fkolla~stable%2Fstein~Iccaf4c0bb79e7e33072e37aa83711b37938d6719","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Iccaf4c0bb79e7e33072e37aa83711b37938d6719","subject":"Bump Openstack Exporter version to 0.6.0","status":"MERGED","created":"2019-10-20 20:59:41.000000000","updated":"2019-10-21 17:10:05.000000000","submitted":"2019-10-21 17:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689659-1571677640176-9904e72e","meta_rev_id":"524021e9e649c9ac2fa8d2194999f2e92db6ffde","_number":689659,"virtual_id_number":689659,"owner":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 17:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":2,"date":"2019-10-21 16:04:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-21 15:21:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":1,"date":"2019-10-21 16:04:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 06:44:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-21 15:21:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-21 16:04:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-21 17:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7663c86381e68f2bf742c129930f5fbfa698ccb8","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-10-20 20:59:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f0896eb3d7aca9006dd09501c98b6640d2b9cf27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-20 22:46:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dfe7761d30e14ef18635c7178da5a87c : SUCCESS in 44m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0487b111c72343f78f4ef43101bbe4af : SUCCESS in 39m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fde2671614f846beb8434e2c0b337df1 : SUCCESS in 1h 11m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3e35ab9b93e444cd906be3dd4c355f70 : SUCCESS in 1h 13m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fe5f8564e90d492f9648a77550737408 : SUCCESS in 1h 45m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/eceba95f73404b3fbde27ad234759e35 : SUCCESS in 41m 28s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/df4783c9240f4961828407054ec6c1b5 : SUCCESS in 41m 20s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/19c0ae7be11642a6995bdfafda0270b7 : SUCCESS in 49m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcd2e007e55843ae94cb56e57c2c9337 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1a5694a72146b4ad8c33c494e38e90 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e77c2e638cfc4d6bae0c04c7ad573eae : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abfde84073f34cf1b1112c7aa0052df9 : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72d6d7ec03ac46f687156223f2e42af4 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c242beddec83404d9d95bdb4115b32c2 : SUCCESS in 2m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/142f85949ecf467fa1f9a63eabf74ec6 : SUCCESS in 25m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f9b366f76f6427385fdec03e98908a0 : SUCCESS in 28m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6280652d181f470aa2a3184d4242b9bf : SUCCESS in 52m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1876126dd8be451f97edffb81eaaabcd : SUCCESS in 49m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8185a1e1574948b5ad3c66c416b5108b : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b96be4289f514277982a2f0ead4dbbfc5e7c6501","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 06:43:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"412a9f8a766169d97019f0e56dbf8800e4377ec4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 06:44:02.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\nneed to wait for master merge to get cherry pick","accounts_in_message":[],"_revision_number":2},{"id":"20719a88ab296a6d426b90314b5dd473cd729ccb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 06:54:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc365cf14e7e2e2e08d4b6f428b2d2a7aee9d799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 08:58:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f82bf980ff14cb2b133bbcee09af699 : SUCCESS in 33m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/05de3dc2b85148e39cd36d9d280470be : SUCCESS in 41m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/217b58052b174917a8b58467304fec62 : SUCCESS in 1h 02m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/90535e95dc5045a6bdbf840e704f5d82 : SUCCESS in 1h 21m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aae688f83b184c65abfffcf5a8e895fa : TIMED_OUT in 2h 01m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ba848f4308a340169e8cf2f0c5ceaa4b : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d6874025f0724bb7a3cd25d105823705 : SUCCESS in 45m 02s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2be1989f4f1a44d192f28bd50206317f : SUCCESS in 53m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334957ef29d74505808451c6874b05f3 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcf4c682afd4cc9aa78e8844d9eb2ca : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eecb55b5b0e481880b068e395047a42 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248a499a73084cd6a841fc535d434934 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c35e3531b9894f5488b1258a8be0689c : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da7a588b94145cf8eef16dd8c39c090 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af4025163a35493abeec4eee5f7e56ca : SUCCESS in 43m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5d5432ecb7c4e0e9c7b9150b0df2823 : SUCCESS in 39m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd3dff1805a746abb385903ad7a29cff : SUCCESS in 1h 13m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/813e0abd7b78496a8d86eaf259b3eee4 : SUCCESS in 1h 01m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b80983ebf1fe41a9a46709425b31da81 : SUCCESS in 1h 14m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c0c1890dced2631feeef0c12a578165b7ad86aa2","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-10-21 11:23:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1e231edbaebf7e7113b48bc8df93ee17e562ff19","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-10-21 11:24:00.000000000","message":"Patch Set 3:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e http://docs.openstack.org/infra/manual/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f82bf980ff14cb2b133bbcee09af699\n \u003e : SUCCESS in 33m 34s (non-voting)\n \u003e - kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/05de3dc2b85148e39cd36d9d280470be\n \u003e : SUCCESS in 41m 13s\n \u003e - kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/217b58052b174917a8b58467304fec62\n \u003e : SUCCESS in 1h 02m 07s\n \u003e - kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/90535e95dc5045a6bdbf840e704f5d82\n \u003e : SUCCESS in 1h 21m 34s (non-voting)\n \u003e - kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aae688f83b184c65abfffcf5a8e895fa\n \u003e : TIMED_OUT in 2h 01m 49s\n \u003e - tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ba848f4308a340169e8cf2f0c5ceaa4b\n \u003e : SUCCESS in 1h 01m 42s (non-voting)\n \u003e - kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d6874025f0724bb7a3cd25d105823705\n \u003e : SUCCESS in 45m 02s (non-voting)\n \u003e - kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2be1989f4f1a44d192f28bd50206317f\n \u003e : SUCCESS in 53m 15s\n \u003e - openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334957ef29d74505808451c6874b05f3\n \u003e : SUCCESS in 5m 33s\n \u003e - openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcf4c682afd4cc9aa78e8844d9eb2ca\n \u003e : SUCCESS in 5m 19s\n \u003e - openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eecb55b5b0e481880b068e395047a42\n \u003e : SUCCESS in 5m 07s\n \u003e - openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248a499a73084cd6a841fc535d434934\n \u003e : SUCCESS in 5m 22s\n \u003e - openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c35e3531b9894f5488b1258a8be0689c\n \u003e : SUCCESS in 7m 15s\n \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da7a588b94145cf8eef16dd8c39c090\n \u003e : SUCCESS in 5m 07s\n \u003e - kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af4025163a35493abeec4eee5f7e56ca\n \u003e : SUCCESS in 43m 22s (non-voting)\n \u003e - kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5d5432ecb7c4e0e9c7b9150b0df2823\n \u003e : SUCCESS in 39m 29s\n \u003e - kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd3dff1805a746abb385903ad7a29cff\n \u003e : SUCCESS in 1h 13m 31s\n \u003e - kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/813e0abd7b78496a8d86eaf259b3eee4\n \u003e : SUCCESS in 1h 01m 01s (non-voting)\n \u003e - kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b80983ebf1fe41a9a46709425b31da81\n \u003e : SUCCESS in 1h 14m 19s\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"26f3188b8b1db931bce5c3de728d689235daf6b9","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-10-21 12:44:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eadcb57ff2969822959d4058ad93fe32555b08db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 15:17:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2b4aa16f98694fc18d19a7d83f8570cb : SUCCESS in 56m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/18348f20efdf498fb3dc4ca4d8be3664 : SUCCESS in 1h 15m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/14434d01af4643f8b6d387a7a5921e43 : SUCCESS in 1h 48m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d082276d75364b368a20e63f84780ba5 : SUCCESS in 1h 49m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3719d67f7e744750b02bbeae27843303 : SUCCESS in 2h 27m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/645ffbed58e040f49eeecc55ddeeb4e2 : SUCCESS in 1h 24m 20s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1bee533fb56648df920ddb42ddb9b9d4 : SUCCESS in 1h 13m 00s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c8543b0329834900878faa62f3a05c97 : SUCCESS in 1h 15m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99f4b71562194c25bbf25a7713f3f0c4 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7997f9cf3e741c18ca22565ea3d9d63 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/835804f570094f68b74f55ca26b9fce1 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/056149162ffd44ffac3c1eaa7689cf14 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f4c27deafa848dfa3c483ee747b8d63 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c20119d81f40518240252389fdd2d2 : SUCCESS in 3m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7e5bd5a4521e4bfa95b7a91b560fe272 : SUCCESS in 22m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/946dae0bc3ba4c3fada95987dd9edb9f : SUCCESS in 32m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/737769a0da6b43b3863d0982ed3c5e04 : SUCCESS in 55m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0b34f16946d4df4a75c69d3d310a2c0 : SUCCESS in 50m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d61414dbc7bc4510818ea678e10207c5 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":4},{"id":"adb9fd660813801799b1999144fc8b07d6232dc0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 15:21:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"beb2fdeb34d379f884db260127bcc9892d0e56de","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-21 16:04:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d74ab68c22279b96871b4adfd0ce9adaab34df49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 16:04:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e1ce3f78636a54a672fb159e57207f4c84ff06a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 17:07:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c91b7f3bc9754e50ab3d289857f7dc48 : SUCCESS in 1h 00m 50s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b6fa0fe47c55491ba63f0acaf0ba3865 : SUCCESS in 48m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68c368d550734d18a0828c5581cba668 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0d1936d41c4c609850c409974500b4 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d9074d962d6489db46a6c6cf119a800 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f7492235e04c17a3d32cb88ac414a4 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc96014104734bd59efb95f28a8fdef0 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885d35ac5f284c2baa76522c04ad6972 : SUCCESS in 3m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/290ac7b48c94433989936a28b29830f7 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":4},{"id":"628fa4eaba92e9d63ee9727d1e865ae7205236ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 17:07:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"524021e9e649c9ac2fa8d2194999f2e92db6ffde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 17:10:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771af29fec5a4a919fc84742655499ec : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0f6980d851a92a6b740a590d0c1229384cc763b0","revisions":{"8b66618a32a9347e0ad434a92d8e6e62307bc761":{"kind":"REWORK","_number":1,"created":"2019-10-20 20:59:41.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/59/689659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/689659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/689659/1"}}},"commit":{"parents":[{"commit":"3806c8f373ec9c6a02a82168159fff6c64a019c5","subject":"Merge \"CI: increase both limits to 3h\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3806c8f373ec9c6a02a82168159fff6c64a019c5"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-18 12:19:39.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-20 20:59:41.000000000","tz":0},"subject":"Bump Openstack Exporter version to 0.6.0","message":"Bump Openstack Exporter version to 0.6.0\n\nThere appears to be a bug in the prometheus openstack exporter versions\n\u003c\u003d 0.2.1 causing cinder data to stop being collected after about 24\nhours.\n\nBumping to latest stable version 0.6.0.\n\nChange-Id: Iccaf4c0bb79e7e33072e37aa83711b37938d6719\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b66618a32a9347e0ad434a92d8e6e62307bc761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b66618a32a9347e0ad434a92d8e6e62307bc761"}]},"branch":"refs/heads/stable/stein"},"6ba86dade6d60b8ac8390fa8a8382aa1a019634c":{"kind":"NO_CHANGE","_number":2,"created":"2019-10-21 06:43:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/689659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/689659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/689659/2"}}},"commit":{"parents":[{"commit":"3806c8f373ec9c6a02a82168159fff6c64a019c5","subject":"Merge \"CI: increase both limits to 3h\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3806c8f373ec9c6a02a82168159fff6c64a019c5"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-18 12:19:39.000000000","tz":-240},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-21 06:43:04.000000000","tz":0},"subject":"Bump Openstack Exporter version to 0.6.0","message":"Bump Openstack Exporter version to 0.6.0\n\nThere appears to be a bug in the prometheus openstack exporter versions\n\u003c\u003d 0.2.1 causing cinder data to stop being collected after about 24\nhours.\n\nBumping to latest stable version 0.6.0.\n\nChange-Id: Iccaf4c0bb79e7e33072e37aa83711b37938d6719\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ba86dade6d60b8ac8390fa8a8382aa1a019634c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ba86dade6d60b8ac8390fa8a8382aa1a019634c"}]},"branch":"refs/heads/stable/stein"},"afeeb47745fcd228aecc16b07e1249a1c251ac7c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-21 06:54:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/689659/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/689659/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/689659/3"}}},"commit":{"parents":[{"commit":"3806c8f373ec9c6a02a82168159fff6c64a019c5","subject":"Merge \"CI: increase both limits to 3h\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3806c8f373ec9c6a02a82168159fff6c64a019c5"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-18 12:19:39.000000000","tz":-240},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-21 06:54:40.000000000","tz":0},"subject":"Bump Openstack Exporter version to 0.6.0","message":"Bump Openstack Exporter version to 0.6.0\n\nThere appears to be a bug in the prometheus openstack exporter versions\n\u003c\u003d 0.2.1 causing cinder data to stop being collected after about 24\nhours.\n\nBumping to latest stable version 0.6.0.\n\nChange-Id: Iccaf4c0bb79e7e33072e37aa83711b37938d6719\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n(cherry picked from commit cd7f78e8f87b194633b60bcce89852f701a77c90)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afeeb47745fcd228aecc16b07e1249a1c251ac7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afeeb47745fcd228aecc16b07e1249a1c251ac7c"}]},"branch":"refs/heads/stable/stein"},"0f6980d851a92a6b740a590d0c1229384cc763b0":{"kind":"NO_CHANGE","_number":4,"created":"2019-10-21 12:44:43.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/59/689659/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/689659/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/689659/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/689659/4"}}},"commit":{"parents":[{"commit":"3806c8f373ec9c6a02a82168159fff6c64a019c5","subject":"Merge \"CI: increase both limits to 3h\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3806c8f373ec9c6a02a82168159fff6c64a019c5"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-10-18 12:19:39.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-10-21 12:44:43.000000000","tz":0},"subject":"Bump Openstack Exporter version to 0.6.0","message":"Bump Openstack Exporter version to 0.6.0\n\nThere appears to be a bug in the prometheus openstack exporter versions\n\u003c\u003d 0.2.1 causing cinder data to stop being collected after about 24\nhours.\n\nBumping to latest stable version 0.6.0.\n\nChange-Id: Iccaf4c0bb79e7e33072e37aa83711b37938d6719\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n(cherry picked from commit cd7f78e8f87b194633b60bcce89852f701a77c90)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f6980d851a92a6b740a590d0c1229384cc763b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f6980d851a92a6b740a590d0c1229384cc763b0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
