)]}'
{"id":"openstack%2Fkolla~689706","triplet_id":"openstack%2Fkolla~master~I396f929ecc11f87fb83585bedd39be93a605c57b","project":"openstack/kolla","branch":"master","topic":"feature/rdo_train","hashtags":[],"change_id":"I396f929ecc11f87fb83585bedd39be93a605c57b","subject":"Move to RDO Train packages","status":"MERGED","created":"2019-10-21 10:03:01.000000000","updated":"2019-11-13 12:08:49.000000000","submitted":"2019-10-26 13:24:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689706-1572096259142-df6c426c","meta_rev_id":"f5cad4ec4f5a428a3b401638c98ea19eeffaafb1","_number":689706,"virtual_id_number":689706,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-23 02:43:54.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-26 13:24:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-26 07:43:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-23 13:28:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-23 13:13:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2019-10-26 07:04:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-26 11:26:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 02:43:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-10-23 13:13:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-23 13:28:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-26 07:43:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-26 11:26:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-26 13:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22143bf909d93a79515ef5b87bffd17e7dd4d210","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-21 10:03:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4227f6072998ae14b7867ef78cd5da6ec805f69b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 10:05:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26a502d9639b7fb5704001a3c80aa6606c88793d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 10:42:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/06/689706/1/openstack-check/tripleo-build-containers-rhel-8-master/de0c718/ : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0d42cab8e48d98466b97180191b89da5d43e6c77","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-21 10:49:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dee0600f070065e33121c1651f68f84c1a669101","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-21 11:09:27.000000000","message":"Patch Set 2:\n\nno centos-release-openstack-train in extras yet","accounts_in_message":[],"_revision_number":2},{"id":"4e4c0f9e007d4454d85fa5403d3c2899e41d1304","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 11:29:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/06/689706/2/openstack-check/tripleo-build-containers-rhel-8-master/ed51240/ : SUCCESS in 39m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f445ee7cbce7aee8ea4eacef3f66eae35a6beabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:48:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9e417e72d813436db7178b0b9e632010 : FAILURE in 6m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5553788f5a85423fa81b79f252b17bd3 : FAILURE in 6m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/282968652e064578b9a31d75883fc30c : FAILURE in 15m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c285e3b6b8844ad4b4273bd7bb8a56aa : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c438d477c0a541fbbefc602ce8462780 : FAILURE in 11m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/73a13419d0cf4a8b98e0d0910b513b78 : FAILURE in 19m 37s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/e3119544c0304075b50e0ce50bab4a8b : FAILURE in 14m 37s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/e0445569789941fa99ffc0058cb99f05 : FAILURE in 14m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9c5c9247590647aaa8e2fdc2e6cf2c4c : FAILURE in 14m 19s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/76669adbf5684ad8b0c981e54c123e00 : FAILURE in 12m 17s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d1dbdf3f50ce4f59b3d23b9000e67cda : FAILURE in 14m 11s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/a727d59bdc7d4a95ae6ed7b57cb58dfd : FAILURE in 16m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5b97e59178804972af830b1b183ab789 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/21c54f68cf894f0b940070484c6b04b1 : SUCCESS in 16m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4cebe74196a64929961342f79de174d6 : SUCCESS in 1h 01m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87df2ba0433440e88e117638b5aec3ca : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5754a58ad34a46e78d3ae2a275ec3b32 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8a10a09cf164af4958263d2fa628576 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7c1301ed56f4bdf919e8b451d719a31 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8f957ced1254e37a0b7c0f6814b98bd : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef9d8242c0b4d73ab7fa43e99469de9 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76c0c684291642bfa16a11e153a85e7c : FAILURE in 22m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/38c9a0966a1348dbba16161db058fefa : SUCCESS in 33m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2afc8f8db96a4f04bfddc6bbd58c53f5 : SUCCESS in 1h 07m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/179eec464f35481780cde955deb7e4b0 : SUCCESS in 54m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28ecb91c0c6947f1b26db4a4aed276c5 : SUCCESS in 1h 10m 28s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a13fd36008af4c9e9dcc221636f9d6bf : SUCCESS in 1h 17m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/f0e1816b3f1946feb71f51528180aeb3 : SUCCESS in 1h 30m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9213572bbd624219ac780486c212a956 : SUCCESS in 1h 55m 49s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/6e113a9ab7a3447c9073aea761d0a8ff : SUCCESS in 1h 11m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7cc1f84d0893454cace6653d4709aaa3 : SUCCESS in 59m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f757703eecc986d3a8408b47114c466bacfd8c96","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-22 07:09:11.000000000","message":"Patch Set 2:\n\nrecheck\n\nready now?","accounts_in_message":[],"_revision_number":2},{"id":"9d22822be4cf7cc0f753345a74868e328b33ead5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-22 07:49:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/06/689706/2/openstack-check/tripleo-build-containers-rhel-8-master/b652e2c/ : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a75a8da01afe41500ab215785df059254f59ecd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:14:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/81e3932b50d848c9bb778e4ca15aea81 : FAILURE in 10m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/703767879d1f41f7ab8e90d719cb94d6 : FAILURE in 11m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/af8700d35a1748ccb087fe45146a4391 : FAILURE in 16m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5f4a49ac975944aeb8c20fb9b591a239 : FAILURE in 16m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/02b5ad2b4a544238be053c85770db479 : FAILURE in 14m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/097823645d3e4ff6884b3e5c332d11ec : FAILURE in 17m 48s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/7e0897de72ab4d03bd220572fb65d404 : FAILURE in 13m 21s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a03ad44ebad04ba4bd13eddb3f659a62 : FAILURE in 16m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2cc2cd922cf246299689e1aa62b73506 : FAILURE in 15m 42s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3d3e65f0662148619216e96b1bd2448d : FAILURE in 15m 31s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d6f6d7046c7e4e47ab4d5ea2aaede1b5 : FAILURE in 12m 45s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3c90c25edcf24f34bac6d26b34afc139 : FAILURE in 11m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a2b4a8a456d347d4af67382d7e58ca9c : SUCCESS in 26m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/09b71a26e487403b8b8c4793d64ad800 : SUCCESS in 19m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ee90d39d28084aad88e65700be35fb5f : SUCCESS in 51m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b441f7935a34150ae8655f1fd87e65d : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74353fb7bc064d71b0886de4c1e53d5d : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81832263cf5e43c6a518993eca0cc6b2 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cee8a3c550a4b37bf8c5542e2803e90 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32bd9595c3ec430f96c35d53b6df4ae5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689d174f0f1d455e9bf2adfca5a90963 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85e2ed38c21c4d2c99d67237ccc4dc8e : FAILURE in 29m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c732d57151c4b68ad0d894f40363c5c : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/daddbf2ea372415996e1b603ae5554a0 : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e53e8883a7454ef38c03b787a43153ec : SUCCESS in 57m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/65dc754a95114ff98cfc4d1f8cbabed6 : SUCCESS in 58m 51s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/3a6c322ca25242048d73f2b60af3a6ed : SUCCESS in 1h 29m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c8b47bf113da4d42b0598598d9defef6 : SUCCESS in 1h 32m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9e94a990b78f4043914f03e3f1c571d7 : SUCCESS in 1h 55m 03s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d156a8eddfa448cfb7878a1574826d48 : SUCCESS in 1h 15m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/665d858838b94532847d56cf1fa9c1a4 : FAILURE in 1h 39m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"deb74ad92a69d483d0413ad9dca961cf60987d8e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-22 09:15:02.000000000","message":"Patch Set 2: Workflow-1\n\nNope, they plan to release in a couple of days, waiting.","accounts_in_message":[],"_revision_number":2},{"id":"0e516552fede6e2b0c291e41098047c8d8c8cbab","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-10-23 02:43:54.000000000","message":"Patch Set 2:\n\n\u003e Nope, they plan to release in a couple of days, waiting.\n\nIt should be available soon(release rpm already exist at:- http://mirror.centos.org/centos/7/extras/x86_64/Packages/centos-release-openstack-train-1-1.el7.centos.noarch.rpm), missing TripleO RC packages should be available in today\u0027s CentOS mirror sync(they are already there in buildlogs testing repo), we shall be discussing it in today\u0027s RDO meeting about it, stay tuned.","accounts_in_message":[],"_revision_number":2},{"id":"59ba983474639d192d44802462ba09ced2cf5112","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-23 08:26:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6b07b04f0fafab4ecf6205e16a27078eb7fb5f4f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-23 09:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/06/689706/2/openstack-check/tripleo-build-containers-rhel-8-master/8c23b1d/ : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ba25cdf0846bb0167961cea8fac7f1cd321f7f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:18:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ec8e07a6adb54766b64d653242782c84 : FAILURE in 49m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/25d64466207046998225ee962becbcb4 : SUCCESS in 57m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6938458447fb42ef86faa90fb902f9d4 : SUCCESS in 1h 29m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2d5b0e542cbb48598e6e6a9474ae8897 : SUCCESS in 1h 09m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4cd0872467fc4307864f05b0ac106f23 : SUCCESS in 1h 23m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e9e4e5a04ad24cc4b917bbd31be0ecd3 : SUCCESS in 46m 34s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/14f6880ecfe544b89c0286285370cc7e : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/db92d66cf27b4721aa2d2cc78f9d814f : SUCCESS in 2h 09m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2a9d846b42a8431992900b8633ff7679 : SUCCESS in 2h 18m 41s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d7ac3cd33491451faf43e01b81fd3f09 : SUCCESS in 1h 51m 14s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1f7c5d19524f406792ba2298baf6e4f4 : TIMED_OUT in 2h 02m 25s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/6ce963f124814fc09219e70053c1742c : SUCCESS in 1h 32m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ace4a704406947bfabc9aabdc341b311 : SUCCESS in 27m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7a39bef8301e492495aaff1dd761c64a : SUCCESS in 20m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2f441ca341be4d14839cbbeb9cd50a5c : SUCCESS in 55m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6b08e811cb148ab858f193e281affa1 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8c9a4e6c3d4485ac959c2872fec71a : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1b47f4a24844772a37cb5eea751de3e : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17587013a05640048255cbcdb47b8537 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/272f3b6b7afe4c699c606c9463d66112 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3687586a1fe49aca6a8144d56ba7ed5 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e4cb3cdfbf14678ad9cad4fa6e417d1 : FAILURE in 24m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1693867b5f345539033af5fbbdfa818 : SUCCESS in 42m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5eb5169b30354000be086cdaf54a5887 : SUCCESS in 57m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b68e9ed0ac82402da2ace5816c46a459 : SUCCESS in 57m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ceb6b0c1e86475f8b35704275eacf0f : SUCCESS in 1h 01m 11s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c7fb8e9121cb455498cbf5fd402ff235 : SUCCESS in 1h 16m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/b2a56893fc57463d836cdcaebef899ea : SUCCESS in 1h 32m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/45f59ad53be34a58b23689c03ae9128f : SUCCESS in 1h 51m 32s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/47d654c561c7415d96755ce1e5d76c1a : SUCCESS in 1h 25m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1f7fc175dbba49ac913d836bf68bdda2 : SUCCESS in 1h 07m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c33f80899e81d1d6d1ed8af5d037094913f5708","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-23 11:37:38.000000000","message":"Patch Set 2: -Workflow\n\nSeems we\u0027re waiting for tripleo now.","accounts_in_message":[],"_revision_number":2},{"id":"188daa8754e37dac369a5123ff41350e1fe2a968","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 13:13:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad3a3208fb718f375ec9ca8fafa2dbdae3817673","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-23 13:28:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"24ad4571ff69b56fef27b4c9d413fb2ad4467954","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-24 04:44:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04e391ebbf600367a4b340e74164a9c450a328d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-24 05:24:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/06/689706/2/openstack-check/tripleo-build-containers-rhel-8-master/1735d86/ : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0da25adf88a46d33e6007afd3e7423bb8ca64402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 06:51:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/96f769d8a6dd4b9c801e52fe98a18e5b : FAILURE in 40m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c20611a5bab046db9cc4a8ee27dddf3d : SUCCESS in 47m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8fe67c02f3774301bb0d1b5b8badc791 : SUCCESS in 1h 25m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e68c0460f8a64b52b803b7d4a9d7cda3 : SUCCESS in 1h 14m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b3905adae5554eb6a78a61d1a3f9df16 : SUCCESS in 1h 23m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fa23d24aa0a2406ca0bfe3518936a35b : SUCCESS in 57m 55s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a3cf5ebf596f44419ec658b9290cea30 : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5cf2b91fc87844d18c23bfd6910881d0 : SUCCESS in 2h 02m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/92694f2f101f4ad297ad22225286778b : FAILURE in 1h 22m 22s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c950aa5fe2c3450b831a995d64506ffd : SUCCESS in 1h 44m 45s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/cb788c977e1d4df49103619268b76716 : FAILURE in 1h 39m 40s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/bf093602e8c84da7b5838ca7e9605f5c : SUCCESS in 1h 27m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/749ccecedb924391bd0046a94000c734 : SUCCESS in 28m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/362e6b42b0334ef3804684ae57b1300c : SUCCESS in 21m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0db982fca7ec4614889d7b0e30223b5b : SUCCESS in 52m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d60b433eee68438691a5a6e2f7c8be3c : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b697dd23b60f456084b6100d754f6aed : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b14463dcf2eb45fbbe7dea51bdb21466 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97cacd1a320b4a8c8abc0e0cde8c8b82 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00f0b25be80343769e17848e3bdd4319 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5325fc84f8ce4b37924cfdedef4f4460 : SUCCESS in 7m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45e3ae7881a64b77be1f383c90cf9269 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8e0c40eb31a4ac8a694fa78df3514bc : SUCCESS in 29m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c7006b294394b198caac14c4c8d674a : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41af3e228c7247d48f6eb4ab7d683d81 : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c571502803148a88bb0e50e6da840ae : SUCCESS in 1h 11m 18s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ef043e7dc571471c89159f4c2b7464d5 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ecdaf8ba4fdb4ab78bced849d6a297bf : SUCCESS in 1h 32m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8964e7cfe25c4c449976a56453d21b53 : SUCCESS in 1h 41m 59s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ec261c66e182461694fb90e0e83b2498 : SUCCESS in 1h 16m 00s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7a1cde1497bc4c4f88b061ffa90670e2 : SUCCESS in 59m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b1bc4466b5be3992b7b311a33cbeafa2079691b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-26 07:04:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dff23ea3df6cd68aab4e58f9b2829ecad166052d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-26 07:43:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/06/689706/2/openstack-check/tripleo-build-containers-rhel-8-master/43c8614/ : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e9a5119b80b1325aedd67c98878d0c8af7a98fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 09:14:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e6086b36d07a432080d3c0dac89d8d0d : SUCCESS in 40m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b65e496851ee4e97a10f13847c8d599a : SUCCESS in 44m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7d9840c8303e43d2ab70708e21d7d90d : SUCCESS in 1h 20m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0aa1767179a3426288a79da57ad6cfdc : SUCCESS in 1h 17m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ed84d95961f248b98f3a0f209cbdf26a : SUCCESS in 1h 23m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d6fee95e159446c3ac8fce9d6d2e71d1 : SUCCESS in 40m 43s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/f56f820a43134731ac86e6c84aab0850 : SUCCESS in 44m 18s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/82355db4194d472386663f828e9362dc : SUCCESS in 1h 57m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1ba298d1ff954d07bbabe815a8d209d8 : SUCCESS in 1h 58m 43s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/a912ff47406d489b914267eda44e5fee : SUCCESS in 1h 58m 30s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8e7ecab9f08b4b22a87aca6d70b1b5af : SUCCESS in 1h 48m 56s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9fc112becbd742499ba1f1e8965a1392 : SUCCESS in 1h 45m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dd4417509a104ee9a872f37f2493d030 : SUCCESS in 41m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/81d473cf512d4e9fa0da9c29ff2a4dd6 : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1be5b472da464004b98b99ce91bb3f24 : SUCCESS in 51m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fe22c2ac24944cdbd5b080f536df04a : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86121f7aa81b43a7a9adde10fe7ad76b : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fd60f4b22534636940bc6a45af2aad2 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cb8d9ffcf1b4bcc886ad9f0ce41e7c5 : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4c6176fe3584139a3736646e6a1a355 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6552696cdf401c86546654bdf8514f : SUCCESS in 8m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de3671b325e2488e8f7f380fb1034b0e : SUCCESS in 50m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bf2d5298e77431094f31924159cf31d : SUCCESS in 27m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/feb4ecaf40144e25b379fefe47753631 : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/400f1e7272394527a8f43a6b9b1bab21 : SUCCESS in 51m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/def11b6d9ec5481d8939b5e7ba51b6fd : SUCCESS in 55m 13s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7336b69791b34baf819591278c10711c : SUCCESS in 1h 11m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/806e6cc407904defb905a6182029fcc9 : SUCCESS in 1h 23m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f436fff69ffa4b86aafbf4a4f108ef49 : FAILURE in 1h 34m 38s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0679a7b0d9a24463955d43df5b8b89be : SUCCESS in 1h 26m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cfb4eedc3a54405e9d38c31a16872e93 : SUCCESS in 1h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6793224b702288b59302ad9b4587629dbf16471d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-26 11:26:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0597f12db88092cc6b216c595fc2fb5dde555be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 11:27:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36d0ab1483b237e963e899b92e88f4ddf4b8cb16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 13:24:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6280e8bcd94646b5aed90c71b6a2f330 : SUCCESS in 43m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/731cf6f8f69445fa8ab2f8adc79bbe42 : SUCCESS in 1h 14m 43s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fa5633fe1ce74d27a8b0502c10161d35 : SUCCESS in 1h 46m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/79d493ea325b4a8e98964872482f8d51 : SUCCESS in 41m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/220deb78b43b44a4b034e02fd0313414 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa3e85663d94289bbb65701eb4456ea : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad74db2214a44f33b1066d20ce3b7351 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a8dccc7c6644290b06d2338b375c78b : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ce97bdfed514611a155cc6a4726fb90 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77671a0a78104b18b7cad4725c3c4340 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/959aee505a9d4bd9b483825abfedcc4b : SUCCESS in 38m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a66ec83004034ee2a788f0e39aa0629c : SUCCESS in 58m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/71b5a36cfaa343b89c670c05b31e34cf : SUCCESS in 1h 44m 18s","accounts_in_message":[],"_revision_number":2},{"id":"87e9a0fc165ebe8426ecb67060bffb370c3048c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 13:24:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7abf461b9b41c12c19c71d93754053bd3c7c39d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 13:25:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891bcf1165ae4565be949d1884416b9d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2},{"id":"0c053fc5b3a120cf0698578089b529d64e08ad5d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-10-28 15:42:11.000000000","message":"Patch Set 2:\n\nis this just preparation for train branching or you plan to keep using official train repo for master branch?","accounts_in_message":[],"_revision_number":2},{"id":"ae1b46c1c64d3a750085504b2b0e0a4463e80378","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-28 15:44:27.000000000","message":"Patch Set 2:\n\n\u003e is this just preparation for train branching or you plan to keep\n \u003e using official train repo for master branch?\n\nWe have not branched yet.","accounts_in_message":[],"_revision_number":2},{"id":"0382a1d812e45aa9830ccca0b1dadcb1881613ec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 16:17:03.000000000","message":"Patch Set 2:\n\nAlfredo, preparation indeed. ;-)","accounts_in_message":[],"_revision_number":2},{"id":"848f45d80316a48188b889c996b9382b5560f0c6","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-10-28 16:19:29.000000000","message":"Patch Set 2:\n\n\u003e Alfredo, preparation indeed. ;-)\n\nMakes sense, thanks for the info!","accounts_in_message":[],"_revision_number":2},{"id":"356920591ca2c1edad0244b1c856be95a082f7fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:08:49.000000000","message":"Created a revert of this change as I61c68bf738673a96dd5ba256e40e8532364ff281","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c78c82c9630b334cbaef040d68bce7e5ae8cc118","revisions":{"8ce0c332edfe88c2da3c10e2a71616e2129193e5":{"kind":"REWORK","_number":1,"created":"2019-10-21 10:03:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/689706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/689706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/689706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/689706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/689706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/689706/1"}}},"commit":{"parents":[{"commit":"3887db89af16822404c37e0ceabb76bb9614722a","subject":"Merge \"Bump Openstack Exporter version to 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3887db89af16822404c37e0ceabb76bb9614722a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-21 10:02:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-21 10:02:59.000000000","tz":120},"subject":"Move to RDO Train packages","message":"Move to RDO Train packages\n\nChange-Id: I396f929ecc11f87fb83585bedd39be93a605c57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ce0c332edfe88c2da3c10e2a71616e2129193e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ce0c332edfe88c2da3c10e2a71616e2129193e5"}]},"branch":"refs/heads/master"},"c78c82c9630b334cbaef040d68bce7e5ae8cc118":{"kind":"REWORK","_number":2,"created":"2019-10-21 10:49:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/689706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/689706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/689706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/689706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/689706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/689706/2"}}},"commit":{"parents":[{"commit":"3887db89af16822404c37e0ceabb76bb9614722a","subject":"Merge \"Bump Openstack Exporter version to 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3887db89af16822404c37e0ceabb76bb9614722a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-21 10:02:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-21 10:49:31.000000000","tz":120},"subject":"Move to RDO Train packages","message":"Move to RDO Train packages\n\nChange-Id: I396f929ecc11f87fb83585bedd39be93a605c57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c78c82c9630b334cbaef040d68bce7e5ae8cc118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c78c82c9630b334cbaef040d68bce7e5ae8cc118"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
