)]}'
{"id":"openstack%2Fkolla~689744","triplet_id":"openstack%2Fkolla~master~I1902a5039cb9318d7d70387dd2dc142d915fd298","project":"openstack/kolla","branch":"master","topic":"kolla-train-release","hashtags":[],"change_id":"I1902a5039cb9318d7d70387dd2dc142d915fd298","subject":"Switch upper constraints in kolla-toolbox to use releases.o.o","status":"MERGED","created":"2019-10-21 13:35:03.000000000","updated":"2019-10-22 11:16:39.000000000","submitted":"2019-10-22 11:14:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689744-1571742899604-b61b701d","meta_rev_id":"12282be472cae5663a6b4518c989653f27bb3723","_number":689744,"virtual_id_number":689744,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 11:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-22 08:25:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2019-10-22 05:58:14.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-10-22 05:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-10-21 15:27:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-22 07:39:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2019-10-22 05:58:14.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-21 13:37:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 15:27:41.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-22 05:23:29.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-22 05:58:14.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-10-22 07:39:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-22 08:25:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-22 11:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e6049509e2db649763e2ffdfe79e842f09f1186","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 13:35:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c596560dc246f25b9a7f46e4146c13d634b9ebea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 13:37:01.000000000","message":"Topic set to kolla-train-release","accounts_in_message":[],"_revision_number":1},{"id":"297bb93986cf3bfb7a3f385797d1af8325281658","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 13:37:07.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"fd5cc85ec1a56bbe0d56bf50916f93388f679fae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 14:15:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689744/1/openstack-check/tripleo-build-containers-rhel-8-master/c9ebccb/ : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"40d95f7522b0019b12fbc042f1317f7d4a14db84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 15:17:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9baadb7bf8bb4c22b26329696e0caa46 : SUCCESS in 52m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/919406a8ffe449799d9519532bfde867 : SUCCESS in 1h 01m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/36e8bcd275cb4588ba112762f8c27e3a : SUCCESS in 1h 36m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ada55bf30af94a3e8668b24fcbcd75c2 : SUCCESS in 1h 16m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e7a5021a89474a27aa8c5478aacf3f14 : SUCCESS in 1h 22m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07453417b5a94f1a823530337bef4a31 : SUCCESS in 1h 01m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f4ac1e4ab3444d508c17345950e3e388 : SUCCESS in 24m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/15d416c6d238401cbc3003abd1827ec8 : SUCCESS in 18m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e525f88675444d16801751656be56f4e : SUCCESS in 52m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff121ae5a17d4950acae88868e45ef23 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b14085741fc4613a95470c16c9924c3 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9045086504c4929b0d08b4a30642973 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c44a97e2ef6641db8ff4659c60d30614 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39a55672a730428f92a8ad3cf39757de : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8cf3d8b1974ae6be6f7590dd3b157e : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/600e4061847c4982aff2445ce2b7c8f2 : FAILURE in 25m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cc968c3234243959a113005af7ebdfe : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad8a2b825b0c4ccd94b0afa7a9a252ba : SUCCESS in 59m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/032a9dbce6514217beb07316a2096310 : SUCCESS in 1h 02m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70fc1b2aee4a4d6c860c94582e0bc748 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"12ce20f2289858e31981b2d38493ad3c2b4da640","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-21 15:27:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f34d6b207139d259ae624cd70e205c440335ed33","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-22 05:23:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd3c399c50b2c1c897295f1214d7cfb92547ff61","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-22 05:58:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11e8fe3219cc9ed7486feafcf808b9a95f53190d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 05:58:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0911104a59172d8e4f833ae9ee3e994c417d9a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 07:33:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/96fd71336d4f4bfa9ee6a0def0f2994e : SUCCESS in 1h 04m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d4c1f2c07d044baa8dfb2c983f797dad : SUCCESS in 1h 31m 10s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a8901e59c54d486ea8cc06d8862b5876 : FAILURE in 11m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a6290a92abc34a4db4456fb3951e6275 : SUCCESS in 57m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65853f9dd9f348479fba27f69bdd1d63 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/970d3c596f124557b26df31599eb282a : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1a3c8465d984184b823d450227d2541 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf016385f3a2465ba5d172d190d4121b : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8879bdb7a024ed48185048a271e5e19 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a724fe2ea4d14226bd526ffad0a3e7ee : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/81f4d9813394446aa348ce1cedd3121e : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bdf2bdcdd8a42f09b31667b9f9860ce : SUCCESS in 1h 11m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7650b953f804129b2e376fe5a4076b7 : SUCCESS in 1h 27m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4898e84cf4299669ec84363c744884a0b1f4e31f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-22 07:39:41.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9219f05ac1cb25ad170bce8acc7569e99c1c072d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-22 08:25:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/689744/1/openstack-check/tripleo-build-containers-rhel-8-master/7cde905/ : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ae34acb961ff8ccef02dcbe5c552fb26f7af5b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:33:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a286c5affc5e4289b16f41e05dae33c2 : SUCCESS in 50m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ccfbb9876a194f8faf789bbc3184b300 : SUCCESS in 1h 02m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6f73a52ef91e4e4f96d2a4fad8ae22bd : SUCCESS in 1h 41m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1d92f0f5573f427ba4d91d8233fa292d : SUCCESS in 1h 27m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/28d1a2b4d3bf4d849ce9e9f5ec2e1bf2 : SUCCESS in 1h 35m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3fd08bb3411f478589cd95df4495d696 : SUCCESS in 52m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c82e44dff45447d19fd67c85b09fe634 : SUCCESS in 26m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b57522053bd9406aadfb73d0a330e445 : SUCCESS in 18m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/04f8bb07cf0d4f0c9a562a2f4d5efed6 : SUCCESS in 51m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f34b565dcca45f7ae869ee297574ed6 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f83b0c054246d7a196128462e813e1 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03086031c54d43719c819712560500ab : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b63ee70580914cec958ec6d53ecad8a9 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4433329b5cbc4244a272197695f659a2 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a634e20ccae84893b064e6663b2accb9 : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3e748822b1c4f519ffc72807ceca98f : FAILURE in 24m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/06c49c8040bb4fe8a7f793767d598b80 : SUCCESS in 40m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0166f48b820c454a8bbfad48b3560e48 : SUCCESS in 54m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c896e6f5bbc436e9b90ab0711ae6f67 : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8a149a273b034023891f62e9ea5ae03a : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4ea5bb2945401a61f259da6d1d46d50ed24814c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:33:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e08005123c3eb52ed40f77107e20bb8728e7f942","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 11:14:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"675f4953a4d8bf23700417f80b1d87139580c5c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 11:14:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d5eeae2c377648b298a44b9b9a30393e : SUCCESS in 42m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ab4c4a7b0827440c82d2243dd19ed78c : SUCCESS in 1h 28m 37s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8d9682e20f3c4b398b9286adaa8398ca : SUCCESS in 1h 38m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d9bae09267824590b901fbad27af718b : SUCCESS in 59m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c9d30f86f0a4fe4b19631489f1a19ca : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a0fd30f11f4260b906a8197950a346 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68436a4b4846464b914ca4fc36df9b59 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66513573ab294beeb3b6de1123652911 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f63da20604941209afb02e46b6707d9 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713d0165f5cb41878217a6f0e8855d96 : SUCCESS in 10m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e48e8486b7404371ad215fa315491e68 : SUCCESS in 39m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/041ef07324c748a4a104b7452f7cdff0 : SUCCESS in 1h 06m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67ca330add194e7dbb354e847324f215 : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"12282be472cae5663a6b4518c989653f27bb3723","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 11:16:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e858a82999834aa4992dbe7b73e08dd4 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b6027a50ec79840ab4d33abe0158e448db25677","revisions":{"4b6027a50ec79840ab4d33abe0158e448db25677":{"kind":"REWORK","_number":1,"created":"2019-10-21 13:35:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/689744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/689744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/689744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/689744/1"}}},"commit":{"parents":[{"commit":"3887db89af16822404c37e0ceabb76bb9614722a","subject":"Merge \"Bump Openstack Exporter version to 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3887db89af16822404c37e0ceabb76bb9614722a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 13:33:34.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 13:33:34.000000000","tz":60},"subject":"Switch upper constraints in kolla-toolbox to use releases.o.o","message":"Switch upper constraints in kolla-toolbox to use releases.o.o\n\nThis ensures we can use just the release name in OPENSTACK_RELEASE,\nrather than the branch name.\n\nTrivialFix\n\nChange-Id: I1902a5039cb9318d7d70387dd2dc142d915fd298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b6027a50ec79840ab4d33abe0158e448db25677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b6027a50ec79840ab4d33abe0158e448db25677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
