)]}'
{"id":"openstack%2Fkolla~689789","triplet_id":"openstack%2Fkolla~master~I508bb082a319955b3e008235e158eaa07a1e8a74","project":"openstack/kolla","branch":"master","topic":"bug/1849128","hashtags":[],"change_id":"I508bb082a319955b3e008235e158eaa07a1e8a74","subject":"Fix python3 compatibility for barbican","status":"MERGED","created":"2019-10-21 14:29:58.000000000","updated":"2019-10-22 08:07:42.000000000","submitted":"2019-10-22 08:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689789-1571731422443-9e4bfd77","meta_rev_id":"745d4cdffdb8f5017a1d85e545cd7c0f8feed0e7","_number":689789,"virtual_id_number":689789,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 08:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-21 17:29:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2019-10-22 05:58:29.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-10-22 04:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-21 16:16:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2019-10-22 05:58:29.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 15:15:51.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-21 16:16:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-21 17:29:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-22 04:56:30.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-22 05:58:29.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-10-22 08:03:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f92c1acc5e0540397109ab562eab609eaaee92d3","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-10-21 14:29:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b30e62d92abc6fa5e137dd4dcbaa3e52c75ff4f0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-21 15:15:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78dc935329a6ad6d0a45763f0cce615213e029ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 15:16:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a49bfc4d5b47d2abd09d28fd8cfd531223530130","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 15:18:06.000000000","message":"Patch Set 1:\n\nBinary image part should be backported to stein.","accounts_in_message":[],"_revision_number":1},{"id":"ff64e866cc614846081ec9ec5a2bfc4c962571e3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 15:20:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/89/689789/1/openstack-check/tripleo-build-containers-rhel-8-master/9890a1a/ : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9ace708c9c8c1e5a239891f3f72b5a01653c92c5","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-10-21 15:58:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9143c9e8504f156c8f78aad81ea01da4bca710f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 16:16:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6e2fb491c86f9fb660f15adb1693aa5b50d4801","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 17:29:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/89/689789/2/openstack-check/tripleo-build-containers-rhel-8-master/62095e2/ : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e19354620bc5726b960a5282caa55b1d27caea6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 18:13:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0acc40e34f8c4994bacf18c5ddb082a8 : SUCCESS in 50m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2c9aec0c979f4d178cef0b1ffc9cbc7f : SUCCESS in 1h 05m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7ce676c5c7094850973af1f9ac815d70 : SUCCESS in 1h 27m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a0c0fc2cd802460da3fa03bdbf1b4422 : SUCCESS in 1h 26m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/01f83a8c8a644f138647ac12dc80235c : SUCCESS in 1h 37m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c664eb70e3b740128c9000e219d90730 : SUCCESS in 1h 04m 15s\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c3ddfcf62ba849f9b47eb19ae0dd7954 : TIMED_OUT in 2h 02m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29001ee60e0d4ed080d51e1d998a541f : SUCCESS in 33m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3e2dabfa99e449158e0a7ade6494822b : SUCCESS in 23m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/16d511b867244ce3b0e8e7559ad3b241 : SUCCESS in 1h 00m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db4e3546005d4184ac04d646ae61cfc9 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d3bec927c5b4242bffd65731ce73c46 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31d92d3931fb4a51bc091e8411c1dba2 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab86a6050f794239a8c7efc57512d9ca : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcc7de1cd6744caf867954bea1e3065b : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e3eaf67d7545d592fbdb3b67f3652d : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/95cf0fe2b1d144208e62626a58bbb1b4 : FAILURE in 32m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8e3a60624814955b22b05d09dc81a5a : SUCCESS in 32m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c42aa1914aa540ee9deaea22cf104bd9 : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3cbdacda9b4147f1955add9b7a93c230 : SUCCESS in 53m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/071cf88f76574f8ba6dccaf1477be20b : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4f294441001ac14946cbed4bd9f149745649402c","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-22 04:56:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7daabc9d0d1f893f36076fe4672b9d3484894acc","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-22 05:58:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bbc02a311ec54a2e35a0db20d2ecafe8d41b537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 05:58:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2e21afd23238f3e2a30153728d399c5c41f71c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 08:03:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ef6beb87515345c2a9dad37943fe6b18 : SUCCESS in 59m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1dad6fdce2f240abbfd8abaa7593029e : SUCCESS in 1h 31m 34s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/377eb29185da4694a7246370d24b0444 : SUCCESS in 1h 43m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b1cb9e11912e4d75948e0ef84ce42dca : SUCCESS in 59m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a8878bb6161461b91466733f16329ad : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c904a1f84b024466975d14faa797783b : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0167f427a80646439a4e4e125a35b592 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace22cf1cee74a2297aa448062597358 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/530cccb4934c49f09769c611ca5f2945 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb62da9230ca4080be5b5920ccc908f3 : SUCCESS in 8m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9aa672105a2442eab89846921554014 : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e401dd75949846339d57e611ff91c518 : SUCCESS in 59m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0cc8bb4cf5d2407c9a679d2e010f60ba : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":2},{"id":"702f26fdaf6d447360f357079de825a615a6f93c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 08:03:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a7b779a8cd1faa1dbceeefc963f4957b2b196c34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 08:05:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21588fd54a9496398408414d87da540 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"8d7b5b6e7b6e171e25e86d6cae92b3ae3836906f","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-10-22 08:07:42.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d6d5790a59e8e5147278054940f362ba0f731bf0","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"823e45a6c909437a6530c9f34f9e4f5cce8ab79c","revisions":{"7317c19be690845169f9ba747af421f5fb43d9ac":{"kind":"REWORK","_number":1,"created":"2019-10-21 14:29:58.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/89/689789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/689789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/689789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/689789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/689789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/689789/1"}}},"commit":{"parents":[{"commit":"3887db89af16822404c37e0ceabb76bb9614722a","subject":"Merge \"Bump Openstack Exporter version to 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3887db89af16822404c37e0ceabb76bb9614722a"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-10-21 14:26:54.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-10-21 14:29:56.000000000","tz":180},"subject":"Fix python3 compatibility for barbican","message":"Fix python3 compatibility for barbican\n\nBarbican refers to python3 but uwsgi refers to python2. This change\nfixes this issue.\n\nChange-Id: I508bb082a319955b3e008235e158eaa07a1e8a74\nCloses-Bug: #1849128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7317c19be690845169f9ba747af421f5fb43d9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7317c19be690845169f9ba747af421f5fb43d9ac"}]},"branch":"refs/heads/master"},"823e45a6c909437a6530c9f34f9e4f5cce8ab79c":{"kind":"REWORK","_number":2,"created":"2019-10-21 15:58:16.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/89/689789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/689789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/689789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/689789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/689789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/689789/2"}}},"commit":{"parents":[{"commit":"3887db89af16822404c37e0ceabb76bb9614722a","subject":"Merge \"Bump Openstack Exporter version to 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3887db89af16822404c37e0ceabb76bb9614722a"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-10-21 14:26:54.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-10-21 15:57:53.000000000","tz":180},"subject":"Fix python3 compatibility for barbican","message":"Fix python3 compatibility for barbican\n\nBarbican refers to python3 but uwsgi refers to python2. This change\nfixes this issue.\n\nChange-Id: I508bb082a319955b3e008235e158eaa07a1e8a74\nCloses-Bug: #1849128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/823e45a6c909437a6530c9f34f9e4f5cce8ab79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/823e45a6c909437a6530c9f34f9e4f5cce8ab79c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
