)]}'
{"id":"openstack%2Fkolla~690003","triplet_id":"openstack%2Fkolla~stable%2Fstein~I508bb082a319955b3e008235e158eaa07a1e8a74","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1849128-stable/stein","hashtags":[],"change_id":"I508bb082a319955b3e008235e158eaa07a1e8a74","subject":"Fix python3 compatibility for barbican","status":"MERGED","created":"2019-10-22 08:07:42.000000000","updated":"2019-10-23 10:00:11.000000000","submitted":"2019-10-23 03:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690003-1571802385130-5cb63f8a","meta_rev_id":"054a7456003d3566c7463a2f2d15a24675b6bd2a","_number":690003,"virtual_id_number":690003,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 03:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-22 11:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-23 03:00:38.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-10-23 03:00:38.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 11:52:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-23 03:00:38.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-10-23 03:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8b1a0bd43ffc60dfbc355c934205814eeeed1a0","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-10-22 08:07:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2bc64a4e49394e27b27cc395365e1f362d0ee886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/401f34d488364344b92125f635d26877 : SUCCESS in 45m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e9c8c7b055284ca0a4d19ffd0c83ed28 : SUCCESS in 50m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b303e1aa474645cba58e67495a44947f : SUCCESS in 1h 23m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6ead9e399f0349eb8db3ef75925133d2 : SUCCESS in 1h 03m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9929a3234c334812831bbf5d72445e90 : SUCCESS in 1h 39m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8aed882b2a9a4a72812eb08083e27eec : SUCCESS in 1h 00m 06s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/34b7e0f202f34c8c8ca47175a54e86c6 : SUCCESS in 49m 28s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5985cb0ee5f64a3ca723fd92a5f3b654 : SUCCESS in 50m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40430ac52f1849ea9edbd948158cb5b5 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b274a12995cc4a8b96cb5767883509b8 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bbb240d390b4457836a2d9c81fe07a9 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4710cf115aa94d569a97181ef42b504e : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16d2d69daace405ab7c9a3315c12d35e : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c6a702c3894d148b28b9cfaa33ad51 : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abadd931b8604b638cce35a3f342021d : SUCCESS in 33m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/daca54ce937a4852b6cca6751e2ded73 : SUCCESS in 30m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27f7b55cd8b14f9eae5cefbf8c0ae165 : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8b9cb293b4c4962855d7a3cd9d01019 : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b09b1e1d13914d30a82991b1f72c74d3 : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7314d585973c8a92dd87b37015af8014b744d109","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-22 11:52:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96a5ea614ff5ef9bdf985835e8a4f067396bc9c2","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-23 03:00:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38d6c67c7bcd1ed10e19e133697312d375311b63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:00:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a795a9555afe4ef8f7ab2a90c56ca1a8945b2335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:46:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5a3d1bb4362144eba302d7ee91c9559f : SUCCESS in 42m 31s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/25bfa06390b84578ab06a9d1d98e9be8 : SUCCESS in 41m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85044148a144401a9d8c4fd1b6fc0918 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d411a186474d4a949ad0744e4f30ea : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7492f5dff2f446da3b7c64c733b3850 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7a8ce851a5d4fc383991afc0d68717d : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5db58b9f955a42c29ab33b069df5b570 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5092ee1ff564461a8b4c460dc3de73c1 : SUCCESS in 3m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea3afa0316404d0aa38fce78e8ac9be7 : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":1},{"id":"168d3f98f4b038600b559be7229f4ed5a16ee279","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:46:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"382abdfa14372f48a4a713941f55bd9ed7d3d99a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02958a8f5c544079812855328a794f3 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"df98cf0a07c1519be2fa99b7e3a8027ba7de770e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 09:53:46.000000000","message":"Patch Set 1:\n\nHold up! As I mentioned in the original review, only binary images used python 3 in Stein. We need to revert the source part of this change.","accounts_in_message":[],"_revision_number":1},{"id":"b0507311e771617f3fab2a408897ab6b72370b47","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 10:00:11.000000000","message":"Patch Set 1:\n\n\u003e Hold up! As I mentioned in the original review, only binary images\n \u003e used python 3 in Stein. We need to revert the source part of this\n \u003e change.\n\nFix proposed here: https://review.opendev.org/690533","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6d5790a59e8e5147278054940f362ba0f731bf0","revisions":{"d6d5790a59e8e5147278054940f362ba0f731bf0":{"kind":"REWORK","_number":1,"created":"2019-10-22 08:07:42.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/03/690003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/690003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/690003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/690003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/690003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/690003/1"}}},"commit":{"parents":[{"commit":"6a26cf5b0a9d6fb8375fa8d6a1bd57bdb462544f","subject":"Merge \"Bump Openstack Exporter version to 0.6.0\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a26cf5b0a9d6fb8375fa8d6a1bd57bdb462544f"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-10-21 14:26:54.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-10-22 08:07:42.000000000","tz":0},"subject":"Fix python3 compatibility for barbican","message":"Fix python3 compatibility for barbican\n\nBarbican refers to python3 but uwsgi refers to python2. This change\nfixes this issue.\n\nChange-Id: I508bb082a319955b3e008235e158eaa07a1e8a74\nCloses-Bug: #1849128\n(cherry picked from commit 823e45a6c909437a6530c9f34f9e4f5cce8ab79c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d5790a59e8e5147278054940f362ba0f731bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d5790a59e8e5147278054940f362ba0f731bf0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
