)]}'
{"id":"openstack%2Fkolla~690525","triplet_id":"openstack%2Fkolla~master~I36c1feaecf2e07e2d9fb8a7f42e899477eaabae4","project":"openstack/kolla","branch":"master","topic":"bug/1848295","hashtags":[],"change_id":"I36c1feaecf2e07e2d9fb8a7f42e899477eaabae4","subject":"Do not build rally image for Ubuntu binary","status":"MERGED","created":"2019-10-23 09:26:40.000000000","updated":"2019-10-23 16:34:25.000000000","submitted":"2019-10-23 16:32:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690525-1571848334329-d71617f4","meta_rev_id":"a1560d3ba350990116225790220d92308e30a73f","_number":690525,"virtual_id_number":690525,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 16:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-23 13:49:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-23 09:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-23 09:45:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-10-23 09:45:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-23 09:32:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 09:42:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-23 09:45:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-23 13:49:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-23 16:32:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ae5b3f57da42a6b953bb1a0f040c5e851f9dfd8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 09:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318da42587146dcb4409286705a0a1253454aef8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 09:28:45.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"0be52b6e453b59299343777f6a16640e833dfcfb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 09:32:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a28c01b049660fcd6ae97ca8f187804bdaff02a4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-23 09:42:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"40bb7edaff4ba265d1467642dc8b499928c3cc91","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 09:45:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTYVM","accounts_in_message":[],"_revision_number":2},{"id":"99b57672f42f5e6819269b9756794ca137c4f4de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-23 10:15:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/25/690525/2/openstack-check/tripleo-build-containers-rhel-8-master/741f88e/ : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e5e9789c558495b039e94cc9da15dc420c69937c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:08:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/71a4053142e4421ab5baa685b23a8076 : SUCCESS in 41m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7d3877bb2e3d4f699142186683a5b9b0 : SUCCESS in 54m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/62773d28c94448978a7269e1d7c97f1c : SUCCESS in 1h 29m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5ac1acb74cc044c7b1f0064143568f83 : SUCCESS in 1h 26m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7637383af42d453699084055f7b92f9d : SUCCESS in 1h 32m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/126b12a270994aa18d3767a489e75367 : SUCCESS in 1h 02m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d3b4fb360e542f180129617c42ad518 : SUCCESS in 32m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/24315f0c8e0a4451a5db9ff97dd2c630 : SUCCESS in 18m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5db291687f2f4d20877d0272098f8ccf : SUCCESS in 51m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09c37807847d43a09f0c7c93e0b11f91 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85764ac4c5e74454959583e69201d30a : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c65356e874b4064b10721f4e189e35f : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6e994ddea0447ce92f7f072071e775b : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f609c0bd65a34f0db7186f561027b02b : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8664af9093e482e9a97d7bbbe618451 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bca32a01e91347848ecdf9dfacf21284 : SUCCESS in 27m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f23dfe2d28134562a28de4f098f384c4 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08d1a15b64e94fc397393faa8b96294e : SUCCESS in 1h 08m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdd7ed934e414fd3b777fe5b6b77c150 : SUCCESS in 59m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4733bd3831ed4a6384d95c6f18395fe3 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":2},{"id":"f9fef50fed86413cba6ed332a76f08d9547fae1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:08:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be46a6b723bf1bb5a48049ee402b9f5ebf0c72db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 12:40:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/180eccc21a7f41448a0671801bc47a25 : SUCCESS in 51m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/abad844991df4985b31edcf5827ce90f : FAILURE in 9m 53s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/72402eccc7a94a4bb38949923f5d569e : FAILURE in 9m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4494482a32d54ed3a62d7479e00725dc : SUCCESS in 53m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7ddfccfb65e404facf660580e4d0934 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b9eb64549448e09086abaeab7b9292 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e83c23086cf8416b89e88d686440dd70 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9497bdb0aec24a7ba425a0a7daba3bcc : FAILURE in 2m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75b0e495111242a798d098e3ef8be2ae : FAILURE in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7e4242380045838992a34d3ad951b6 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/099912f6344449138918717b6c71f4ff : FAILURE in 3m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1346f7873bd42e5b8e2f45cf48c7e7a : FAILURE in 7m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a210ef82dea546f7b53d993999356de5 : FAILURE in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"50a8b089e0530e0cd780bdf5f4cbb9d8de7c76ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 13:10:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4f5e14cd8b4be0b8f853121239c694bb8552c0a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-23 13:49:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/25/690525/2/openstack-check/tripleo-build-containers-rhel-8-master/b925128/ : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3d65fdfd6ec43dfc75442163d0525aab82c3aa38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:48:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1810f68183a84440a3e987c730832f47 : SUCCESS in 42m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/38873e88444e411a82d83e2055c56a8d : SUCCESS in 1h 06m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/748311f366964725b9140838ef4bab23 : SUCCESS in 1h 23m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b71b722d0f794d58bca56e00b8ebe33c : SUCCESS in 1h 29m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d4c7f26816a8470385fb2d864414fb4c : SUCCESS in 1h 31m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ae8315d7b4b346d9b1ec11b9ee36fc75 : SUCCESS in 1h 04m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/03991711711147999847b653287513f4 : SUCCESS in 26m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba415ec8636e46b1b3f999afdb05388f : SUCCESS in 20m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1fdda09f5b174d00bca50599d782bb7a : SUCCESS in 52m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe9717f6cdcb4846991b7c3dca557103 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f78b12b32d14490c8a4e31b8c07f9e6e : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58255950bb794dab84e22409c6678995 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ea897ecf154c74bc3f6a9ceff33b88 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cce3ed7d9e614773a64f8d0569840551 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81967daac7f848968bc83be6cee1ac8f : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5705016ab8f4ababc42090bdb357169 : SUCCESS in 31m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a4ad4f832254e0b8d740ddd39f93a15 : SUCCESS in 36m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/083c5cf92dec4e18b829ecf2a856ef0f : SUCCESS in 1h 13m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41b7308ed5cf4a3f85ef94f6b44712e2 : SUCCESS in 1h 02m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b8e31c32a054c8a98ec2b6a3b30175b : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7694d412f37bd0dd07893fbc4b87eae3e3293d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:50:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"202fff18a0571ede9bef125cd94a85cdef15647f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:32:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"700ca7bf141e48cf5766942146c31cac671f98f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:32:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3955a8eccfad4e5aa81564d34f49bb31 : SUCCESS in 54m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bead7bd24bc4457aa4c09851bd9a9f91 : SUCCESS in 1h 25m 00s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ad939297918e4535960a5ad9018ef9c1 : SUCCESS in 1h 39m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/03858274a00341ca8dd2c060a260bfa6 : SUCCESS in 54m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4134318b456470b89b3adb55a484fcf : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9ee75f382a4318b4134192ff01a1f0 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8764a8c7024408d948fbcd26d0bb10f : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08815cc09ef0447bb0f350d05e2b6c7c : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c58c5268714d4cbba80d3b144021e858 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ee1c3f28e540debe2fccd3517154ea : SUCCESS in 4m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/930a767d07134d8fad8b48c8da634d35 : SUCCESS in 26m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1913bede7c5c4efa8f09a52a6bcd1e5f : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f4b2f7a3f1da4f59a73730aba61fe6d0 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"28aecb8b82f8501f4bbf4d7435996beea793b09d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:34:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb4a47204c44c18a0277fe1961fdf5e : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0647a062ba1260820e3b94a4c510a4b16a218528","revisions":{"098d2afd3672140dbd0a3559f55aad62ffe40993":{"kind":"REWORK","_number":1,"created":"2019-10-23 09:26:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/690525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/690525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/690525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/690525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/690525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/690525/1"}}},"commit":{"parents":[{"commit":"e239169e13319ca51437604f152bb61d82ab81e9","subject":"Merge \"Switch OPENSTACK_RELEASE to train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e239169e13319ca51437604f152bb61d82ab81e9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-23 09:25:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-23 09:25:15.000000000","tz":120},"subject":"Do not build rally image for Ubuntu binary","message":"Do not build rally image for Ubuntu binary\n\nUCA does not provide rally and the Ubuntu rally depends on python2\npackages which conflict with python3 packages.\n\nChange-Id: I36c1feaecf2e07e2d9fb8a7f42e899477eaabae4\nCloses-bug: #1848295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/098d2afd3672140dbd0a3559f55aad62ffe40993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/098d2afd3672140dbd0a3559f55aad62ffe40993"}]},"branch":"refs/heads/master"},"0647a062ba1260820e3b94a4c510a4b16a218528":{"kind":"REWORK","_number":2,"created":"2019-10-23 09:32:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/690525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/690525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/690525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/690525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/690525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/690525/2"}}},"commit":{"parents":[{"commit":"e239169e13319ca51437604f152bb61d82ab81e9","subject":"Merge \"Switch OPENSTACK_RELEASE to train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e239169e13319ca51437604f152bb61d82ab81e9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-23 09:25:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-23 09:31:53.000000000","tz":120},"subject":"Do not build rally image for Ubuntu binary","message":"Do not build rally image for Ubuntu binary\n\nUCA does not provide rally and the Ubuntu rally depends on python2\npackages which conflict with python3 packages.\n\nChange-Id: I36c1feaecf2e07e2d9fb8a7f42e899477eaabae4\nCloses-bug: #1848295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0647a062ba1260820e3b94a4c510a4b16a218528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0647a062ba1260820e3b94a4c510a4b16a218528"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
