)]}'
{"id":"openstack%2Fkolla~690531","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ied5d2cb8ffa47b90833ee3cf241797601906f9b2","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ied5d2cb8ffa47b90833ee3cf241797601906f9b2","subject":"CI: Increase job run attempts to 5","status":"MERGED","created":"2019-10-23 09:51:48.000000000","updated":"2019-10-23 14:56:29.000000000","submitted":"2019-10-23 14:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690531-1571842430667-d762e54f","meta_rev_id":"d995327adde46751e583adb7b9ed6fd17369f13a","_number":690531,"virtual_id_number":690531,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-10-23 12:45:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 14:53:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-23 13:37:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-23 13:11:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-23 13:37:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 13:11:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-23 13:37:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-23 14:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6de36561fbe33c671f01e911c6d315b4cd75c15d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 09:51:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a9c2dd10ef2af435ad8be1c64d4a56498ff73ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:33:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/45551ed8f3554b15894c97adf677ace6 : SUCCESS in 49m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/88fe444baf51466b9cea99e6e41d8e6e : POST_FAILURE in 27m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5244e0914a5f4d99b370202998c74353 : POST_FAILURE in 1h 04m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9134320757b14bdc9d86a2430a887d53 : SUCCESS in 1h 13m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/df4dca83b7f1431282ba612d37d874a2 : SUCCESS in 1h 37m 59s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/748770ff71a848d1a452d98e3f5cfcaa : SUCCESS in 54m 02s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/59f7748e01e64656bda8ae3cfe0ea985 : SUCCESS in 55m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65458c3798904299ac5bb697b06354c0 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45356fe38f7444949d79416d59f85735 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa2cba3608a64b20b0e52131bd2dafa7 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9b03ac315b948629d2be98dd4270667 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a7ac6d239bd470fac7df0bfda88a68a : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9b3c99eefa4bf5823ef1c8e0e5b89d : SUCCESS in 3m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca1c2823873e4da29c60d66ddfd17af0 : SUCCESS in 31m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd8841a9fcae4d27b4ed21f31cb8c099 : SUCCESS in 30m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19033f7f6139434294f39e88ef329a7e : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f7a014143fe4d37966404591f84ef0d : SUCCESS in 47m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/469c7baf77b24b47bf9fc25a879f38bf : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":1},{"id":"09b890013784f2e2a847bd09df227a53c6dad7b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 11:57:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6877e53fd1e753f7e07bdb082cfe5e2c556f6a3f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 12:45:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0dba2bbf667206194eef78396a688a33e6e975ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 13:11:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85b422410d493e5b2a094c93b8e951c1c2db24f3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-23 13:37:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5e34e4a557212624f6a997da5dba9d65ed48218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 13:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/98419d330ff14d5f9887ca1b1b059f99 : SUCCESS in 52m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/68de3c8cfc284895a516616b465ceba7 : SUCCESS in 1h 00m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/888ae1534170457f95a80a0e22d707e9 : SUCCESS in 1h 35m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7064cbd4b74a4abfbb7a17e53b597f42 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e3211ff2c9044342845cafca602862ce : SUCCESS in 1h 49m 58s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9cc8fefc54874bb3b926823c403d07b4 : FAILURE in 59m 47s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e6281d922d0e4b3fa3537bdfd35b8013 : SUCCESS in 55m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd45e92f62514446811640daa229dbfc : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3409b19a35904ae9ac4b1ec71f41985a : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3abb9d7eabc74539b4533dbd103d1d8f : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f94f26c302ef45cb96c294a754911585 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/701cfb80f896462188b4a38a6a29ea1a : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a531dab3662f4da9ab5b2feb6ea10014 : SUCCESS in 3m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7bf1e80f145f405ab613a33fd4828076 : SUCCESS in 36m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10a1afffd47e430ebbd12e03d3089de2 : SUCCESS in 39m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f0af9529703472d9972db5a8d686bbf : SUCCESS in 1h 13m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02d76a3d7bff47519edd8d82f2793537 : SUCCESS in 55m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c1e4dc9cb6c54e90ade75653def1ea85 : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":1},{"id":"70e0155b0e0719a76bc9cb39582e204c685088ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 13:53:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27cd9fae469befc60e40d396cd7e5e0be93eb46e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:53:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4762daa06e5980fcb4012d59be79ca38d51d2671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:53:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f425a7e0be5f43f5a1ce151a7e0cee87 : SUCCESS in 55m 07s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2fe009151fc045dab162226a2f403f94 : SUCCESS in 54m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f080690014154db19dfb13704cdafcc0 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02bf4c9e1a5c4e5bbe40b57ccdb1b7e0 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df7695ecb8344609afc5eab70acf255c : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f94abfd48314994b6e9997ce8339f6f : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/623d579096f64daf92eac3a1ca28da94 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a6b4a0c9c84740aa6cbca2865e60ec : SUCCESS in 3m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/93fe71b9a5704b0ba7e8a4d1f96b0fd7 : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d995327adde46751e583adb7b9ed6fd17369f13a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:56:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab807c14d94f43bfa36589f981bcfb0e : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff815baf44b0ed4348b4c9c63eef704e05e10c37","revisions":{"ff815baf44b0ed4348b4c9c63eef704e05e10c37":{"kind":"REWORK","_number":1,"created":"2019-10-23 09:51:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/690531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/690531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/690531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/690531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/690531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/690531/1"}}},"commit":{"parents":[{"commit":"d6d5790a59e8e5147278054940f362ba0f731bf0","subject":"Fix python3 compatibility for barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d5790a59e8e5147278054940f362ba0f731bf0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 14:48:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-23 09:51:48.000000000","tz":0},"subject":"CI: Increase job run attempts to 5","message":"CI: Increase job run attempts to 5\n\nAttempts affect pre failures.  This means we can increase stability of\njobs by rejecting nodes that fail pre without failing runs at the same\ntime (unless we are really unlucky and hit b0rken nodes 5 times in a\nrow).\n\nThis change is adapted from I17b7f878c742fa8db66f738526855a02ab9f1905 in\nkolla-ansible.\n\nChange-Id: Ied5d2cb8ffa47b90833ee3cf241797601906f9b2\n(cherry picked from commit 98fb7dc8ff2136d658c9a79ed40f6e42e7893662)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff815baf44b0ed4348b4c9c63eef704e05e10c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff815baf44b0ed4348b4c9c63eef704e05e10c37"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
