)]}'
{"id":"openstack%2Fkolla~690532","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ied5d2cb8ffa47b90833ee3cf241797601906f9b2","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"Ied5d2cb8ffa47b90833ee3cf241797601906f9b2","subject":"CI: Increase job run attempts to 5","status":"MERGED","created":"2019-10-23 09:55:02.000000000","updated":"2019-10-23 14:36:58.000000000","submitted":"2019-10-23 14:36:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690532-1571841418462-70f2cce9","meta_rev_id":"f883007a7148af0054cdb46c6ff4e08158bbdf4b","_number":690532,"virtual_id_number":690532,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 14:36:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-23 13:37:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-23 13:13:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-23 13:37:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 13:13:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-23 13:37:38.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-23 14:36:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"278eab605c388317c91d9793b488058d1e97afb8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 09:55:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce0f070adfcea618429a0e6cd3ca725200eb9620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 12:03:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5b8e093ee67b4ab78d4b40c4344b7b08 : SUCCESS in 38m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/48d363a11c134412a8da67dce437a55e : SUCCESS in 50m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/39d41bfec64a404f85773ddad6eb92c7 : SUCCESS in 1h 24m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/91f955217aa347418715e259f8c02b75 : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d1796465b74c498980d115fd31b52db9 : SUCCESS in 51m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f773a0599a394a9f9e6e55906422b784 : SUCCESS in 40m 57s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ee69297770424d68a67e71016f615c9f : FAILURE in 1h 31m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4dea0df516a04c82ab959ee823a725cc : SUCCESS in 1h 28m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e86de486e09840f8836447b22d265104 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bcf6ae05d164ae19e96ba63b32c7cc8 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/018207fdc44441da9ae969dff005cb4b : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/351a400b5e3a41cca63126ea415a9354 : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/804267de594d44cda85c05c805534e6f : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/063011fb3da449a2b492b78f011fd75f : SUCCESS in 1h 42m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/15094419f63540b697965a9d0affe31b : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b2fe42207b44b7c96274b484aad41a5 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d441968501e842d1a133228344efd831 : TIMED_OUT in 2h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6bd9d7b64dfc5cf8427cb5a20bdfab6f9c96da4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 13:13:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30580f366151880c263267d7dfe866c1bd4d6363","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-23 13:37:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"877eb61b9daac7f89794cad63f768e7b2d790d19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 13:37:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd6b5b2f3a21e9e672d30a5fe837e4cdf773de0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:36:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f883007a7148af0054cdb46c6ff4e08158bbdf4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:36:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/720ccca325c34a82b61894ccd2732f36 : SUCCESS in 54m 04s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3516cd3583ed4128a8c44504dd305c77 : SUCCESS in 47m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/668a186e14e34831b548bc4b625f819e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9bb36182f314defa2598a1e55435da6 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a124b9fe57574310af9baeb89579fa8f : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d913af332f14712bc67d4dadddf50c6 : SUCCESS in 4m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1b0e5a9c3f2040c3ba2115b58f17a510 : SUCCESS in 3m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ddc61e330cf5461eac023eab7dc2fe4d : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f639e85df24a8d6d17cd5c90db4fc977fbe40b9","revisions":{"4f639e85df24a8d6d17cd5c90db4fc977fbe40b9":{"kind":"REWORK","_number":1,"created":"2019-10-23 09:55:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/690532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/690532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/690532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/690532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/690532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/690532/1"}}},"commit":{"parents":[{"commit":"728818cd8ad016a5925ded3070d655dfd2c3db9f","subject":"Merge \"handle push error properly\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/728818cd8ad016a5925ded3070d655dfd2c3db9f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 14:48:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-23 09:54:48.000000000","tz":60},"subject":"CI: Increase job run attempts to 5","message":"CI: Increase job run attempts to 5\n\nAttempts affect pre failures.  This means we can increase stability of\njobs by rejecting nodes that fail pre without failing runs at the same\ntime (unless we are really unlucky and hit b0rken nodes 5 times in a\nrow).\n\nThis change is adapted from I17b7f878c742fa8db66f738526855a02ab9f1905 in\nkolla-ansible.\n\nChange-Id: Ied5d2cb8ffa47b90833ee3cf241797601906f9b2\n(cherry picked from commit 98fb7dc8ff2136d658c9a79ed40f6e42e7893662)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f639e85df24a8d6d17cd5c90db4fc977fbe40b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f639e85df24a8d6d17cd5c90db4fc977fbe40b9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
