)]}'
{"id":"openstack%2Fkolla~690533","triplet_id":"openstack%2Fkolla~stable%2Fstein~I62c1a42c29ce7273028f99edbb77f4362a2452ce","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1849128","hashtags":[],"change_id":"I62c1a42c29ce7273028f99edbb77f4362a2452ce","subject":"Revert to uwsgi-plugin-python for barbican source images","status":"MERGED","created":"2019-10-23 09:59:54.000000000","updated":"2019-10-23 16:13:01.000000000","submitted":"2019-10-23 16:10:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690533-1571847019647-70f2cce9","meta_rev_id":"06a43426242d84272606fd653c709c9daf3c74c3","_number":690533,"virtual_id_number":690533,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 16:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-23 10:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-23 10:05:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-10-23 10:51:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-10-23 10:00:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 10:05:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-23 10:51:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-23 16:10:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1ddaec6823897fd197b5969a429de5010d54c1a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 09:59:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65bf18f7fca084eb3b64b59f6a2ca8eeb8ad523b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 10:00:23.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6d62427a0544d7405b7df74cd7b9beb2cfd9c5d9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-23 10:05:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"95becebf85bcb16b54b2b54860b5d9df362b326e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-23 10:51:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1dab6773f64a08dd276d566d74fa4edd92bde8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/936fc660bfbd4fcbb002a7d897d2d9bf : SUCCESS in 57m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5d761872ca84497ca5d48611c32c84a5 : SUCCESS in 59m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/de71086dd2a8416e93e752949399f912 : SUCCESS in 1h 11m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5b9b5b4a3e00448eb750d6ffe1c17dd3 : SUCCESS in 1h 08m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bc8730d8bedb4da5b6d4c8e7327327a6 : SUCCESS in 1h 17m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/87dfeb36c77b469f8bf905b38b13ba6e : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8af97486ef074d8bbb5e53e14202596c : SUCCESS in 45m 46s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ff00bf7cfdd14e2187263a52490d1ae6 : SUCCESS in 47m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9542c74b72641259f19b47e5a292892 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe184e32ab8484d830582e2d4ebcbaa : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13d6ced699534fbba12593e2879314be : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ae140d90334c74afbc0f4a0bb8bcf0 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b001f27e4ac547399a745db3b70746b5 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0188f56c6840169c38b6d4bab6c8e7 : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebb400ebef3f42fc8f7a1afccc069668 : SUCCESS in 35m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/846403b4658f490993dfafee478fc5b4 : SUCCESS in 34m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fbd0f5b53f94c7888a455a2baa14bff : SUCCESS in 56m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd26c885d8554692b55d740faaef0651 : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fa3cd5c3421644caa16e7ed525631dc0 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"408720a139825f041b3d33a9c0624e26951a2ba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:21:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"827662d46d04e6c141e0eac4f26a289bbbda397d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 12:41:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e70b2690eb76419386772bae7ac503f4 : SUCCESS in 47m 53s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c181e629b5cc4ecab6c78fbe16c1a22d : FAILURE in 51m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1c210d3957a4162af41502984ebdd0c : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e15e59b7274109b942c3a7783f4947 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a261803c6984f58a885882767315ad9 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50749ccde54c45458889e7486f4a03a3 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abe4eac3a15a40119258ba4ba237339c : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9344721179f544c498644a4d1ac0f3a5 : SUCCESS in 2m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0fac28d5509f41f683cc01e00b62c32f : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":1},{"id":"aff672010999f3131e4937dca14f4b4f3f53188c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 12:45:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3370bf2a60d94751ebe6ad2f4a8e337ae164781f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:32:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/46b6b6d175e54ec9a8dc59e0977ad82e : SUCCESS in 53m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5500cef236784daf9f43bf67f9211ddb : SUCCESS in 51m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6e65f9c31abc42e5877d5338548a2565 : SUCCESS in 1h 13m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a7430453291e4add88f6b4e1a430fc5e : SUCCESS in 1h 15m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c23f750a770e4863bf9808b2a27003ca : SUCCESS in 1h 40m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2818115289924c4cb9c1825579c632fc : SUCCESS in 50m 23s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/94338a48851a441694299560fafccf4d : FAILURE in 58m 54s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dde925850dac4871836bbbb440f0f042 : SUCCESS in 58m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc8474f7b7ef46fe8115e49fc5a8d239 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e6f7cf7f124750a3e9e9f412b5f46d : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7624a613b67f4cf284b56086b3b32163 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/651cbe10ad1e4fdd8efa0a450b9d91e5 : SUCCESS in 2m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6c03a4e04f741179b2d17628d7d12c4 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da85ee12183f4a4782c8edb5e65a23c3 : SUCCESS in 3m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90845e10dfc04e51acb16705bff479d0 : SUCCESS in 22m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/427dca60c6104e559ad02d552e930977 : SUCCESS in 32m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e256f26f78994f19a9b0ca292e49de59 : SUCCESS in 1h 10m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6bb8b8575cef487da946716b0706aeb2 : SUCCESS in 50m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/81fdc26e02664acab98e6e284773b43a : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"98025b27c6085ff3c964cc9ebda808b26d49b38a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:32:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"478deaafdbd2c663d3489f21cb7b0f5302283ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:10:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1df9f3a175e241a4a1272086871021a8 : SUCCESS in 48m 15s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7ff3c420c7ca4290bc19f5979ba41663 : SUCCESS in 1h 01m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92b1d086e85f4b05aa39630f53228626 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a520536a36746879fa88f973231d77f : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/725e96a5f6b8452da547c06bbcd88eb2 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/162dff791d2543e3bf8efb5ebd588a23 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1bd198b75594695a21e9031094cba3a : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3c7862a10d4e85be75f17982e60e15 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cc87a3488e24e3f978653715d71e8df : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ea4fdd95db905aebf259a89f40f114ca5c5794a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:10:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0e4ee934f3564c5840d9ac3de717e9c9b262859c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c3711b32944d53a42c3a9696ebf892 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d65eab1eaf7a86774ac0b2f3c98646b85fd0e641","revisions":{"d65eab1eaf7a86774ac0b2f3c98646b85fd0e641":{"kind":"REWORK","_number":1,"created":"2019-10-23 09:59:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/690533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/690533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/690533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/690533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/690533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/690533/1"}}},"commit":{"parents":[{"commit":"d6d5790a59e8e5147278054940f362ba0f731bf0","subject":"Fix python3 compatibility for barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d5790a59e8e5147278054940f362ba0f731bf0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-23 09:56:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-23 09:56:55.000000000","tz":60},"subject":"Revert to uwsgi-plugin-python for barbican source images","message":"Revert to uwsgi-plugin-python for barbican source images\n\nStein only\n\nI508bb082a319955b3e008235e158eaa07a1e8a74 was backported to stein to fix\nan issue with barbican images which contained the python 2 version of\nuwsgi-plugin-python on Ubuntu/Debian, but should have used\nuwsgi-plugin-python3. However, in Stein only Ubuntu/Debian binary images\nused python 3 - not source images.\n\nThis change reverts to uwsgi-plugin-python for Ubuntu/Debian source.\n\nChange-Id: I62c1a42c29ce7273028f99edbb77f4362a2452ce\nRelated-Bug: #1849128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d65eab1eaf7a86774ac0b2f3c98646b85fd0e641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d65eab1eaf7a86774ac0b2f3c98646b85fd0e641"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
