)]}'
{"id":"openstack%2Fkolla~691270","triplet_id":"openstack%2Fkolla~master~I29dbcea169a6e338bc9a8fb4fcde7a77c1d6fac6","project":"openstack/kolla","branch":"master","topic":"ci-drop-the-gate-profile","hashtags":[],"change_id":"I29dbcea169a6e338bc9a8fb4fcde7a77c1d6fac6","subject":"CI: drop the gate profile","status":"MERGED","created":"2019-10-25 09:21:58.000000000","updated":"2019-11-05 14:00:01.000000000","submitted":"2019-11-05 13:58:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691270-1572962313169-f546210b","meta_rev_id":"5112189fb5f68fd906c8b4ed265d39e0c7f630ca","_number":691270,"virtual_id_number":691270,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 13:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 10:58:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-28 07:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-25 09:59:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-10-28 07:57:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 09:59:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-28 07:57:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-05 10:58:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 13:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cc70899d9e6f83277f4c7b70179912ec8c691d3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-25 09:21:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30cd23a3b2307b4a58b638b9955e048760b47694","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-25 09:58:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/70/691270/1/openstack-check/tripleo-build-containers-rhel-8-master/965d787/ : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d7e21a23b3a049eec20a09f464d9712f7006a5c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 09:59:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5855f9d5e7956c316a89a8c1095c035a2dfc7911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 11:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7525966a73bc4347b1d1f86c90e626ef : SUCCESS in 45m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/21717a85eae44661bb922dfdcdd7a8c6 : SUCCESS in 1h 03m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/72ed99c9d3f34d7c9d94e088577e64b0 : SUCCESS in 1h 40m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9fded48dfb40449bab5f14763adf7497 : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a42fffe5414341aca247ba550c4d99b3 : SUCCESS in 1h 22m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/55b1ff1ce819431c938811eda88b6e03 : SUCCESS in 54m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9807526d18394358be798ea9cdc859e0 : SUCCESS in 24m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/103cd4db7c224a62958738d6cddaefe6 : SUCCESS in 19m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f34e213417b34d63b731889b27fba7ec : FAILURE in 23m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e4ad1be4e054bc182a242770f5eaf35 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf9ded30b4e44c1abb62c4b63ded480 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3e515634ee44bcca82ab7d1e60fba90 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3662b74e9a5435a8463cfc8920382e1 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b5950b4928d40149a75784c90d44728 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec9be84b74646a494446489d6b6cc7c : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc2d6b75b96d4eeba3c98c772de500c1 : SUCCESS in 32m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00abddedfecf41ce9aacc2dddf9c5a27 : SUCCESS in 35m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dad4b863f16d4df48b67f8f63eb957ab : SUCCESS in 56m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/134ce1ecd8494d13a9027f3d6900c221 : SUCCESS in 1h 02m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/414242e4dc954d81b3aec41d5d339fed : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0ddc3353d3aaaa656b37422881c36d3bc090186d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-28 07:57:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"714457caf644e67d21b6bf3d0bfb5bbf39e6c62b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 07:57:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3f54ae929dbbeb3b0685272d09f2f0ee7fddf61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 09:09:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/15eadc31c8da4e24b5257bd1d55eccf0 : FAILURE in 1h 07m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/437a4ad51ec649868b8a5f760069afdc : FAILURE in 30m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8327af3f78e049cc8bb430df93f9d30d : FAILURE in 51m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2c9458895f574dde84abcf5fedc4d764 : SUCCESS in 53m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d499b41723f7498280d5e4b1ab0431cb : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd4af786784441ebce31a80483e4efe : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4730d4c430504e7fa67785a576c3845b : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a273de0c47749fc869d57eb07f232a6 : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b91ef5b854e94398a4cbd388ba2c4014 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ddde35407341fcb07a5619083c4bfb : SUCCESS in 4m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eaf6f28372e74341b86e77db79113ba1 : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/195c17786d2c41d68de9b8cb8a9dbcfc : SUCCESS in 55m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3cd16d28d39c47e99a3ffa7b5afd678f : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":1},{"id":"11301d0020f6953e8724fc6c6f1391c9041a4a77","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-05 10:21:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e9d72e44793591850de1ce6446a546912973c02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 10:58:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/70/691270/1/openstack-check/tripleo-build-containers-rhel-8-master/7cfb0fc/ : FAILURE in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c401796b1b5869dfd7550835c96d302d7e1f2325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 12:21:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/322b1ca7601f40a4951e0e566d9aa8a7 : SUCCESS in 48m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/476462d242074b52b248d2765b4ae770 : SUCCESS in 1h 04m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/90117ceb7fd94c3d814e053a9a216778 : SUCCESS in 1h 33m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/669bc1f2e0d34267884059eef810df35 : SUCCESS in 1h 48m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/63e7fa32dd32403ab66f41d0f2896457 : SUCCESS in 1h 30m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c04298df81e34cad955bb7181eb5899d : SUCCESS in 58m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0828b84b9ebb4321a9fc9d35a18aa519 : SUCCESS in 33m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dc8b7cf7d4434524a40a6110819edfc4 : SUCCESS in 17m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5e2deac436a4472a95d3ff6695fdd734 : SUCCESS in 56m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ce6f80de5aa4cd8b680c31cbe434372 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a017dae41d0444fabeba3f396067cae : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7753b58b54dc4317800835ebfd2840f0 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0214c5153d349acb7c5a821c03f0575 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b38f20bf95c4742a2f1cb0eb00e6d6f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611871930c994a99b0cd8ee687a420d3 : SUCCESS in 7m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c38c3fe8cf8d485190cbeee8963bb9c4 : SUCCESS in 34m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/381a7c0e6b5e4616b788115a0fb2a104 : SUCCESS in 29m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7388645e95747778b049f1c8ee6f044 : SUCCESS in 50m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d31bebecc2ae4969ae1bee9d57c68cb4 : SUCCESS in 54m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9227c1b96bdd4fd7a8103ca6a330415b : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":1},{"id":"86121c4109a8c84ed6dc8f9807d4148271a30f78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 12:21:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4f26ea5dbc0df92fe33dd993596c5c1de67cc81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:58:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dcb82dc877d547ec8abc7db6bea40b72 : SUCCESS in 59m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4d7d0cafda85478388bffd508ce72dc5 : SUCCESS in 1h 20m 41s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/66220662a44a45ff9f5f6d01b9bd360e : SUCCESS in 1h 35m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78fe39fa5c2f4ab2b319769bbe21516c : SUCCESS in 55m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f8b5d0314d463a9533b00554b287f0 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c01fda28c6f461eaf4f7aee4d9c3fb6 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bf4be3bb8a7413ab4089bbdf3a16d59 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfaf2a9e7fa24e389f325e6e0b7c00d5 : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87c944ac3fad42358ee7719a57fbed74 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35aec5990d514ee59815feb7fd9952e4 : SUCCESS in 6m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/39e9a87e10ef42ffa29670d50afe31b0 : SUCCESS in 32m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c223fd0b11d14325ba7b7f4fa1dcceea : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7e8a02fddc440e29a678714413daf9e : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1780509032cf94328dc6ab97d54a9ee913b6387e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:58:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0be5fdd6e7cd5e7c9db36d7996fa8a8ad1cab794","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 14:00:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324431108e9047a39b41e63b6d0dbe99 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a465df591ce1541f9ff970adf86eb2f7e59845e","revisions":{"7a465df591ce1541f9ff970adf86eb2f7e59845e":{"kind":"REWORK","_number":1,"created":"2019-10-25 09:21:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/70/691270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/691270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/691270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/691270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/691270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/691270/1"}}},"commit":{"parents":[{"commit":"e239169e13319ca51437604f152bb61d82ab81e9","subject":"Merge \"Switch OPENSTACK_RELEASE to train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e239169e13319ca51437604f152bb61d82ab81e9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-25 09:20:21.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-25 09:20:21.000000000","tz":120},"subject":"CI: drop the gate profile","message":"CI: drop the gate profile\n\nThis profile has not been used since Kolla Ansible was separated.\n\nAlso removes the relevant \"tests\".\n\nChange-Id: I29dbcea169a6e338bc9a8fb4fcde7a77c1d6fac6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a465df591ce1541f9ff970adf86eb2f7e59845e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a465df591ce1541f9ff970adf86eb2f7e59845e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
