)]}'
{"id":"openstack%2Fkolla~691290","triplet_id":"openstack%2Fkolla~master~I84536b302991e8fb04d5fa67ea266bc428437ce1","project":"openstack/kolla","branch":"master","topic":"bug/1843043","hashtags":[],"change_id":"I84536b302991e8fb04d5fa67ea266bc428437ce1","subject":"Set pipefail in kolla_mariadb_backup.sh","status":"MERGED","created":"2019-10-25 11:34:37.000000000","updated":"2019-10-26 11:15:00.000000000","submitted":"2019-10-26 11:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691290-1572088434980-cc9bba01","meta_rev_id":"b700992d2774af342ed964a8d9fe65190ec075b5","_number":691290,"virtual_id_number":691290,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-10-26 07:00:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-26 11:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-26 07:42:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-25 11:55:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-25 12:36:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-25 12:36:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 11:55:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-25 12:36:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-26 07:42:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-26 11:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"facb44ff40a0c2908e5ca5d04875bbf931fff444","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94590812c4e121d67d299863fa63482374f98867","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:35:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2ff156a8542cb27ff9ad0bc57fe1537e711d851c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-25 11:55:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a51ab403229172f17d32316694c5137d7f99982","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-25 12:15:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/691290/2/openstack-check/tripleo-build-containers-rhel-8-master/dbbb97b/ : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":2},{"id":"34bf47c7e3181132de78eeec344eb9472b9e7ff0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-25 12:36:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a28c1cca8320d40168945def875b01b1383f4f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:48:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e6ecc60c0d724779ab96d3d8a276d984 : SUCCESS in 56m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/90883c5dc96f48cabf3dd9b1809304d8 : SUCCESS in 1h 02m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e1d05817924647c2941f2ecd3c8d9c2e : SUCCESS in 1h 26m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a07ab200bdc24305951b288bc947f758 : SUCCESS in 1h 32m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e3b87af1b09841daaa9ef2fed86c15f7 : SUCCESS in 1h 35m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2ac9877a38fb4541b11dd242e60b88f2 : SUCCESS in 50m 03s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/016f9b35d70c49e0a5e106dfa1588183 : SUCCESS in 1h 57m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f19edc1d92a34d248647a6ed70099717 : SUCCESS in 2h 09m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/54e10867d66a43afb3e5b4122f15a17b : SUCCESS in 28m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0999ab887cae492296ca363fbf397355 : SUCCESS in 27m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f62b05b6d944780b30682b6f5191127 : FAILURE in 22m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/993523c007db4161a7cb0b13ab774d5a : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959402a93d524a9cafc63411b6df2c4e : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fce8e47ad3914807bc48ddb97edde592 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49f46b3f65f54b2bac46e87b9316e162 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af698d26e1af4775a47bbeef062994ef : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80079a0f351340e8835963e3d9048d36 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb91bd075d7943a3a882f12e993eadb6 : SUCCESS in 30m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac5c0b8468fa4f629c894470c5a0cee9 : SUCCESS in 36m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8e1a9989b1c4978920795c9c1c7399c : SUCCESS in 1h 00m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a05b5083c95469782a3c6b38c7a703c : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea95cb03d70b4803b1ac5b1b23b2e5e4 : SUCCESS in 1h 08m 53s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/5551bb8282de4f209e5a2362ed6206c7 : SUCCESS in 1h 24m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2293903a292c49f791259e1499fd0ad8 : SUCCESS in 1h 42m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d477305a1f633a6507a1faedc2115ecb0c8fdd17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:48:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"71381f2bc0ab4885668dff381c4bb2b0a0047eec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:57:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/364d0982c3ca471692c4d3a983b4ab11 : SUCCESS in 1h 01m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/89fe6d2e24be46f2bd1aa69dccf9d7a1 : TIMED_OUT in 2h 01m 55s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d394bb3a733e414aa4e6abf70e2642fa : SUCCESS in 1h 27m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8dc2ab6a44424b71a7baf44c12cde2fa : SUCCESS in 59m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/609c90be802c4b9aac5e25dc8a5650f1 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06fcced649064258839d5e4a6dfc6f7e : SUCCESS in 8m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6f9bbfef46546dc8c5be2150e9d75c5 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70a645f8140a4e118059f655e2a8b288 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e41675e2857d4c098479aeda023ac822 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5610381a174546fbaa7c01d2ebc31e5d : SUCCESS in 7m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29bdec7520274a979022b0f216b53ebd : SUCCESS in 30m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d12e6af53774d69b354aa89a28617fb : SUCCESS in 57m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c90a87eb60074e73a987732622050f8d : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":2},{"id":"54cac1f93427996a10d020d77d24f71cb1305bc3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 17:08:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1aa1cf20e049b55a8bddc474290e22ade0ad3e4c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-25 17:47:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/691290/2/openstack-check/tripleo-build-containers-rhel-8-master/a8568c3/ : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":2},{"id":"493d4bc8bddf917d5f6aa65128def4bf37612fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:12:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2306e4a705f24f63b1165b6a9e95559d : SUCCESS in 1h 35m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/568aa22a61034228931147288341efcb : SUCCESS in 1h 46m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b8f4b487bc48464e8dd966c47a3fd6e1 : TIMED_OUT in 2h 01m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ba58a30cce644ced9ae596ad81e6770c : SUCCESS in 1h 12m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f2d41ae1f3ca46208d711530265b93ac : SUCCESS in 1h 40m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7b5cb82c05e7452fbae0c23b686b8574 : SUCCESS in 43m 57s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b7c2be39d00e4e70bfd331a1937eb99b : SUCCESS in 2h 04m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7d7bc09fd9f7430aaf387c8f0b3c2553 : TIMED_OUT in 2h 32m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9389bab2a58448c3bc70cecebe0d8228 : SUCCESS in 24m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/31832723899445098261b89992364ce7 : TIMED_OUT in 3h 01m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/479bad7a3f9846d8bbf09f83428b3d12 : SUCCESS in 51m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c6242ef8b78414494e8040e54db0f6f : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad3091a2e1cc4ce8a33f400e284bee0a : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5eba368cb1b04689aae21328b8ef806a : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cae139de93f4767aa09bd3cc856c07d : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d7fe2374d624d5cbf6ae45e40eb0115 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106f654575434aca8332ad1e32a5ccec : SUCCESS in 7m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4e53144d6de4cbba0f8b95e0299beb1 : SUCCESS in 35m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44676da7cd934c7c9c28161900f28ccc : SUCCESS in 35m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86f8a6139d5143e2b472305ea9f08fd6 : FAILURE in 1h 04m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/898f48cd8efb4ab5b09d8efb0332919c : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b721035abc1442c5b979ddbfebf17967 : TIMED_OUT in 2h 31m 32s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/657733d71ed043338914b8d236a73fee : SUCCESS in 1h 41m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0304c58b8cb648d1b18ccb81f974683e : FAILURE in 1h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ae28891cee1e6f13b6101cad9cba05b33d16968","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-26 07:00:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"876d92e5f2a90354f525c6be734801e7fd0967b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-26 07:42:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/90/691290/2/openstack-check/tripleo-build-containers-rhel-8-master/1117de4/ : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":2},{"id":"54d988134753f9bf6484812c5727fc07d726bdd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 09:36:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/285795eb96ea412cb96f952e3240ffcd : SUCCESS in 50m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fa046e76d4bf46f2ab494c7fb39e8a88 : SUCCESS in 59m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cf2c1ae81e3f478482a78bab00f4bd1d : SUCCESS in 1h 25m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7a319883fc82431bb35c329d07c5e796 : SUCCESS in 1h 04m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c753235f4b2549c2aa09c7b94706dc20 : SUCCESS in 2h 03m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/26041021979f4edfb4958dbb0459d0bd : SUCCESS in 54m 20s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ab0ce361eb234d01a3b52f49b44573ca : TIMED_OUT in 2h 32m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c69445d5e1f545358ca4925fd31b06d3 : SUCCESS in 1h 54m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c9aea88e3097462ba35436b9c6b92a38 : SUCCESS in 21m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1fadc7188cd74f9c9949b51e548832dc : SUCCESS in 17m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/785bacaf3b354f50ad533f291be92ef0 : SUCCESS in 53m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e8878ebf5764683b1c81f3a986aebd1 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7a1b9f7f2040608ddd3315a6360177 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/825c0058f8f7429eb88ce5f5e7cd6e5f : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71feeb6a10c34d3681dc7a166b523954 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c745316962144643a591bb745b30b693 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0fd3308c00470abe6d475a6b302dfd : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ac14885915f4dbabaa33f83066ccb1c : SUCCESS in 1h 12m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2aec22d407cf469bbeb8b15c70b44347 : SUCCESS in 25m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1caf439b0a9c43939f68423a8604f7a4 : SUCCESS in 1h 40m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3eb4ade66f24209ac42e776e019db58 : SUCCESS in 1h 00m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cd849f11732240d69e64c6b6244e8984 : SUCCESS in 1h 40m 55s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/a29d64ecf57f4cc391a7dcad4adc8f57 : SUCCESS in 1h 21m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5d02ec7ae47941cfae1b45ce4752589a : TIMED_OUT in 2h 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3109907ddccc63f935c5c5e2497b34e330ce943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 09:36:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f66e69083c079f82f35b32f93f6478f3d64ca2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 11:13:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/14485705d6ea477ea02720c59d68e418 : SUCCESS in 40m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/07577e6e8fed4847940f97b085711787 : SUCCESS in 1h 10m 10s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a727b754d454482bb20d9c36479a8357 : SUCCESS in 1h 36m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/733b131ba0c24cd18341a1ae62198fb8 : SUCCESS in 39m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3816651f83c94ae1a6bc46c9dd81c4b6 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d026dedd143241a5b8f6456f66c8777d : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df717016dbe744519a37db02f232c10d : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41e466461cc94481a6bb4f3b7b00534e : SUCCESS in 2m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a29445370f90453a84e9c7845fa1354a : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfffb34c2a8546449745cf43878645e0 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7428bc81e9a545eb81f2f4e39f079728 : SUCCESS in 23m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01058e8ba0674ef8bf7a16fe6ffb70a0 : SUCCESS in 50m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f552ef4460749a99fd819d330e8a01c : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e088f3747f151d565d8b3e223c2ab83e2c264bb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 11:13:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f48c7cf240e6d910e5b14cec90fbc9e668d9e8e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 11:15:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023232970cf6470fa4646091e5f33fd3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88f374c914c4d6709a1edf407a912c8cf566038b","revisions":{"f609d762d54550f755f967f6cb2f0725b29e07fa":{"kind":"REWORK","_number":1,"created":"2019-10-25 11:34:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/691290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/691290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/691290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/691290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/691290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/691290/1"}}},"commit":{"parents":[{"commit":"2e908723ac0ae1bf066a7c53a2bfd219fd3df475","subject":"Merge \"get \u0027apt_*\u0027 options of kolla-build.conf working again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e908723ac0ae1bf066a7c53a2bfd219fd3df475"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 11:33:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 11:33:01.000000000","tz":60},"subject":"Set pipefail in kolla_mariadb_backup.sh","message":"Set pipefail in kolla_mariadb_backup.sh\n\nWithout this, a failure in mariabackup would be ignored due to the pipe\nto gzip.\n\nChange-Id: I84536b302991e8fb04d5fa67ea266bc428437ce1\nRelated-Bug: #1843043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f609d762d54550f755f967f6cb2f0725b29e07fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f609d762d54550f755f967f6cb2f0725b29e07fa"}]},"branch":"refs/heads/master"},"88f374c914c4d6709a1edf407a912c8cf566038b":{"kind":"REWORK","_number":2,"created":"2019-10-25 11:35:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/691290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/691290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/691290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/691290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/691290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/691290/2"}}},"commit":{"parents":[{"commit":"2e908723ac0ae1bf066a7c53a2bfd219fd3df475","subject":"Merge \"get \u0027apt_*\u0027 options of kolla-build.conf working again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e908723ac0ae1bf066a7c53a2bfd219fd3df475"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 11:33:01.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 11:35:18.000000000","tz":60},"subject":"Set pipefail in kolla_mariadb_backup.sh","message":"Set pipefail in kolla_mariadb_backup.sh\n\nWithout this, a failure in mariabackup would be ignored due to the pipe\nto gzip.\n\nChange-Id: I84536b302991e8fb04d5fa67ea266bc428437ce1\nRelated-Bug: #1843043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f374c914c4d6709a1edf407a912c8cf566038b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f374c914c4d6709a1edf407a912c8cf566038b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
