)]}'
{"id":"openstack%2Fkolla~691300","triplet_id":"openstack%2Fkolla~stable%2Frocky~I40605679783649e9851d897d305ce6aeab1ecef2","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I40605679783649e9851d897d305ce6aeab1ecef2","subject":"Bump OpenStack versions for Rocky","status":"MERGED","created":"2019-10-25 12:57:34.000000000","updated":"2019-10-25 17:17:18.000000000","submitted":"2019-10-25 17:17:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691300-1572023838171-b61b701d","meta_rev_id":"51b812543e30cf58eb9b59585d54f39b84022b5c","_number":691300,"virtual_id_number":691300,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 17:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-25 12:58:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-25 12:58:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-25 12:58:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 12:58:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-25 12:58:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-25 17:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"735c6486261f5ec1e4983a337a27e88da9a809fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 12:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faeae305901a273ecdc18bc1277da74654de9be1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-25 12:58:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3d81daa8874a58c63a7b251ff44e5ca9ee08f49","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-25 12:58:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ed52721562e6f92d38ea7a294d6fdced1d15e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:05:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e7c6a08abb324212907fc9843daf1b6d : SUCCESS in 37m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1b8ac1d7803f40b5b2c386c1329d2270 : SUCCESS in 55m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1c7358bae4be406b8d6d7d0350b38480 : SUCCESS in 1h 35m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/21f76f1e63a14d6685e89abc7b176220 : SUCCESS in 1h 37m 29s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/59106ae0f7f34073900003e5ad833fb9 : SUCCESS in 1h 02m 00s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e99dc659829247038618978b9e00385a : SUCCESS in 1h 04m 26s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b6b47cd53a4b4770a29eb2f998df2355 : SUCCESS in 1h 34m 01s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/443abed5523e4e54a18bc05909be6e7a : SUCCESS in 1h 27m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28285c5d850e4f0eb113b5410d8f8041 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e3d52dc2a841159f1e0b182c9aca2e : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/faf762853cac4214900f27019a8899d4 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bc6b1a63c31b46f39ab74b42fe7e09c0 : SUCCESS in 3m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a81f7c8aae264c4182fb494b2f39cf65 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c603123e5fab45fbb144f8812268acca : SUCCESS in 1h 20m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5296b11991844ebca972ede456765507 : SUCCESS in 37m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d23f4d4e50f6437fbc5720df85bf151b : SUCCESS in 1h 04m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/179f8c1893474bc7aefb9da91c1e12ea : TIMED_OUT in 2h 01m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d3d07dc7cf44ddaa40de2fc8fb9dde93c71d954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:05:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d4b1aaf585c3b173a8645d76539808c6827c550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:17:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4a7a7510f5e344ac9fc421cbc127abe6 : SUCCESS in 1h 01m 08s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dbeb870130754a4dbf48e6be5dcb73cc : SUCCESS in 54m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35e8d280c68c438a99bfaec239e5381b : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1f03b988b540c5a6306869169a6d23 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c65a884ac6084989ad7e3705f8142a06 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/64b09eed1c42487c969b200f7880bf55 : SUCCESS in 4m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a5c39e6841764041bd8adea7a9febde9 : SUCCESS in 3m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f2e1501a74d4376b8b398761cd7de8c : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"51b812543e30cf58eb9b59585d54f39b84022b5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:17:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97e85ff2f6378ae922416bf16c6030af97697510","revisions":{"97e85ff2f6378ae922416bf16c6030af97697510":{"kind":"REWORK","_number":1,"created":"2019-10-25 12:57:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/691300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/691300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/691300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/691300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/691300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/691300/1"}}},"commit":{"parents":[{"commit":"4f639e85df24a8d6d17cd5c90db4fc977fbe40b9","subject":"CI: Increase job run attempts to 5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f639e85df24a8d6d17cd5c90db4fc977fbe40b9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 12:57:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 12:57:24.000000000","tz":60},"subject":"Bump OpenStack versions for Rocky","message":"Bump OpenStack versions for Rocky\n\nChange-Id: I40605679783649e9851d897d305ce6aeab1ecef2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
