)]}'
{"id":"openstack%2Fkolla~691512","triplet_id":"openstack%2Fkolla~master~I0302ee1dfc10a2b7df127ac6a7764a84b1d74a33","project":"openstack/kolla","branch":"master","topic":"bug/1850036","hashtags":[],"change_id":"I0302ee1dfc10a2b7df127ac6a7764a84b1d74a33","subject":"Remove /var/yum/cache from CentOS base image","status":"MERGED","created":"2019-10-27 11:43:57.000000000","updated":"2019-10-29 10:39:04.000000000","submitted":"2019-10-29 10:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691512-1572345474310-c91a571c","meta_rev_id":"43a751dcee879e0d3405cd0613678fc7eb96f707","_number":691512,"virtual_id_number":691512,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 10:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-29 07:45:18.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-28 17:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-28 12:27:02.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-28 19:44:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-28 17:09:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 12:27:02.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-28 17:50:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-28 19:44:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-29 07:45:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-29 10:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9f107ba164b40d02e2cf5845fa8849321911db7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-27 11:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d4fcbbe3883e783df47063ef2ba5bf7b56fb1ee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-27 12:21:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/12/691512/1/openstack-check/tripleo-build-containers-rhel-8-master/7347ea3/ : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4c724d7ceead9b16fca90ef441e5228d909cb0a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 13:28:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/78f31af072704f00bc5cf440abaf9548 : FAILURE in 5m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/faa7b31e261f4d1d9cf2edb639556a46 : FAILURE in 4m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/59236be20ee74d1c95553765ffec93fb : FAILURE in 9m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/006f5355733a4a4a9e6088d6ecee461d : FAILURE in 10m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/df9059098165480b95e249b8060350f9 : FAILURE in 10m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/074314276c084f7a98469ee187379f17 : SUCCESS in 57m 42s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/80b1c6515d30478fa20e4f7561399558 : FAILURE in 9m 18s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/e1a2be1e7ec94927b3c073568919d832 : FAILURE in 12m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b7073b34211a4d1585d6d156dc55dea3 : FAILURE in 11m 21s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/a211e2de248b46e3bc9e22505e764727 : FAILURE in 11m 52s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/75d38732831f467ca66a640eac9c4e25 : FAILURE in 11m 13s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/11883aa050c7464981357380a0121b77 : FAILURE in 9m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1778e290ee754499a412d75d5d33da0e : SUCCESS in 26m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ddbbaaa6dd3a4dab9fe2bf488bb9b323 : SUCCESS in 17m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d8a3688246f4ec896e3337abafc8b5d : SUCCESS in 57m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74964abf69c84669863047c9a994037c : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2e0304836b46a5a504f1310b736a87 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00291522b5a24c8882250e6fc26cadbb : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5872e1546047486c9bdc519dbb0996c6 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/487af0c752c545779eb6069c920da477 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00edddced1a44980832938e9960e99c1 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9383345e6aff4b888344cbbce5b78370 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/155eacd74c384ee7a398476c4c15bd9d : SUCCESS in 25m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba3fe67dcf5d40ac9e214a78ae17b86f : SUCCESS in 54m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37fda588b7f34487a4b57b1b98561d0b : SUCCESS in 57m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba64bc2c0a5e4faf867283cf33089454 : SUCCESS in 56m 52s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/9fe29ce340ef4a3893d30cbeaf2c816a : SUCCESS in 1h 07m 40s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/35ce289cd259447f9f0f09bc295d9487 : SUCCESS in 1h 20m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6d2ce425a6c64c63886a80535310acf9 : SUCCESS in 1h 37m 03s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/5db0a8917e7547c2a0357aeb8cf5a0c3 : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/21b93e38cb2f4a66a868d9554c22a4e3 : SUCCESS in 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"763cd33c9bd2b70696c91acaa18c7b8284541d70","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 12:17:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ba68cb4196b166cc60d70a42aada95afc485b83","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 12:17:19.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"cc1dc2e452804397091347f0cc37db629c94cb00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-28 12:23:02.000000000","message":"Patch Set 2:\n\nWhat is wrong with \u0027clean_package_cache\u0027 option then? \n\nEnabled by default.","accounts_in_message":[],"_revision_number":2},{"id":"11828d052a0decad111e9c8aef6327acdbe39c79","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-28 12:27:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"099bb4d8b60e0f00247b7644d2f9b2ff8b01a644","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 12:27:45.000000000","message":"Patch Set 2:\n\n\u003e What is wrong with \u0027clean_package_cache\u0027 option then?\n \u003e \n \u003e Enabled by default.\n\nThat will result in an unnecessary drop of cache + we still need these to be in one step to spare the extra layer with update differences. ;-)","accounts_in_message":[],"_revision_number":2},{"id":"4fe46b520cb10d95acabead04b0183ae25d24dad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-28 12:57:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/12/691512/2/openstack-check/tripleo-build-containers-rhel-8-master/f47ba37/ : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":2},{"id":"4fd9afb092bc2e061471afabed9e15512ecec6e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:57:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/105180fe9e1840758d408d5cfb520550 : FAILURE in 50m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/47636397caab496aa4d06d4058c191c4 : FAILURE in 57m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fe38e3853e8840b3b087f99fe8b80dbc : SUCCESS in 1h 34m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6d28e94ec56347fd8340825d84e0a454 : SUCCESS in 1h 25m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/069f9509bebc44d19ac31dd04eccadff : SUCCESS in 1h 41m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/15cb2ac2e61a4fcc96d353b1c7354903 : SUCCESS in 59m 58s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/4a69ebf4f8d04d2ebce4a74cab6e62ee : SUCCESS in 57m 04s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5b343bcfaa824c37936241d90a5dd9bd : SUCCESS in 2h 03m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/71f7acdbd3a9427694267f6f57eab416 : TIMED_OUT in 2h 33m 27s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7914d42343474846a5ba056117bd0ce9 : TIMED_OUT in 2h 02m 17s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/00d0e3ab0eaf43cbabf9139294b1b029 : TIMED_OUT in 2h 02m 20s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/76b58c9e03064c00a3e4a5a2f1d1131c : SUCCESS in 1h 47m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9733414baf0d4e16bda2b1f390e9e5f4 : SUCCESS in 30m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/789a85672d2640d6ac473a27eacf0417 : SUCCESS in 22m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a09415098c3d4c2eb3c2d3610fe7f3d9 : SUCCESS in 55m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83efb8c1fc4043b2ab4709a8bc6c138e : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a93203f72444898a420ac62f8a7877 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f26e98faba3943f08e4648f117a5b8f2 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24332fd74c6a4ea2991d4bc4f11ea5e3 : SUCCESS in 2m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/877dc1bd4c8b4a38ab0834982b06809a : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3304416b4747bfa88d228d616b66da : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5ecbf2290464c6c897985b3850dcfc2 : SUCCESS in 25m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78140bb99e704f4fb420e33147288410 : SUCCESS in 30m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24f1561272584494a7104ea1add71236 : SUCCESS in 52m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e3001c7950f441a99d99ef1134b71145 : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/baffbaca3f61406684743fe17326325b : SUCCESS in 1h 08m 01s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6adde99b0b2a452d967716ee26a474d6 : SUCCESS in 1h 10m 47s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/96c469b513334348878bc566937580c4 : SUCCESS in 1h 23m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8d0f2b7a25de4f699c1ddf6e39f905ca : SUCCESS in 1h 47m 51s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/69fd570a335e43288c118e5cc3a7b4e7 : SUCCESS in 1h 16m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/36b55d3be16043a8bd7e1cfe459883d3 : SUCCESS in 1h 04m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"594aa69515f9c0c592c3ec97cad2556251fe6467","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 17:09:28.000000000","message":"Patch Set 2: -Workflow Review-Priority+1\n\nlooks working","accounts_in_message":[],"_revision_number":2},{"id":"ee1917008b3184d14a1e6c4793fc32051e4f4b56","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-28 17:50:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f05f6015805d00fdea070414ce21face522037c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 19:44:45.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8fd471ac986e694b17d89a4a0471c69b8154717f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-28 20:24:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/12/691512/2/openstack-check/tripleo-build-containers-rhel-8-master/9194048/ : FAILURE in 38m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1cb1b00687bdc404b97a42397e4ac918c8875e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 22:20:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/22be78ed911c474a9531375646fc7a45 : FAILURE in 42m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0218da1b2d784b5a8d6ccd1361f79c64 : FAILURE in 4m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1e239171fa6d4876b06c58ec89fb725a : SUCCESS in 1h 25m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cb4e47bc988f4ab685c68b6330cf3dee : SUCCESS in 1h 27m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6cc0fc225c4a4fd9aa03b6aa2ef1110b : SUCCESS in 1h 38m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c9c48944795a49d2991a5ed9968585ca : SUCCESS in 58m 25s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/8b8948dac36e4195a407593576ed4711 : SUCCESS in 47m 22s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b7213ff5cdb148229eb07c2f86d4ec3a : SUCCESS in 2h 15m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/007251627549454583356444270a7906 : FAILURE in 1h 51m 14s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/da491eca1017434eb89bd9ab7e061735 : SUCCESS in 1h 56m 44s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5b6ff4d0f12442538f60dfce120c3184 : SUCCESS in 1h 58m 40s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ada73cd32d73495aa7710026d659d77a : SUCCESS in 1h 38m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/47552e261c2a4fc1b706e9dee6d8aaa2 : SUCCESS in 40m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/adf2c7d6d1a04a228762e2dea426128c : SUCCESS in 30m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/47ebfefa2eec489fb55174b3b1aa4dd8 : SUCCESS in 1h 23m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9f9585a61fe46449fd9e97b2b49d6e0 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1fc7eb83c4a45c2bcbba411d8e73fe0 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f26652e57d55459a94abb906aa84af19 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93e44475d88041ccb217e97c80b7c616 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/faf18f238c9c460db783219d9e27b261 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be82241a2f13419ba917cfd6fa1f928a : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb6b5e0dcf744646ad608f7fa28bc492 : FAILURE in 33m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2506a4513cc447b84d7d47d96efb69b : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/21ba578ac6d14644a617f0c88bd994d4 : SUCCESS in 58m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/86186d73fdcf494f8d0d619fda13e607 : SUCCESS in 55m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad3d47381ee747e3ab2af72b56141260 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/66467b36c7644cb9a628f0119c156c66 : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/038bf4e230ee48709aadedd31fe8d0a3 : SUCCESS in 1h 33m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/581a82d5bc7d4530829bc535ac5aca4b : SUCCESS in 1h 41m 18s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/67b9d56d7fa34bc2aa1a70200cf5589d : SUCCESS in 1h 28m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/83d4b6c21d4d428985481312200f9c17 : SUCCESS in 1h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c575b3b25c2d3e88254bd9e4ac2215c7e79cb7ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 07:07:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b0211994424d13c01d47ea878e72ab8b1ddbdee5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-29 07:45:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/12/691512/2/openstack-check/tripleo-build-containers-rhel-8-master/0a97f4c/ : FAILURE in 36m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0d82bcc9eccdf3982cc8f06f9ac1202f612e213b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 09:15:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/df7e0f352f0b45fc81db9037fa8b9ed0 : SUCCESS in 42m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8225554a209d496f9622cc01c7808b4a : SUCCESS in 1h 01m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/811c357465a8448ab16cb87bb8daf325 : SUCCESS in 1h 17m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9ffcbcd626f442b9814405dc55490f5d : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9d04520fd46847419e16c293333c2435 : SUCCESS in 1h 29m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4fd951c977d04a03a11c97020b8d3004 : SUCCESS in 49m 15s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/bbef4ecc786046c79bfca4a7f53934ac : SUCCESS in 50m 40s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/f2d7b4b98f6f4725ae06c1b32925052b : SUCCESS in 2h 01m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2264ff4a6e614ce4b30271924ec1ecb8 : SUCCESS in 2h 01m 02s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/60c1240518c240d3bd8a5dd55f6d5cd2 : SUCCESS in 1h 38m 51s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/75739840a6e048519dc760e772840751 : SUCCESS in 1h 45m 56s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5d410be8eb304f3fbb2c2c941b59c1b1 : SUCCESS in 1h 31m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b9281acb20dc4e73a9fa6f35349101be : SUCCESS in 23m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/56f8906210de4970bae9462b3b53bdf7 : SUCCESS in 25m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d2efe2b6bacb4be189ea9748fa04e0db : FAILURE in 19m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c5a67366ba742848d1168fc00dfff49 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d295b44e4164335bc1c951cee5751b0 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f96883fb5d234da293e9e7aed0cf1d6f : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3907c5f3e7454ddf9596ca1ba904ad2b : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3079f60b49b44cbb88094b8198fc9c7e : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7d93ec75344c23a40cf7a2a5c28412 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44aef554b78e466ab45fc2dcfff5ea02 : SUCCESS in 23m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/20384b696b4544619195057ff8bba011 : SUCCESS in 36m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc8ee2930a7c4a6c83b925fcd7e74020 : SUCCESS in 55m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/319399b2f2094c72abf4c8bd47cafa04 : SUCCESS in 53m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a42ab7dfc18d41e3b1e7bbc1c71adb09 : SUCCESS in 59m 28s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/121d92f641d64cd3b345de7e746a9fd9 : SUCCESS in 1h 18m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/98b0a4f3f5c8431b9e1a43d84b9b8cf5 : SUCCESS in 1h 32m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1fc123ae55c34ca99d9419db46c63788 : SUCCESS in 1h 38m 32s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b7925254c3d046cd8f1d30da7d9d5ed1 : SUCCESS in 1h 16m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3de0823c3f9846f99fcfa2bad591d3ca : SUCCESS in 1h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f97fbcba86bafa9369b8868fe73e9ad736868d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 09:15:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40ebc64c64effb429d2a361cca7d926451068ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 10:37:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d37190cf3f534f7aa2b9fce98ee52a25 : SUCCESS in 51m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/578dd6db862d4a0bb9d773416f77ce74 : SUCCESS in 1h 14m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8dae4d82cc8a46e59e6dfe2e0426afbc : SUCCESS in 1h 22m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/924824fdab264f5bbb98e1f7da9b40b4 : SUCCESS in 49m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a76f869970bc48be951081b78c069a5a : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/149dd58e3e9c4fdbacc7469aa9b6368f : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1e7a0d44b8446098eaa229b0f468642 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a458d7dc194022bd4ba86feb2fa81f : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c34dc951e02348ddb9f72cea5fe39fca : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1740466a77644d2684023d89a3ee68b2 : SUCCESS in 6m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4e39276fbc446f8a37f89d8aed71218 : SUCCESS in 33m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e38abb64986d476ca11abdd5783d3f03 : SUCCESS in 56m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9479e9c2599841be95598db5efcaee4e : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6ef77d321a68ad2fc1e30e141ce2f8f1cab7de39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 10:37:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"862368de5afe3edaecfd163665edb711bf159817","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 10:39:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e961776933f94513b11297ef51b177fe : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe00a87ec8e6257ad1724b342b70811c03117b58","revisions":{"b999346a1aa5a16f387e1bf8d67bf65f837dadd0":{"kind":"REWORK","_number":1,"created":"2019-10-27 11:43:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/691512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/691512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/691512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/691512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/691512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/691512/1"}}},"commit":{"parents":[{"commit":"2e908723ac0ae1bf066a7c53a2bfd219fd3df475","subject":"Merge \"get \u0027apt_*\u0027 options of kolla-build.conf working again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e908723ac0ae1bf066a7c53a2bfd219fd3df475"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-27 11:43:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-27 11:43:01.000000000","tz":0},"subject":"WIP: Remove /var/yum/cache from CentOS base image","message":"WIP: Remove /var/yum/cache from CentOS base image\n\nSince https://review.opendev.org/#/c/682600/, the following lines result\nin /var/yum/cache being present in one of the layers of the base image:\n\nRUN {{ macros.install_packages(base_centos_yum_repo_packages |\n        customizable(\"centos_yum_repo_packages\"), chain\u003dTrue,\n        clean\u003dFalse) }} \\\n{% for repo in base_centos_yum_repos_to_disable |\n    customizable(\u0027centos_yum_repos_to_disable\u0027) -%}\n        \u0026\u0026 yum-config-manager --disable {{ repo }} \\\n{% endfor -%}\n{% for key in base_centos_yum_repo_keys |\n    customizable(\u0027centos_yum_repo_keys\u0027) -%}\n        \u0026\u0026 rpm --import {{ key }} \\\n{% endfor -%}\n    \u0026\u0026 /bin/true\n\n{% block base_centos_repo_overrides_post_yum %}{% endblock %}\n\nRUN {{ macros.rpm_security_update(clean_package_cache) }}\n\nThis is because is it not cleaned in the first RUN statement. This adds\n76MB to the image. Also, using two separate RUNs which results in an\nextra 23MB due to updates to /var/lib/rpm/Packages.\n\nFound via https://github.com/wagoodman/dive.\n\nChange-Id: I0302ee1dfc10a2b7df127ac6a7764a84b1d74a33\nCloses-Bug: #1850036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b999346a1aa5a16f387e1bf8d67bf65f837dadd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b999346a1aa5a16f387e1bf8d67bf65f837dadd0"}]},"branch":"refs/heads/master"},"fe00a87ec8e6257ad1724b342b70811c03117b58":{"kind":"REWORK","_number":2,"created":"2019-10-28 12:17:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/691512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/691512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/691512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/691512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/691512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/691512/2"}}},"commit":{"parents":[{"commit":"2e908723ac0ae1bf066a7c53a2bfd219fd3df475","subject":"Merge \"get \u0027apt_*\u0027 options of kolla-build.conf working again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e908723ac0ae1bf066a7c53a2bfd219fd3df475"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-27 11:43:01.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-28 11:53:20.000000000","tz":60},"subject":"Remove /var/yum/cache from CentOS base image","message":"Remove /var/yum/cache from CentOS base image\n\nSince https://review.opendev.org/#/c/682600/, the following lines result\nin /var/yum/cache being present in one of the layers of the base image:\n\nRUN {{ macros.install_packages(base_centos_yum_repo_packages |\n        customizable(\"centos_yum_repo_packages\"), chain\u003dTrue,\n        clean\u003dFalse) }} \\\n{% for repo in base_centos_yum_repos_to_disable |\n    customizable(\u0027centos_yum_repos_to_disable\u0027) -%}\n        \u0026\u0026 yum-config-manager --disable {{ repo }} \\\n{% endfor -%}\n{% for key in base_centos_yum_repo_keys |\n    customizable(\u0027centos_yum_repo_keys\u0027) -%}\n        \u0026\u0026 rpm --import {{ key }} \\\n{% endfor -%}\n    \u0026\u0026 /bin/true\n\n{% block base_centos_repo_overrides_post_yum %}{% endblock %}\n\nRUN {{ macros.rpm_security_update(clean_package_cache) }}\n\nThis is because it is not cleaned in the first RUN statement. This adds\n76MB to the image. Also, using two separate RUNs which results in an\nextra 23MB due to updates to /var/lib/rpm/Packages.\n\nFound via https://github.com/wagoodman/dive.\n\nChange-Id: I0302ee1dfc10a2b7df127ac6a7764a84b1d74a33\nCloses-Bug: #1850036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe00a87ec8e6257ad1724b342b70811c03117b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe00a87ec8e6257ad1724b342b70811c03117b58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
