)]}'
{"id":"openstack%2Fkolla~691637","triplet_id":"openstack%2Fkolla~master~I5d2f200c440d0088f4a92f25da607862404ca16e","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I5d2f200c440d0088f4a92f25da607862404ca16e","subject":"opendaylight: enable it\u0027s repo only in opendaylight image","status":"ABANDONED","created":"2019-10-28 09:12:33.000000000","updated":"2019-10-28 18:22:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"61fcec567e47ec9aefe57e55c9ab8563092c0fc1","_number":691637,"virtual_id_number":691637,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-28 17:10:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-28 11:23:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-28 09:54:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 09:54:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 11:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 17:10:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"1464b588eade569d86ba8aa17e43d7847cc2aef4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-28 09:12:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e16ad8172839b509409d5f7f9c81e2a2e01d237c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-28 09:54:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/37/691637/1/openstack-check/tripleo-build-containers-rhel-8-master/b775722/ : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"90f0347a75fc25bf41b75f08407264365b41c9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:23:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3066b3ff19a947b0b8f6420fab6ea565 : FAILURE in 56m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d5541d36f67948f6b4a9451112e487e7 : FAILURE in 51m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/230ca2b927b24a0ca499cb4cdff4ed2e : SUCCESS in 1h 14m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a49b2953e0b44e0c8011dcd3647be3a2 : SUCCESS in 1h 28m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/61f7b58bcd104885ace7078e043dd835 : SUCCESS in 1h 28m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4c25e25beafa4b269cee437dd1c9d749 : SUCCESS in 59m 40s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/bc41757f55b64a9da7f8e5382aeb2a2f : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/2a8edc9046304cb4b746ccf4fa2ded33 : SUCCESS in 2h 03m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e9dd0ac5dfcf41a8a883561f18a5240f : SUCCESS in 2h 03m 41s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/90e320e4dede4dec95df82ddde8d7f1e : SUCCESS in 1h 57m 36s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/60c6677bc8cb4db6a27901947a343acb : TIMED_OUT in 2h 03m 12s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/06dd6b770ab4416d8def6855bc4038df : SUCCESS in 1h 53m 37s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/90da178459a647b1bc5ac66dc333e2ed : SUCCESS in 24m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b4bd5d0d997d4ad2908280447422350c : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e588ff3e0eab435fb098a8fbdb8d476d : FAILURE in 17m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca0f1a8c932b4b8ab4dbfc50341231ea : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9036ff1aa4c64e6eba17400b53911422 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07138f1c4a4c4667aeb2b61d9531488d : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4165d888b93e4d858ae8ac36df8bc5c0 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53f5e509a7824e918a9a416b54040d2b : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f63b73492e42e6b44b7798a43e1e62 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2aca468b2f6c4247b02f0ef119e277bd : SUCCESS in 27m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/226089a0d4f54fc488f592339131eb87 : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/337e73035f884d0f86f3b788d2627836 : SUCCESS in 52m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/63460c4bc59e4bd8b4e3fc79f70458b4 : SUCCESS in 53m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca6cc2ba6cd34e6ea1da5cedd58416a6 : SUCCESS in 58m 27s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/580a6115e6864ae79a487805390019af : SUCCESS in 1h 42m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2d83f82696644744bd0ae27af6e89e95 : SUCCESS in 1h 17m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/20fc8318fbd94f51a423a2c42f24c78e : SUCCESS in 1h 40m 49s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e180adee3da941a19e542b19410e0a2b : SUCCESS in 1h 10m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/74624e51cac04dbdbbf21a19a6dbac49 : SUCCESS in 1h 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d523b12e0f5439b527f2edefdb560ad04fa9b2cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 17:10:11.000000000","message":"Patch Set 1:\n\nnot helpful due to depending on all images","accounts_in_message":[],"_revision_number":1},{"id":"61fcec567e47ec9aefe57e55c9ab8563092c0fc1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-28 18:22:46.000000000","message":"Abandoned\n\nCBS works again","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5583f67d7f37070752e177bd88472c825ea89338","revisions":{"5583f67d7f37070752e177bd88472c825ea89338":{"kind":"REWORK","_number":1,"created":"2019-10-28 09:12:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/37/691637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/691637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/691637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/691637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/691637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/691637/1"}}},"commit":{"parents":[{"commit":"7b309f830034319e3f7db8efb424448e43ccae21","subject":"Merge \"Move to RDO Train packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b309f830034319e3f7db8efb424448e43ccae21"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-28 09:10:41.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-28 09:12:21.000000000","tz":60},"subject":"opendaylight: enable it\u0027s repo only in opendaylight image","message":"opendaylight: enable it\u0027s repo only in opendaylight image\n\nPackage is unsigned, fetched directly from cbs.centos.org.\n\nChange-Id: I5d2f200c440d0088f4a92f25da607862404ca16e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5583f67d7f37070752e177bd88472c825ea89338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5583f67d7f37070752e177bd88472c825ea89338"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
