)]}'
{"id":"openstack%2Fkolla~691742","triplet_id":"openstack%2Fkolla~stable%2Frocky~I508b8258e0d0aacfa0e4ff56706b5f273d2fbf0a","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1850044-stable/stein-stable/rocky","hashtags":[],"change_id":"I508b8258e0d0aacfa0e4ff56706b5f273d2fbf0a","subject":"Fix centos qdrouterd - use RDO qpid packages only","status":"ABANDONED","created":"2019-10-28 19:46:22.000000000","updated":"2019-11-04 12:26:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6d35ee1373ae531953a9109bf76cd9c9aada814","_number":691742,"virtual_id_number":691742,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-29 16:18:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-28 21:38:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-30 03:36:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-28 19:48:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 21:38:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-29 16:18:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-30 03:36:54.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"}],"messages":[{"id":"9cfa67c7ee6ebecf5696447ce7bf89ff132c22c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 19:46:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"27e2234ff418a84788a8004c1305833431877f04","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 19:48:55.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"757acf618b16817f998261674ddb3b8033f58896","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-28 19:51:33.000000000","message":"Patch Set 1:\n\n(please see my comment on stein backport)","accounts_in_message":[],"_revision_number":1},{"id":"ca9f495279340c27d2e5c2efa083daf8c9c16c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 21:38:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/723d4f73ed124a90ade0b08724161f7c : FAILURE in 38m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8375a7d9de5945e2b2df434ce37f16bc : FAILURE in 44m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2d4c1b1c0b734c3dad03bb4a8bafb97d : FAILURE in 42m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d885c6b44300426b8142a428f285544d : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/176b595fc45f4fc1a3d6d093d93b9eb5 : FAILURE in 39m 48s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/371a161cd8334b809c7ac99acf24e37b : FAILURE in 40m 00s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/79b967c6164f4c02ad147e9afca09012 : FAILURE in 41m 36s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d2a72b4ace8c46d9897491713dd90e12 : SUCCESS in 1h 19m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4383e9a2ef24a0fab83b2df8156adf8 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8b3aead5414ec8ac9c0aa0d5bc3a5c : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c88afdc7d51f41c4b77fa7edb8754727 : SUCCESS in 3m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/943c24e17ad44e49abb556f38bab9542 : SUCCESS in 3m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/98d9f91741574d32915ec73651bcb377 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/461b2fbb9d724122bfa136533139cad6 : FAILURE in 1h 23m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c483761e394846eea0957ea76fa3eced : FAILURE in 34m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc87c6450cc14507bb42e10a9753ee7e : FAILURE in 21m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37b49c6831c34da697c039814c84b132 : SUCCESS in 1h 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"392aff912ff26939c183a062fa90f49f5dbbb0c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 16:18:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f48b2b46c2961539be7cfbbcceb7cfe27df99d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:39:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/14a987d6a7d945948c5521c3d2aff71d : SUCCESS in 36m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/678f499c8453483db72005fdd8066639 : FAILURE in 50m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/10e95c7022454ab5a364b2c127bc916b : FAILURE in 44m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b33c6699aadd4428b1ad4b38dafbf425 : SUCCESS in 1h 13m 28s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/77cae23dbeee496a92b1c7f7bf23c90c : SUCCESS in 47m 34s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/194696e6143846c0a724728763038473 : FAILURE in 47m 13s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7ce8595f083041298f1b1f8e8f4e82ed : FAILURE in 41m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ac3f102df723498ebb05692c707f728e : SUCCESS in 1h 18m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f35f066dfed4418a8b68e03bbcc88f7 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5b2726810d44fd989045a30fa07e5e : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26cb0f3f4ed54f39ae5fc855b9a8d729 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/12cf74c8a16043079e9c70b716fbf9e8 : SUCCESS in 3m 49s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bef6764b2f2a40f7934dc9a7d8693f08 : SUCCESS in 3m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27534ca4e0cc463f9d499fb5c130b15a : SUCCESS in 1h 21m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d29c08b07d14981bee27a74d5daabc9 : FAILURE in 34m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c44aa014b30a40a78938303d4d63e52d : FAILURE in 26m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be47210d07064ffbb4c8330f57175dc2 : SUCCESS in 1h 58m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"774bcdc678409c954f93327cc6718db62acf5037","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2019-10-30 03:36:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e01bc4d7f904a38b3a86cb33a4223d7eb516246f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 12:26:41.000000000","message":"Abandoned\n\nnot needed, no epel present","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02571352b0481a18a4cd27ca111b5695bfa95796","revisions":{"02571352b0481a18a4cd27ca111b5695bfa95796":{"kind":"REWORK","_number":1,"created":"2019-10-28 19:46:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/42/691742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/691742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/691742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/691742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/691742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/691742/1"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-28 08:39:30.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-28 19:46:22.000000000","tz":0},"subject":"Fix centos qdrouterd - use RDO qpid packages only","message":"Fix centos qdrouterd - use RDO qpid packages only\n\nDisables other sources with qpid packages.\n\nChange-Id: I508b8258e0d0aacfa0e4ff56706b5f273d2fbf0a\nCloses-bug: #1850044\n(cherry picked from commit e41e86c805167b9ab2f5c6217f8f8ad5cf773f52)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02571352b0481a18a4cd27ca111b5695bfa95796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02571352b0481a18a4cd27ca111b5695bfa95796"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
