)]}'
{"id":"openstack%2Fkolla~691810","triplet_id":"openstack%2Fkolla~master~I45238b087647e733544561d333aa560f82e2a07c","project":"openstack/kolla","branch":"master","topic":"ci-remove-cruft","hashtags":[],"change_id":"I45238b087647e733544561d333aa560f82e2a07c","subject":"CI: clean up setup","status":"MERGED","created":"2019-10-29 07:22:45.000000000","updated":"2019-10-31 12:23:07.000000000","submitted":"2019-10-31 12:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691810-1572524483410-2bb4bfd8","meta_rev_id":"a8978f0204af5f52c81ed144c1c9773bc30861bd","_number":691810,"virtual_id_number":691810,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 12:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-29 12:35:00.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-31 10:43:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-31 10:43:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-29 09:14:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 12:35:00.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-31 10:43:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-31 12:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6db477a1bc2461585399694310c111518507b6cd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 07:22:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f994e38ec6048fe01d654bf8dab3130a6e1599a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 07:33:38.000000000","message":"Patch Set 1:\n\nI wonder whether the disablement of IPv6 in building gate was not the cause of odd behavior of images.","accounts_in_message":[],"_revision_number":1},{"id":"9caba1896551d45837e6bc11c7e867ceb918dec1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 08:17:40.000000000","message":"Patch Set 1: Code-Review+1\n\n+2 once pass zuul","accounts_in_message":[],"_revision_number":1},{"id":"78524b7239dd24a0524ccd78235c5669cf465fd5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 08:49:57.000000000","message":"Patch Set 1:\n\nNope, IPv6 still broken in those images.","accounts_in_message":[],"_revision_number":1},{"id":"3f85ab49af97b45a07a05fe363fd0f2f18a16edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 09:01:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1b565b7f9ea3445b81d9ad2a2d4a57a1 : SUCCESS in 48m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d5de9dad802c40fabcdb8c95a2d4ca6f : SUCCESS in 51m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9eacb5646c7641b9a2126108e70a6bb1 : SUCCESS in 1h 33m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/841f0c49a0d743f4ba0e6d9ee9569ce5 : SUCCESS in 1h 24m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/35e5f1b1a5244264b0443be1f478e1f1 : SUCCESS in 1h 34m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4496c864cc8b40b3abd3f7e9127e4c65 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e93e7a106dd4ef1af6c46276340cea5 : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8725decd357e400683ae3094f7e4caf7 : FAILURE in 17m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/517be5a0059f4f1b99a19f3442552add : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc615b6cac140088efe1d0e9d790489 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d0d29b86a5f458aafb6e6ca1b37cafc : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/244921c7c30f4d55b938402208f24917 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59ae521a5e1f45ad8fff5ccaa0092244 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a002a0f3e79f4a98872326a929ecfcef : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/950f47adcae04a4680b4dba88f4140d9 : SUCCESS in 22m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/229db6f742384ea1a631bc44a67a510c : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae4a37fed9b349889542824e641b4319 : SUCCESS in 53m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c1b54cc4e074c609956d5cfcac4ff6d : SUCCESS in 1h 00m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c3aa73c88e0443dbb730eea5147430ff : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2864511bb5934d918d37dea24b4c9f98b6806222","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 09:14:18.000000000","message":"Patch Set 1: Review-Priority+1\n\nMaybe it helps with docker breakage","accounts_in_message":[],"_revision_number":1},{"id":"6080098d782b083d6984af01def5cd9852935654","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 12:35:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4793d386a3d913fc3622daf1e48521f61bb8f70","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 10:26:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b7ceea7a040ea063bc6a61fad260343908a6c6a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-31 10:34:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93d153e3ce516aaa1e47cfba561a7c9c46e139f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 10:43:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80ae1ead1984932b16732e11d3fc9ce1c3382dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:43:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2735ff1b357ef2bbb1dbd86593c281d8c874922","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:21:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5267e63c135f49c6b790bda24f8c66e9 : SUCCESS in 54m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/eb61a465aa1e49d6a3ca3ddfd19d6047 : SUCCESS in 1h 21m 48s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0ad36865eb53431b8cf8713810ba0177 : SUCCESS in 1h 32m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/431bd4208d53410584d3aecfa79f9d9f : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2adab96f2384173826b7199c6ed80a6 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2365c87a7ae84e94bdfc9f1c750d9817 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16c1b45c07714bdc876fa0776affe056 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dd0dde4d0ff403c8c309fb7a2f8e395 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a0c5fa237d42a2a04012589d426ec5 : SUCCESS in 7m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/228fb43c7401443081d509234b1bf5e9 : SUCCESS in 37m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e6772ea10c04dc08f5cacf59a64c1fb : SUCCESS in 1h 01m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ddc5acbf8a34b99a903e195398eea16 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"94f03258a66844544a7de590591afe3431585cd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:21:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59ef6f04c6aafd00b0eb847562ef2c63195c11ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:23:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aae39cc30f444a092fcb7cd019d9236 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff6e229ed4526ec7cf4f2e333c6ba93e1f716a51","revisions":{"ff6e229ed4526ec7cf4f2e333c6ba93e1f716a51":{"kind":"REWORK","_number":1,"created":"2019-10-29 07:22:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/10/691810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/691810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/691810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/691810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/691810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/691810/1"}}},"commit":{"parents":[{"commit":"7b309f830034319e3f7db8efb424448e43ccae21","subject":"Merge \"Move to RDO Train packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b309f830034319e3f7db8efb424448e43ccae21"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-29 07:20:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-29 07:20:19.000000000","tz":60},"subject":"CI: clean up setup","message":"CI: clean up setup\n\nRemoves stuff which does not belong here, along with the docker\ndaemon systemd service overrides.\n\nAlso re-enables IPv6 on CentOS node.\n\nChange-Id: I45238b087647e733544561d333aa560f82e2a07c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff6e229ed4526ec7cf4f2e333c6ba93e1f716a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff6e229ed4526ec7cf4f2e333c6ba93e1f716a51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
