)]}'
{"id":"openstack%2Fkolla~691914","triplet_id":"openstack%2Fkolla~master~I2c249633cf07479ab808a3aa261ab683af8f9abe","project":"openstack/kolla","branch":"master","topic":"mariadb-centos-upstream","hashtags":[],"change_id":"I2c249633cf07479ab808a3aa261ab683af8f9abe","subject":"WIP: Upstream MariaDB 10.3 on CentOS","status":"ABANDONED","created":"2019-10-29 16:50:07.000000000","updated":"2019-11-02 11:14:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8520b126d71968eeaec6a19ec87d92ef8b63f71a","_number":691914,"virtual_id_number":691914,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-29 19:56:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-29 22:36:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-29 20:09:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 20:09:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-29 22:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca704ad4bf91db6991ca80e66de02be279e360fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 16:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"859d49cb607e51358cb9e147b30eb7c5c4c6b1ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 18:40:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6bac3cc7a8cf1efea39edcc13316ec59e53a9873","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 19:56:17.000000000","message":"Patch Set 2:\n\nit fixes ipv6 for galera","accounts_in_message":[],"_revision_number":2},{"id":"733ab46718e046a9aa49198bb7f272219138c1a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-29 20:09:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/14/691914/2/openstack-check/tripleo-build-containers-rhel-8-master/c420b97/ : FAILURE in 42m 04s","accounts_in_message":[],"_revision_number":2},{"id":"2918af4408aa7a1b32b3484d189f11b2d210e46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:36:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bce69c576ab740ccab192c6b24a32863 : SUCCESS in 43m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cb6c7e0db4054a7dae4785a13b7c8e05 : SUCCESS in 1h 04m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b89616ee4e72416cba45d038cf9325e4 : SUCCESS in 1h 31m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b91fe6d751c9481883be47ead1807913 : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0733fe9d3a584821833687407df61afe : SUCCESS in 1h 19m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ce1bd439e96a4887819184e43cbbd171 : FAILURE in 55m 49s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a2758cd5377747e881819f8a1f4e2858 : SUCCESS in 45m 28s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/198e2da880464ceebcf91f611d64c067 : SUCCESS in 1h 52m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/14f9194e1ff14f00b2fc6ce244ceb265 : SUCCESS in 2h 01m 16s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/6dbd4a7427894286aa64f683fd69caa9 : SUCCESS in 1h 41m 27s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d80ec722001d4b248097b94f92194f34 : SUCCESS in 2h 00m 17s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/a86868e22516413695c264e2eb9d1913 : SUCCESS in 1h 38m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e524c688d65346cba2d1ed81b0e9360a : SUCCESS in 26m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7f5f3f2774ea44bab8c449785e837f65 : SUCCESS in 18m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3bdcdffac106480c816c2e80248aa1d1 : FAILURE in 19m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ea63a882aaf41f393e8e2f1307cbd6f : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40dccba6f517414bb17568aad4cce30a : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d4c2fe7918c491190f005c97c300b92 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2162bc9a90543e6b9896b6d0cf319a8 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d6d7b6f7d6b426f8812c599a6eabed4 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2506d88427c34e2ba7b478dbd73a2358 : SUCCESS in 8m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4fc43248ee344321977eef885d589ee5 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc5653b3aa484c54840ece3f47f75182 : SUCCESS in 33m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fef27c41754548e39639bbc3a26d0699 : SUCCESS in 58m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/298bfa736103464f887006aabd61abd2 : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/586d209d5cb14f638976fa31c0b99759 : SUCCESS in 1h 06m 24s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b7b0aa93d17b49128e0eb6494cc5a667 : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/3ec4213ac3eb4a7190ca9630ad86d07a : SUCCESS in 1h 50m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e88e08982c7f4e4283aeac09b05297eb : FAILURE in 2h 09m 57s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7d215c8c6fad4803ac485362a2a84cb1 : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/046ef561817d4bb1a3652a1e1fcfad07 : SUCCESS in 1h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"729598262302ee330c7c19fea52586f9a56d06ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-02 11:14:03.000000000","message":"Abandoned\n\nnot worth if rabbitmq still fails","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06c55a8ba2acf80ec68868015462020578c1a72b","revisions":{"a405decbc33c263f5d682d0a83ba78e85be8d2ee":{"kind":"REWORK","_number":1,"created":"2019-10-29 16:50:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/14/691914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/691914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/691914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/691914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/691914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/691914/1"}}},"commit":{"parents":[{"commit":"912ccf0658e88f5f8cea69c93a4ff1a830149e44","subject":"Merge \"Remove /var/yum/cache from CentOS base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/912ccf0658e88f5f8cea69c93a4ff1a830149e44"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-29 16:49:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-29 16:49:23.000000000","tz":60},"subject":"WIP: Upstream MariaDB 10.3 on CentOS","message":"WIP: Upstream MariaDB 10.3 on CentOS\n\nChange-Id: I2c249633cf07479ab808a3aa261ab683af8f9abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a405decbc33c263f5d682d0a83ba78e85be8d2ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a405decbc33c263f5d682d0a83ba78e85be8d2ee"}]},"branch":"refs/heads/master"},"06c55a8ba2acf80ec68868015462020578c1a72b":{"kind":"REWORK","_number":2,"created":"2019-10-29 18:40:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/14/691914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/691914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/691914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/691914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/691914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/691914/2"}}},"commit":{"parents":[{"commit":"912ccf0658e88f5f8cea69c93a4ff1a830149e44","subject":"Merge \"Remove /var/yum/cache from CentOS base image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/912ccf0658e88f5f8cea69c93a4ff1a830149e44"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-29 16:49:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-29 18:39:34.000000000","tz":60},"subject":"WIP: Upstream MariaDB 10.3 on CentOS","message":"WIP: Upstream MariaDB 10.3 on CentOS\n\nChange-Id: I2c249633cf07479ab808a3aa261ab683af8f9abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06c55a8ba2acf80ec68868015462020578c1a72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06c55a8ba2acf80ec68868015462020578c1a72b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
