)]}'
{"id":"openstack%2Fkolla~691926","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ib290fb2ec80f408c9d1e134595a5bb7ace922952","project":"openstack/kolla","branch":"stable/rocky","topic":"691926","hashtags":[],"change_id":"Ib290fb2ec80f408c9d1e134595a5bb7ace922952","subject":"Fix CI failures (ODL and searchlight)","status":"MERGED","created":"2019-10-29 17:33:43.000000000","updated":"2019-11-03 18:08:30.000000000","submitted":"2019-11-03 18:08:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691926-1572804510423-4445955c","meta_rev_id":"41e7d42e2296578b1d9eb0a75abeced37f7cff3e","_number":691926,"virtual_id_number":691926,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-03 18:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-03 17:00:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-03 17:00:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-11-01 16:19:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-11-01 19:24:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 19:24:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-03 17:00:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-03 18:08:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6118afd1de70dcee322be3408198798f627a9ee","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 17:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70bb30ba24b7cc1e3e70377a4916cad26154bb28","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 17:34:03.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"3b72dc9a40817186014ee2819ef7434f2f1fd1f0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 17:35:21.000000000","message":"Patch Set 1:\n\nbug: https://bugs.launchpad.net/kolla/+bug/1850479\n\nand needs to be merged with the qdrouterd patch or else both fail","accounts_in_message":[],"_revision_number":1},{"id":"effb9ef5aafb48507e5b8f3d235ed7bdebd5d47e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 17:36:17.000000000","message":"Patch Set 1:\n\nI also think it might be a little abrupt to mark it unbuildable - can we use the only release tag and be done?","accounts_in_message":[],"_revision_number":1},{"id":"1081467df2e22596c566e78cc365bd110bf074e1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 17:44:38.000000000","message":"Patch Set 1:\n\n\u003e I also think it might be a little abrupt to mark it unbuildable -\n \u003e can we use the only release tag and be done?\n\nIt was not released. Who knows what\u0027s there.","accounts_in_message":[],"_revision_number":1},{"id":"eb233ab13062ad7761f0cdb21388a0929bf1151d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 18:26:29.000000000","message":"Patch Set 1:\n\nbugs for sure :D","accounts_in_message":[],"_revision_number":1},{"id":"e66892c40333fd1af89005c421e5845f5b2ebbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 19:45:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c5175a137e244c32abff6bd680738398 : SUCCESS in 43m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/af234aeb76434daeb9405ced6a93ef10 : FAILURE in 55m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cc57f5a548e84cfd9ddb0d9e2d4e0188 : FAILURE in 39m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/099a72b1e1f14f4fa5d99ececeb08735 : SUCCESS in 1h 29m 03s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d5355b2ffaca4436b21429420ee09517 : SUCCESS in 59m 26s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/95ed3cb807054bd6be1b30a374b86366 : FAILURE in 46m 05s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/de23fc5eecdd4e29a84395728841eb30 : FAILURE in 47m 06s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9cd7b16da6e34804b5a88cc712013ae1 : SUCCESS in 1h 23m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e1472c3f8de439ba3ceecaf201ecacc : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b99a4988b247158e68cf3878c8394d : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4348633234604dc886354c9fedcc06f0 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/35c99872db0c41ea9270cd94ade7d9ab : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b8bfd5d97ba34c2ca865d0401f22ce08 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06c7b8b7d3034fe39a652a7762b7b059 : SUCCESS in 1h 24m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9011507259734cf2af06e86331f893da : FAILURE in 40m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d77298e4efa34736b663a079dfb7aea3 : FAILURE in 24m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/40bcb37483dd4dfe931f4d800a1c9b51 : SUCCESS in 1h 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f0fe887518a1385702fc89131fc3e4ee65bd43a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-30 18:45:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a04c5b2fd55e8295b21ce31592ea97434ed3a5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 20:34:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/40f5b75aff0f47b3a263463023551537 : FAILURE in 8m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d3afa9b69e2a4ab9b728d83169a9d1d8 : FAILURE in 8m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a6e7c2402fa04e798ce10094784d9989 : FAILURE in 11m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7f932cf98db94fe4b4404ad3ad773e48 : FAILURE in 9m 05s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4d439cffe30d47df83b159321cba60a1 : FAILURE in 7m 17s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a2c9e266f83a41379bb1940a316c27d1 : FAILURE in 6m 01s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7b54c9e3ef28417186c3dfbea3f82542 : FAILURE in 13m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c42d285735854958be1d7262b8049517 : FAILURE in 10m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13f357e7e0594b8d85fc691bfe3bbffa : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae3b63a22d44394ac00a7196b368aa8 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e17d82b7eba04e7fae156ec4caeb2891 : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d003e93809e43bcbbcb4478d659e91c : SUCCESS in 3m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9bca32e5ad2b42429f2f87e9a952e397 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18214fdc8ca546df9e0a9146a7389dd9 : SUCCESS in 1h 17m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/54ebb12bcbd943758c48293c66016e07 : SUCCESS in 52m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd4bb7d3728646f7a90b86f708267092 : SUCCESS in 58m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/397f472c57c0499aa7c40a41ef8d36e3 : SUCCESS in 1h 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24f3d93a49c19baf265d0db2fc721d96fd25b786","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-31 09:22:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f447d3a1fc59b869c5310993094257cad9695b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:44:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7b1cab3239ed4fd1bb49a1212f3d8ccd : SUCCESS in 49m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e6c11a65c0764446b4f90db8741854f5 : SUCCESS in 56m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9eef8e6255834a5db862683bf5875c2a : TIMED_OUT in 2h 02m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/962111b62dd0426fa07adaa40974a8f2 : TIMED_OUT in 2h 01m 57s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3a0a94d7693f40ed83158f755d15ec08 : SUCCESS in 1h 00m 54s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a5c62f8936e14f7db82494ed579d0672 : FAILURE in 1h 06m 41s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ee5e4f67a0fd4fee88f92d90f287ef78 : SUCCESS in 1h 17m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4593d67ec45b48e69b7f2bf0717c0dce : TIMED_OUT in 2h 03m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b8f8f13caea4c258791df68abc8659e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8691b52fd645239bc75741f15b49d7 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3df8104c12234286af9364ba1d13f107 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9f56b5577bd944f99eb2c0ca9ffd7b0e : SUCCESS in 3m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8a2130f9aee04742ba34d643b40dbe81 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/296b0f8dc88c485d8e9e5e14aa51d0b6 : SUCCESS in 1h 22m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0a325f9bbbc4feb809920c9394fad0c : SUCCESS in 49m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5800b152068b4d3c912b7c60b3b814b6 : SUCCESS in 1h 00m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/909a86c417e84ab7a1f0224dbc08d8da : SUCCESS in 1h 58m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30925e922db763de82709743a31649e06d1170ba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-31 12:21:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"519d3b1fa44adc121a595ead3e9615261fe4f7bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 14:35:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/59f1bd5d68cf456893929cc3114c037a : SUCCESS in 42m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5d91aa780fd0486297b3d0043a5d7073 : SUCCESS in 49m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6a093bf6a44f4e2f8f2480a5b9fcbebe : SUCCESS in 1h 22m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4c713550a56f41c294a77c0d03a41c30 : SUCCESS in 1h 18m 01s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0d249263c2c048b0af09616e9577988c : SUCCESS in 54m 44s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/54a3bc94089c4006beee14af954e666b : FAILURE in 44m 09s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/57833af08a8345e6bf0369b7f4e102fd : SUCCESS in 1h 32m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9bf7d2e588014fc098c7f837730131a6 : SUCCESS in 1h 22m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8bc33da760f44f1a61f8af7dafa9157 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2af1c6079c42fe9b1e382c94b6b709 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92b4b6ac294d49b7a74cb4925bf18a47 : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/debc506ce9fc4991b26c2174d7719bcf : SUCCESS in 3m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a3c1e46dba640d6ae1ea1d351483e84 : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6913365d144d42de92dc6a1953805ae3 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e3930886d2044b09c541ccbfea4b359 : SUCCESS in 34m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11e5235d3baf436f87889aa75ed94a08 : SUCCESS in 59m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/500444c6e3ad4e97a73a2a1178587077 : TIMED_OUT in 2h 03m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"012b214a2e23e7c1e88b8ecd95912140e1ae7148","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-31 15:53:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8478b14455a10b9f6b92e7e012141692dde25206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 17:42:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1a1d4e05da8b43fe8ee0d7d31f25f213 : SUCCESS in 42m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9f3fb2038c634103b06f8a4a8ad048a7 : SUCCESS in 53m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f431979e99cc459cac0ecac0f6b21971 : SUCCESS in 1h 19m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7ba859fc78c9445db6b0329de284c821 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0cc8d264dad44b0fb411cb7663e9cc83 : SUCCESS in 50m 34s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/87cd9dab5d8742a6895627567bb4e205 : SUCCESS in 52m 05s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/644fa62149414452a5b711d3f91cf023 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/e041d065be0b46d4bd37523c9b41f6b4 : SUCCESS in 1h 29m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1296e047dceb40ffb2c41835fd604720 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71e2a491d9b45249c9d0e6c4e654046 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f4cf670d2640caa5746047fcaa5cc3 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72aea9f599284557b350b3cf897ab7cb : SUCCESS in 3m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0799b850f98a452e8df24687c2134b90 : SUCCESS in 3m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/805608c2910647939af3c78fc3a9989d : SUCCESS in 1h 14m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9dfe5cd6eb6d4fd4971f9806a48f4ae5 : SUCCESS in 28m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7eacaa68d404e3a962db4cdf5eb36e8 : SUCCESS in 49m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49c14663fb12406e852fd646465c986e : SUCCESS in 1h 46m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4d27fd0de69583255d68117cebd8d5c0099d6615","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-31 17:45:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6bf21f8f4602a5bb6acdd5bb4a16ff6ea7108502","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:25:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"664ed3818110f4e38d22475c02014130a127a467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:02:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/37d207128acf4373b453fb8f4534ad19 : SUCCESS in 51m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cc868f7c6cf84d52a3878ca4887a184e : SUCCESS in 44m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fb27a8dbfc734722a568f765bf4dd24d : SUCCESS in 1h 33m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/16faf3060bbf4a2d98b61bd67bbc7bab : SUCCESS in 1h 30m 31s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a09b918201e745338438fd65fdd304d0 : SUCCESS in 38m 18s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/91d0456ab30b4946a1b1f0416a02682d : SUCCESS in 40m 24s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b83ba3e04be3480c8d5da549e9901de2 : SUCCESS in 1h 25m 33s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a361603c70884e18b2527296c933eab6 : SUCCESS in 1h 23m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f867f5eca3848da8cff00cc3c86f227 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d64245bf27d4f0494ade788ee375682 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae2ca85b6cd44838ab0012f0955b47a9 : SUCCESS in 3m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8cc53e4fa09c41e495b08f3dd1e770a5 : SUCCESS in 4m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8706ee1707b742669a2b07dbabc01e3f : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/694db8f56d7a4d3e82effb6d179ce5d8 : SUCCESS in 1h 18m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b90aa97f7f8743ca8bbdcf7b8c5e0830 : SUCCESS in 47m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07802e69af7e47dda64176e1b3a41f53 : FAILURE in 4m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5808086f30b40caba82acfc4bf3aa7c : FAILURE in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"396713769d6889142d71fd3220b85a4529de71ab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 16:19:21.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"072b74271c39e2eee733ef5c0cfc2c74b086c05b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 18:10:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c598c68f72cc4b8bbd6304e369915c4b : SUCCESS in 52m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/45f4fecfed8f4af1b1634509a210f880 : SUCCESS in 43m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca5c9a13bdf14c52af99c292e70332f8 : SUCCESS in 1h 30m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ac706d3d0a6046eea40b848ceb5e9f29 : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1b6ee3a4c5c54cac9485d73e2b16d26c : SUCCESS in 39m 56s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5206f8e872384916ae197959653e1dbb : SUCCESS in 48m 02s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5338f69be233470892ce33d41b688956 : SUCCESS in 1h 28m 55s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/dd833a9d47d14ca08b51708de4ab62e9 : SUCCESS in 1h 17m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4c9d49579af4a12bac650cd10c7b02e : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a48ca2dd8241e98df5e08a66b5cc9f : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54a83d41f1844a4c9090ae56b33f5048 : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/048a1577bb544bfe9dadc66748da5ac0 : SUCCESS in 4m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/70ae069b2d544147942bf407c42b0b99 : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd3e4582f8cd4b5892ba322ae6fa3f50 : SUCCESS in 1h 34m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4a2f849af0741068661d268dd9adbb4 : SUCCESS in 35m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/49854297c9d74b208e5d29d9ec9b0940 : POST_FAILURE in 40m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbbe8d1b77e14e4ab63ce73e3fbca853 : SUCCESS in 1h 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2c0b7d29657ad38cad3501f978d8582f9464647e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 19:24:07.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ac3fb78290db88d013c045927f76767bc5c6cf53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 21:16:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6e4cd6fc2e90471297dc271839e6245c : SUCCESS in 43m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/70dc5ea9cc364b53a07a2da53b0ac2fb : SUCCESS in 52m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/96be22cf34664931a281c4b68dfec836 : SUCCESS in 1h 28m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ea43b0a5f9684aceb02a6c606be9ebe3 : SUCCESS in 1h 14m 25s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a3ac3c9531a84479b8b114c3bbaab108 : SUCCESS in 44m 23s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/077de2f5814d4d9894cf6b4f29e8be23 : SUCCESS in 1h 00m 46s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/04c7322c03d04b17a611618633e42b76 : SUCCESS in 1h 16m 58s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b033569ee3924f0db6e8156af721e07b : SUCCESS in 1h 22m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33e9c1de54f642afb4a6054e83b0f133 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4df35bb45b4d12b7978306f074d58e : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bbe8fef36bd4d549c3981e2ddb9f9f0 : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db16c8766f5b4858a7aa12008fa448d4 : SUCCESS in 3m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8d59969bbab146d197b4c3a6ba5a9bd9 : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38eb9bfd100c4560a69af1580ce2b89f : SUCCESS in 1h 19m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/077a6dd38df54b93ad7265672afbac29 : SUCCESS in 31m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/61944aec97a644f792b544608ad7cd2c : SUCCESS in 1h 06m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f9ba7efa84a458386c057579f142aec : SUCCESS in 1h 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3f8975128df39ebed87975bf431360a3322b9a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-03 17:00:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"da5c21b21c9702b336854c35c0e717745ed41e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 17:00:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"01c5c7ff23e3780a6ed1f90c662601751f0c93c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 18:08:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/54f5552b8c7b4535aeb479bbb5a7d26f : SUCCESS in 1h 06m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c35feadbfe664988b2987f0205605c1c : SUCCESS in 46m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f52fb298057f4e62a0f2871e0e244758 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0096c4dd4b4f21b958555619499613 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6eab4f09c10b40b2b1046aa836f92fb0 : SUCCESS in 3m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8008e9cbe0884eb9ab02b551165ba7fe : SUCCESS in 3m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0e30c8050b5b4d239e03d70c80f74895 : SUCCESS in 3m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0cc2985e52434292b02814a349a1c930 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":6},{"id":"ed66e9eef9ff5f0c4029d9003a9d9237f3e82f46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 18:08:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"356096d0d8949bf23444b9873dbd6f96e3a0e32c","revisions":{"1b8c264aa19db82ed63a684d7457c4819b32686e":{"kind":"REWORK","_number":1,"created":"2019-10-29 17:33:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/26/691926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/691926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/691926/1"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"subject":"searchlight: add to SKIPPED_IMAGES due to CI failures","message":"searchlight: add to SKIPPED_IMAGES due to CI failures\n\nSearchlight did not had release during Rocky cycle. So we left using\n\u0027master\u0027 tarball. Now Python 2.7 support got removed in Ussuri and our\nCI builds fail.\n\nChange-Id: Ib290fb2ec80f408c9d1e134595a5bb7ace922952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b8c264aa19db82ed63a684d7457c4819b32686e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b8c264aa19db82ed63a684d7457c4819b32686e"}]},"branch":"refs/heads/stable/rocky"},"ae861422e3ea426f0a7c3cc601d7927080c97f26":{"kind":"REWORK","_number":2,"created":"2019-10-30 18:45:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/26/691926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/691926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/691926/2"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-30 18:44:33.000000000","tz":60},"subject":"searchlight: do not try to build it","message":"searchlight: do not try to build it\n\nSearchlight did not had release during Rocky cycle. So we left using\n\u0027master\u0027 tarball. Now Python 2.7 support got removed in Ussuri and our\nCI builds fail.\n\nSo searchlight is added into SKIPPED_IMAGES and it\u0027s plugin is removed\nfrom horizon image.\n\nChange-Id: Ib290fb2ec80f408c9d1e134595a5bb7ace922952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae861422e3ea426f0a7c3cc601d7927080c97f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae861422e3ea426f0a7c3cc601d7927080c97f26"}]},"branch":"refs/heads/stable/rocky"},"9cf48143a10e51e0bbb895140bf739f0cb383a6f":{"kind":"REWORK","_number":3,"created":"2019-10-31 09:22:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/26/691926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/691926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/691926/3"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-31 09:22:09.000000000","tz":60},"subject":"Fix CI failures (ODL and searchlight)","message":"Fix CI failures (ODL and searchlight)\n\nsearchlight: do not try to build it\n\nSearchlight did not had release during Rocky cycle. So we left using\n\u0027master\u0027 tarball. Now Python 2.7 support got removed in Ussuri and our\nCI builds fail.\n\nSo searchlight is added into SKIPPED_IMAGES and it\u0027s plugin is removed\nfrom horizon image.\n\nChange ODL repo to nexus.opendaylight.org\n\n(cherry picked from commit 939a56035b8f15b74dff2d68154963735b0f73b4)\n(cherry picked from commit 5a08221cc4270dec3af5ff613731cc7be7576124)\n\nChange-Id: Ib290fb2ec80f408c9d1e134595a5bb7ace922952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cf48143a10e51e0bbb895140bf739f0cb383a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cf48143a10e51e0bbb895140bf739f0cb383a6f"}]},"branch":"refs/heads/stable/rocky"},"4f269da3330044b1410d7829fa61be4f8a8b172b":{"kind":"REWORK","_number":4,"created":"2019-10-31 15:53:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/26/691926/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/691926/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/691926/4"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 15:53:08.000000000","tz":60},"subject":"Fix CI failures (ODL and searchlight)","message":"Fix CI failures (ODL and searchlight)\n\nsearchlight: do not try to build it\n\nSearchlight did not had release during Rocky cycle. So we left using\n\u0027master\u0027 tarball. Now Python 2.7 support got removed in Ussuri and our\nCI builds fail.\n\nSo searchlight is added into SKIPPED_IMAGES and it\u0027s plugin is removed\nfrom horizon image.\n\nChange ODL repo to nexus.opendaylight.org\n\n(cherry picked from commit 939a56035b8f15b74dff2d68154963735b0f73b4)\n(cherry picked from commit 5a08221cc4270dec3af5ff613731cc7be7576124)\n\nChange-Id: Ib290fb2ec80f408c9d1e134595a5bb7ace922952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f269da3330044b1410d7829fa61be4f8a8b172b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f269da3330044b1410d7829fa61be4f8a8b172b"}]},"branch":"refs/heads/stable/rocky"},"677302eeb16752953cd8703889b90a7f381af6d2":{"kind":"REWORK","_number":5,"created":"2019-11-01 14:25:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/691926/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/691926/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/691926/5"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 14:24:29.000000000","tz":0},"subject":"Fix CI failures (ODL and searchlight)","message":"Fix CI failures (ODL and searchlight)\n\nRocky only\n\nsearchlight: use the 6.0.0 Stein release\n\nSearchlight did not release during Rocky cycle. So we left using\n\u0027master\u0027 tarball. Now Python 2.7 support got removed in Ussuri and our\nCI builds fail. Pin to the 6.0.0 release from the Stein cycle.\n\nChange ODL repo to nexus.opendaylight.org\n\n(cherry picked from commit 939a56035b8f15b74dff2d68154963735b0f73b4)\n(cherry picked from commit 5a08221cc4270dec3af5ff613731cc7be7576124)\n\nChange-Id: Ib290fb2ec80f408c9d1e134595a5bb7ace922952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/677302eeb16752953cd8703889b90a7f381af6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/677302eeb16752953cd8703889b90a7f381af6d2"}]},"branch":"refs/heads/stable/rocky"},"356096d0d8949bf23444b9873dbd6f96e3a0e32c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-11-01 16:19:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/691926/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/691926/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/691926/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/691926/6"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:31:58.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 16:19:18.000000000","tz":0},"subject":"Fix CI failures (ODL and searchlight)","message":"Fix CI failures (ODL and searchlight)\n\nRocky only\n\nsearchlight: use the 6.0.0 Stein release\n\nSearchlight did not release during Rocky cycle. So we left using\n\u0027master\u0027 tarball. Now Python 2.7 support got removed in Ussuri and our\nCI builds fail. Pin to the 6.0.0 release from the Stein cycle.\n\nChange ODL repo to nexus.opendaylight.org\n\n(cherry picked from commit 939a56035b8f15b74dff2d68154963735b0f73b4)\n(cherry picked from commit 5a08221cc4270dec3af5ff613731cc7be7576124)\n\nDepends-On: https://review.opendev.org/692575\nChange-Id: Ib290fb2ec80f408c9d1e134595a5bb7ace922952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/356096d0d8949bf23444b9873dbd6f96e3a0e32c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/356096d0d8949bf23444b9873dbd6f96e3a0e32c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
