)]}'
{"id":"openstack%2Fkolla~691930","triplet_id":"openstack%2Fkolla~master~If859427648e8fe3db7fef7c662ff338859e71131","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"If859427648e8fe3db7fef7c662ff338859e71131","subject":"Bump OpenStack versions for Train","status":"MERGED","created":"2019-10-29 17:55:32.000000000","updated":"2019-10-31 12:47:25.000000000","submitted":"2019-10-31 12:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691930-1572525957232-df6c426c","meta_rev_id":"f14a045d1d96c531a492455a2696852dd735a68e","_number":691930,"virtual_id_number":691930,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 12:45:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-29 19:03:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-31 11:06:25.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-29 18:40:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-31 11:06:25.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 18:40:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-29 19:03:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-31 11:06:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-31 12:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c809045eafa203d0feb2cfa1a4f0cf8d3b7ea796","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-29 17:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af8f04c53b59867fa87b83908c1c9a7d78c6d4f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-29 18:40:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3e4e851f94b9966237a0a63211e0ec11db97d82","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-29 19:03:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/30/691930/1/openstack-check/tripleo-build-containers-rhel-8-master/25e276f/ : FAILURE in 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"78c0b955664943ee4b41d79a1910cadf9e049c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:05:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f91b5ad3d4e049f8acdabbfdaa5df8a2 : SUCCESS in 46m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/127a253d2c30422fbd1802046e864ace : SUCCESS in 1h 06m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8293e6c2110a4d428992358c99cc6e1a : SUCCESS in 1h 18m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/97cb1ee574ea41a48fc6a5647f8fe21f : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e74cddb128294f7d8a084261e823429a : SUCCESS in 2h 00m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/42702be6dadc4e46bbf531fd34598b19 : SUCCESS in 57m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/251b964c815548f1beaa3c14a2c3afe4 : SUCCESS in 41m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d236b3bbeecd47499a012667d3b988c3 : SUCCESS in 42m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aeb78767babe4cbb8178fadff9033675 : FAILURE in 21m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c24a21bf4ca459993160c804e7d1d35 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c5a3c3ba162447abb95e3256fb77c1d : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de974e4f9b364feba1b8139b37c75f35 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bab4a3b00c304688bc9ac48998ff2ede : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1505b16879114f748f98f94a3ff38599 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5ccf2d3e8441b58589b8e07a07b5b5 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/869ccaee7cd54185bf7d81e14fd52f9d : SUCCESS in 39m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0be958ff2e44c2f831098eaebe4aac6 : SUCCESS in 35m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/db7a96650e30476fbe2dc84bc65fed90 : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee64e8ea9ef746e3a9b5681e0afdb979 : SUCCESS in 57m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9797dfa662dd4cd09b9bb6202b299f99 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a757ad9664f77c3e7a34159ef2e560d170e60816","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:06:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e9e18d261f3ee4819e1853bf47aa529449c077e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:06:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71e05d79ee04cd9f94e1be486939dd3e34dfa2e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:45:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"77aea1a6309b4266da21f07f70d9b62cb688a329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:45:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b4022bf8d4134ab5b3c66a41b0bf3c3d : SUCCESS in 55m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4e0faaae0f7b408c9f74963378b3ffbb : SUCCESS in 1h 29m 49s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e66d68506a164fd7a873519cc5a57d02 : SUCCESS in 1h 19m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a1ec867089d04edfb7d9816681f53a6a : SUCCESS in 48m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4776d37585b340bdb2633c61eddbc370 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2948c6b5100140518900e23805d89012 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afdf9a3c02b74f5da36702a4195baf9b : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fcb445b2485460da3b7d0ea94a14fe0 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e678011fa874c0fa7d7491ed967c6d6 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fad9074de064046a32e8e0372a65ad7 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1b7455f14e748259198bb3d22a0bee4 : SUCCESS in 39m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f53984c8c264278aa5174095da8de49 : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8fd78cd8ed4401aaedb17fbecb4fd11 : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f14a045d1d96c531a492455a2696852dd735a68e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:47:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30604ecee9644457a7f2abe7a0fc8948 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2466ce01300849fb329beb88f7b6024fb150af55","revisions":{"2466ce01300849fb329beb88f7b6024fb150af55":{"kind":"REWORK","_number":1,"created":"2019-10-29 17:55:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/30/691930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/691930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/691930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/691930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/691930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/691930/1"}}},"commit":{"parents":[{"commit":"ff40bc6e05082524057188c9757058ced5c21a37","subject":"Merge \"base: use en_US.UTF-8 locale on all distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff40bc6e05082524057188c9757058ced5c21a37"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-29 17:54:38.000000000","tz":60},"subject":"Bump OpenStack versions for Train","message":"Bump OpenStack versions for Train\n\nChange-Id: If859427648e8fe3db7fef7c662ff338859e71131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2466ce01300849fb329beb88f7b6024fb150af55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2466ce01300849fb329beb88f7b6024fb150af55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
