)]}'
{"id":"openstack%2Fkolla~692092","triplet_id":"openstack%2Fkolla~stable%2Frocky~Iaf3414464f3dd747427247339c6b201b352063cb","project":"openstack/kolla","branch":"stable/rocky","topic":"xfsprogs-rocky","hashtags":[],"change_id":"Iaf3414464f3dd747427247339c6b201b352063cb","subject":"Add xfsprogs to nova-compute","status":"MERGED","created":"2019-10-30 09:37:37.000000000","updated":"2019-11-04 12:10:44.000000000","submitted":"2019-11-04 12:10:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692092-1572869444388-9a54f4b4","meta_rev_id":"1b05a8931ecc1ad664390d5f44eea0bf0029e09a","_number":692092,"virtual_id_number":692092,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"date":"2019-11-04 10:10:30.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-04 12:10:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-03 17:04:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-03 17:04:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-03 17:04:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 11:35:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-03 17:04:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-04 10:10:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-04 12:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e26a7e3d384be7a78e24939b65d09d5c5d178026","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-30 09:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f56daddfb144dbd6ed1c96f0200cdb27c98e291a","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-30 09:41:41.000000000","message":"Patch Set 1: Workflow-1\n\nwait for master","accounts_in_message":[],"_revision_number":1},{"id":"bc4c89cdaf9b3ddb970983213dcea7186993c17e","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-30 09:42:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"79500f7054b2cbf65fec7c3e6947df5baf437b12","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-30 09:42:55.000000000","message":"Patch Set 2: Workflow-1\n\nwait for master","accounts_in_message":[],"_revision_number":2},{"id":"9c02c0d8084fae3ede1f51accfa47d86bf555336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 11:43:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2fa5744d03134ea7accee5867656976d : FAILURE in 7m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/afa5af7ac889467d8492c639af9709e4 : FAILURE in 8m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/50e34b969b304546ab0d88e3d5bb6291 : FAILURE in 13m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c6c695da6c824b12aa1bb2f20801d98d : FAILURE in 10m 58s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b1d9711526b44c39bb3af44927800635 : FAILURE in 6m 18s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/97dba4cf4cb740fea85f12d394cb8408 : FAILURE in 8m 36s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ff62aacd92d54070954e28a87a845d17 : FAILURE in 13m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b6ce40b7f97945d994a0262738fdbb9d : FAILURE in 13m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6097235ae334bd4a9d2b223077a0297 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73c562a76714bdab78b179928641b1e : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8eb77dae0345446b9c6326a1f281da9e : SUCCESS in 3m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b39200a0d4c242d6a226d13ef56982c8 : SUCCESS in 3m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3d4d6cc8075b4c28894d1a8eef97510e : SUCCESS in 4m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a3ffafd30b5458fbeeb59ad2511b13c : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7134fde188e34204a7f9c27b787eaf7c : FAILURE in 33m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/42ad1a2ab8394960aa88a470bad039fe : FAILURE in 23m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/864e205fe6574362974cf0c710e8d24d : SUCCESS in 1h 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"282c37afd3a9e5e79a14aed85ea86f71bc62da0a","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-31 09:49:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ee4da27874ddcedcff372045df6c7b9d9764787d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:48:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/107b631b000e4955981e546f49e9eed6 : FAILURE in 6m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/01cab5c81ba7454e8b9c5b044344644a : FAILURE in 6m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dc711f6545a34f59b38d6dec1757a1d6 : FAILURE in 11m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1737b2db03fe44c5a85dfeefa3c8ebfc : FAILURE in 14m 45s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/10415324c6f44e05954c76f3402debc3 : FAILURE in 6m 32s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bfbcc32f703548ecb53a2b0fecccf5fc : FAILURE in 8m 02s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e1bd8807ce794301ae642e7e522ae12c : FAILURE in 14m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3d1abc35404b41878088f426b03e4eef : FAILURE in 9m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/834cfd6cc28246d4a4b619340ca10f75 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4af2b118cb24774969192b03ff07cdd : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24a92654be4544c1a69c9e0cf21117ed : SUCCESS in 3m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3bb96f59205645e1850ccd38e1e4c1a5 : SUCCESS in 3m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/cb88a666c573494eaf752bfd54ec3aa2 : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e73afb067ce54d31bef3749b18dc2c3a : SUCCESS in 58m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b265e501ee0143979a4ab8c83f21b9d4 : FAILURE in 36m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0380d8a4910a434a8b5d542238767db0 : FAILURE in 21m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f7a61218f9a4045b490e6a49d8e261b : SUCCESS in 1h 45m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33f97f7c1bb49dd040978231df2db1f15962d697","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-31 12:17:50.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"f941671347d4de75cc5f443a37dee534422efb07","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-01 11:35:34.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8b380956cd77696aaab7e708916f46c0e1242e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 13:02:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c9b82397822f461fa0fa27248b789f47 : FAILURE in 6m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8e8243d12c3749729d65864a0a915632 : FAILURE in 6m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/60cc004fdbc84af493c0337ed30596f3 : FAILURE in 9m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0c26f10514f94381a54bda19c8b64776 : FAILURE in 10m 13s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/71a260f145d84b8e9b1b9a392f0a8296 : FAILURE in 6m 45s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dd6d4fb2bcca450e98469dab2f6f5c81 : FAILURE in 7m 55s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/346e4a1ad03f4280a13184bdf58f5aed : FAILURE in 10m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3c6b2cd0705848a3af8fc71c8f476d62 : FAILURE in 9m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9193d2c37d44239b111f5b105ecf6bc : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eb66fd2ab248fba0c348eda90725bb : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c0e0549e9da4f118db7001f0bc9d53b : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/161ef82944d54bf588572711c9f99fbd : SUCCESS in 3m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/76e9d25311744264b3a5d3192eda4177 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3d815416a644752849cc49c47d47d42 : SUCCESS in 1h 25m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f5b58a3d9ae456392cd4f8edc102f9b : FAILURE in 35m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c56d7b9b222c423c8e7b8ffca05e9170 : FAILURE in 4m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e091cd73e16b4506aa22b98ad81a5784 : FAILURE in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91180763e8f60c874e7ccfc781a0e6b61fa10214","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-11-02 20:56:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f5a5a50bf534824710838bfcfa6008dda262c14f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-02 22:15:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d1bb874444224428aa4d63d20cb1a107 : FAILURE in 7m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f6334075d4054f3da1b1f1f2b726a238 : FAILURE in 6m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/200c7ad213f64af09580593dec21e1fa : FAILURE in 9m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f5c9bc8599e7468d91ec260d2508ca22 : FAILURE in 10m 52s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/402189463d1f4dcebb93fad16ae5116d : FAILURE in 7m 27s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0dc3b86c099f445f9b0eb0c3ea448893 : FAILURE in 7m 57s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3d17dd9ce3f6435e84a38064de7fd8bc : FAILURE in 10m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/362d891f4af74a7ba8a4dc50f2744c96 : FAILURE in 10m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11ad0aad88424aa3a97b6dcdca3b58c7 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d42b7c5bf140dc94bd24760b77438f : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc7a9757afc24fec8d10f7f17f5c6b9f : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3dc4e003cccc4cfdae1e8b6c197016c9 : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b28a95cbcce343009b8c7f7b6734c146 : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31fad507d8224c2e8038694e0d02f41e : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79f68c49c163433d9a99f30016e064d8 : FAILURE in 32m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd88ed29a18d49e4aeeadad439286256 : FAILURE in 4m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b942ca19eb834b87bce8d4de33a942a4 : FAILURE in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cddf4ef55795ce2d88aec69f3a6cdbfc2c2227dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-03 17:04:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3a7331da88017bb566027112217e19c9b07ae303","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-03 17:04:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"937e1fae068fcdb05aa1eb9e51398ae2d3d6db72","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-03 18:15:14.000000000","message":"Patch Set 3:\n\ndid we fix push failure reaction in rocky?\n\nansible failed because of\n\nERROR:kolla.common.utils.kolla-toolbox:Parent image error\u0027d with message \"push_error\"\n\nnot treated as failure of build...","accounts_in_message":[],"_revision_number":3},{"id":"55d2eeee09d6e0f389f953f4f030166528a96b19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-03 18:16:19.000000000","message":"Patch Set 3:\n\nwe did - we have another unhandled case eh","accounts_in_message":[],"_revision_number":3},{"id":"778aaf027ffbdf79235675d5052cf6dae944a173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 19:03:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a174f9a8b8114fc6be8eb3043c09a5c8 : SUCCESS in 57m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/732639a46c5540c3ae21634646e43807 : SUCCESS in 47m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ec5318251e8d4d5a9c6282032143bc2b : FAILURE in 1h 04m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3b3ff50100ef4a7a90e38afaa52cc044 : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7c4db890bd344ccb843367d110720d49 : SUCCESS in 51m 33s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d8b0a7668d5844828f88c9cf09fca941 : SUCCESS in 1h 01m 03s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5e1e7bb0471549d3ae5ff7a71e03e6c8 : SUCCESS in 1h 30m 06s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/550e34a1724d483c937c0aa4bc05fb8f : SUCCESS in 1h 38m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ee314c85e7646e7a5f4e2ff13bf59c8 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e70bda8cb8144eaa57197c57a6a270f : SUCCESS in 9m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3838843c745c40a3b01b87b7d39769f1 : SUCCESS in 3m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6bc1fee3af924cdf84d12020ec037e43 : SUCCESS in 6m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/da08b6d7241940738db4006ae6a191c3 : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a2eb0e1f1d14d2ca3962f70ad07a8ce : SUCCESS in 1h 00m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fcf7f82bb0142f0aa942b11f1fa7761 : SUCCESS in 32m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f142a34722e44f8ba9484fdf7c510730 : SUCCESS in 1h 08m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65d20ba0169747079ba522e7f2bc5bc7 : SUCCESS in 1h 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cfbc8281a69ed9deddc0a54a1d5bd84737a3410","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 08:41:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a1c0b50092878a2844a5d9b85a65c6d7e5a3737e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-04 10:10:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0f6655dcbe974ef2bf30029f215a662a5ed59e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 10:47:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/89b0429a1e1c4b6aa36dc1ca8e3546cf : SUCCESS in 38m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4b5d8fc76b3340468bc9596b2b89a8b0 : SUCCESS in 45m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/21a7c3b3415447d78699ba496c118250 : SUCCESS in 1h 25m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/580758a637d441cea705b18d32a8c244 : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1db2badfc252446496bf54ec19d73c4f : SUCCESS in 49m 26s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/24b2e94f5d274eafb97946591742e816 : SUCCESS in 46m 32s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8d0aa9b0fd634743bf53db318bddb104 : SUCCESS in 1h 12m 13s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/af6d1ce48cb94bb39455bdb7c5b3ee15 : SUCCESS in 1h 11m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b05e5a4926a045bfa16adc4ca253bdda : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe7a5b31b094bcb9b765bc395ee5dfa : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b059a2468b9b4272a2af34b304a5e5cf : SUCCESS in 3m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5a533f1d4e24713ab161b0f17c27943 : SUCCESS in 3m 27s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c1aea7d9df354348827a0a0b28b45c35 : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5140ba2f7e64a5fbc82778270093625 : SUCCESS in 1h 32m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dfd06f64d5714424bb9a662761997be4 : SUCCESS in 31m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c9a1f6b307b4aa294d497ae3eff8256 : SUCCESS in 49m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c332168aa98e4986a3e9225a14902bca : SUCCESS in 2h 00m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d75dfca0ea48e39d5d6a9266d902a62d713e26bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 10:47:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3dbb68a0cf9c500a0791627395a2b3d8f57c3f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:10:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c588dd4f020440f685418702d9706a3c : SUCCESS in 54m 00s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/25eae144731e4004be5059dfb08e947b : SUCCESS in 1h 07m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2f3fd80bb0c46b7a93faaa3de0e0729 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c417e991014c6aa5ef3859e378a66f : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ceccc5c59ba4a3c8807f695a732994b : SUCCESS in 2m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4620bfcace44dd19fc45b5c61ebd9b7 : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f96af94cc57543688978c6c09cc58d6a : SUCCESS in 3m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/37013c53879e4c8fa46f96bbb65e683c : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":3},{"id":"073960108d43ef105ec37b730a0da6289614f473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:10:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1731da6330c11f80e9cb875d642d5208b81fa98","revisions":{"f0705fcfbd38564d3cb2efa666a67e6b7b757d18":{"kind":"REWORK","_number":1,"created":"2019-10-30 09:37:37.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/92/692092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/692092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/692092/1"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 08:43:23.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 09:37:10.000000000","tz":60},"subject":"Add xfsprogs to nova-compute","message":"Add xfsprogs to nova-compute\n\nxfsprogs is required for formatting XFS ephemeral disk partitions\nwhen format\u003dxfs is specified as instance create failes with\n\u0027mkfs.xfs: No such file or directory\u0027 due to missing xfsprogs\npackage in the nova-compute container.\n\nCloses-Bug: #1850610\n\nChange-Id: Iaf3414464f3dd747427247339c6b201b352063cb\n(cherry picked from commit 9b1a243a3a1669532f25bbdcbc56b78f0f72882c)\n(cherry picked from commit 23e488dfe73a9273698f2a33d8b74f770d9a83f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0705fcfbd38564d3cb2efa666a67e6b7b757d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0705fcfbd38564d3cb2efa666a67e6b7b757d18"}]},"branch":"refs/heads/stable/rocky"},"2c81d08e48bfdae0474988377f2c351093e2211b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-30 09:42:28.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/92/692092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/692092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/692092/2"}}},"commit":{"parents":[{"commit":"97e85ff2f6378ae922416bf16c6030af97697510","subject":"Bump OpenStack versions for Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97e85ff2f6378ae922416bf16c6030af97697510"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 08:43:23.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 09:42:10.000000000","tz":60},"subject":"Add xfsprogs to nova-compute","message":"Add xfsprogs to nova-compute\n\nxfsprogs is required for formatting XFS ephemeral disk partitions\nwhen format\u003dxfs is specified as instance create failes with\n\u0027mkfs.xfs: No such file or directory\u0027 due to missing xfsprogs\npackage in the nova-compute container.\n\nCloses-Bug: #1850610\n\nConflicts:\n      docker/nova/nova-compute/Dockerfile.j2\n\nChange-Id: Iaf3414464f3dd747427247339c6b201b352063cb\n(cherry picked from commit 9b1a243a3a1669532f25bbdcbc56b78f0f72882c)\n(cherry picked from commit 23e488dfe73a9273698f2a33d8b74f770d9a83f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c81d08e48bfdae0474988377f2c351093e2211b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c81d08e48bfdae0474988377f2c351093e2211b"}]},"branch":"refs/heads/stable/rocky"},"d1731da6330c11f80e9cb875d642d5208b81fa98":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-03 17:04:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/92/692092/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/692092/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/692092/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/692092/3"}}},"commit":{"parents":[{"commit":"356096d0d8949bf23444b9873dbd6f96e3a0e32c","subject":"Fix CI failures (ODL and searchlight)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/356096d0d8949bf23444b9873dbd6f96e3a0e32c"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 08:43:23.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-03 17:04:07.000000000","tz":0},"subject":"Add xfsprogs to nova-compute","message":"Add xfsprogs to nova-compute\n\nxfsprogs is required for formatting XFS ephemeral disk partitions\nwhen format\u003dxfs is specified as instance create failes with\n\u0027mkfs.xfs: No such file or directory\u0027 due to missing xfsprogs\npackage in the nova-compute container.\n\nCloses-Bug: #1850610\n\nConflicts:\n      docker/nova/nova-compute/Dockerfile.j2\n\nChange-Id: Iaf3414464f3dd747427247339c6b201b352063cb\n(cherry picked from commit 9b1a243a3a1669532f25bbdcbc56b78f0f72882c)\n(cherry picked from commit 23e488dfe73a9273698f2a33d8b74f770d9a83f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1731da6330c11f80e9cb875d642d5208b81fa98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1731da6330c11f80e9cb875d642d5208b81fa98"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
