)]}'
{"id":"openstack%2Fkolla~692093","triplet_id":"openstack%2Fkolla~stable%2Fqueens~Iaf3414464f3dd747427247339c6b201b352063cb","project":"openstack/kolla","branch":"stable/queens","topic":"xfsprogs-queens","hashtags":[],"change_id":"Iaf3414464f3dd747427247339c6b201b352063cb","subject":"Add xfsprogs to nova-compute","status":"MERGED","created":"2019-10-30 09:41:10.000000000","updated":"2019-11-03 18:16:42.000000000","submitted":"2019-11-03 18:16:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692093-1572805002057-aeac7fa8","meta_rev_id":"7e79c4e69ceab9798397dfeae61ca797d2ff857a","_number":692093,"virtual_id_number":692093,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-03 18:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":2,"date":"2019-11-01 11:35:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-03 17:20:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2019-10-31 12:17:58.000000000","_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-03 17:20:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 11:35:51.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-03 17:20:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-03 18:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"562b3589769c4e4b59998cb3e284126d0eaa10e9","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-30 09:41:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c39947e44accde9cbfe191bb7c1d186f4b16224","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-30 09:41:51.000000000","message":"Patch Set 1: Workflow-1\n\nwait for master","accounts_in_message":[],"_revision_number":1},{"id":"b4574e79f33efec6873213f650a764f9b47cb601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:29:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/930d1d0d24ea4935ae93e2d67af62b2f : FAILURE in 7m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1362927889b4473caf6e0e87162498cd : FAILURE in 6m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1cdd521b8cfe4b79beeabca02f166fed : FAILURE in 9m 30s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/17d509e5a06a4bb49ccb176f1278845a : FAILURE in 9m 17s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4df3364d6a63463ba21758b6dd7689e0 : FAILURE in 7m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/42c06d805d6340d493529dd3816c11f1 : FAILURE in 6m 51s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3e06f1d8336d46ceb5365388aa1f4593 : FAILURE in 9m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9a53d5c345c64714aa52c87db52eeacf : FAILURE in 11m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40b4512ca2c40f488a5f5f61376ec15 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5cc4c460c9246bdbc80ec3fe61bea3c : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c22c639585641e9892741a47ede1207 : SUCCESS in 5m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d0c538b0d2ba4a318d7af2f67e50e4f9 : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/435eeb1a022f49e1977e09e50d2c4453 : SUCCESS in 33m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/63f930ad4f1e46dcad1eeb64d99420f3 : SUCCESS in 32m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/135e09dcb2434386a9f6aee182444d55 : FAILURE in 24m 43s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/08ae5abe74a744adb7acf0571b08468d : FAILURE in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd6f8693282c6481f748635f24f0670b8f9e0493","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-31 09:49:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4091e906957d21866afa705469ee8ad9466f34fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:36:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2884147116734836a1a8dc335beb8849 : FAILURE in 8m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0d6ad0606aac43eaa4d5c6a7b3eae6e7 : FAILURE in 7m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0d99e8f5f9a94073974160f001392e8d : FAILURE in 9m 23s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8788877838de410682427e449d98a4c1 : FAILURE in 9m 14s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ba4e9f48ae6d48c39a3b02e24ebb2533 : FAILURE in 6m 53s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5a3331f9fdd64839b7ff3179cf83834f : FAILURE in 6m 50s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0462f34b9b054e5f9b95a4ca90d7c520 : FAILURE in 9m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/58871ed84efd423ebc48987fd09dd95a : FAILURE in 8m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05801094a6546bc942e800511110ade : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58676824c35842088890f4bc8d8a2a38 : SUCCESS in 3m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/95ecd851639a4743bc1f0e7bf1816730 : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d30aefa1696644cba5193d35d9031a04 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29f062e6242f4eb698e008eb2baf64c7 : SUCCESS in 25m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c76b6c9518624e198458670562d2aa5a : SUCCESS in 29m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fbbc7f1763744d0390ec29438602bde3 : FAILURE in 25m 51s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b463e9e296546149fa229469c8de552 : FAILURE in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e665d6bff17ec6ef34e7ef407a15b091c356335a","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-10-31 12:17:58.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"bc80ee8f03f55bd3e6e95221f9ad2166d94dca19","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-01 11:35:51.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d9b07682965a5e683ce4731c8e718ededf7bff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 13:42:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a461718388854b668ef0ac74c1b2d1a8 : SUCCESS in 43m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dafb84e3ccf2410e9d46c83b8bf891e7 : SUCCESS in 58m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9a3bdaddf306463e8233988a4c22694d : SUCCESS in 1h 53m 04s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/968c14dac4e6429fba0e6d734cc3421f : TIMED_OUT in 2h 02m 02s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8d3e756b64e94f38a7d244e261232654 : SUCCESS in 40m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/079e1cff218844f9a7be4b9f645d0ed0 : SUCCESS in 59m 05s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cb344d1eb8bb420bad02c1fae840929c : SUCCESS in 1h 27m 00s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d5715bbfec8e4ae594b230e78e08b55e : FAILURE in 1h 32m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338de0b7d40740369f4f3eb5c8a1be42 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46ea25bf91a84bb9951b9980cdfd51b4 : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5203f997d654cf7a7877dcf58c9d85a : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7bec4aeda7774a8e99cda973d32700f2 : SUCCESS in 4m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83676ff2dedb485196ed9aacc27f29db : SUCCESS in 31m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/315f1f334ea242c4a947962a411517dd : SUCCESS in 27m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0f252a641f948b5b7743c317176d96d : FAILURE in 29m 14s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8dbd686f3503477996bf7e301774ec40 : FAILURE in 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b30ff18c24fad6ad79bc524dd13232bc4689637","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-03 17:20:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1bca4dc1a79b57da395d2a3b43d04621b8a2777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 17:21:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ae12ad01efa00232f54d7bdd378918cbfa62836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 18:16:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9a74c7356a034f00bcdcdaa631e7a3ad : SUCCESS in 43m 05s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/09eae308d49e4ee293ae2329ea1c9ad3 : SUCCESS in 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3187a23019ed477a8f03b66fc2b97952 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e93878bed3bf4d0c8b2b5c1208c04dde : SUCCESS in 3m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/521e77c9fc794f06806b1ddb7310b894 : SUCCESS in 3m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/40e61b6639494f5e90506dcbc3a8e7b8 : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a642db2393224d318715f82d2dbd1afa : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"85b080eed10348a86ff485f87b578e8f458687a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 18:16:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b103210a86dfafbaf9e10d114d27acf5c3c319aa","revisions":{"b103210a86dfafbaf9e10d114d27acf5c3c319aa":{"kind":"REWORK","_number":1,"created":"2019-10-30 09:41:10.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/93/692093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/692093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/692093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/692093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/692093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/692093/1"}}},"commit":{"parents":[{"commit":"26ec009bfd386040c0f2ff2aa067c7a25772b37e","subject":"Bump more OpenStack versions for Queens EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ec009bfd386040c0f2ff2aa067c7a25772b37e"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 08:43:23.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-10-30 09:40:32.000000000","tz":60},"subject":"Add xfsprogs to nova-compute","message":"Add xfsprogs to nova-compute\n\nxfsprogs is required for formatting XFS ephemeral disk partitions\nwhen format\u003dxfs is specified as instance create failes with\n\u0027mkfs.xfs: No such file or directory\u0027 due to missing xfsprogs\npackage in the nova-compute container.\n\nCloses-Bug: #1850610\n\nConflicts:\n      docker/nova/nova-compute/Dockerfile.j2\n\nChange-Id: Iaf3414464f3dd747427247339c6b201b352063cb\n(cherry picked from commit 9b1a243a3a1669532f25bbdcbc56b78f0f72882c)\n(cherry picked from commit 23e488dfe73a9273698f2a33d8b74f770d9a83f6)\n(cherry picked from commit f0705fcfbd38564d3cb2efa666a67e6b7b757d18)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b103210a86dfafbaf9e10d114d27acf5c3c319aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b103210a86dfafbaf9e10d114d27acf5c3c319aa"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
