)]}'
{"id":"openstack%2Fkolla~692450","triplet_id":"openstack%2Fkolla~master~Ie45cf3465fedddbde7856961527421883ba3d5c9","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Ie45cf3465fedddbde7856961527421883ba3d5c9","subject":"Remove support for CentOS 7","status":"MERGED","created":"2019-10-31 20:43:31.000000000","updated":"2020-04-15 14:05:05.000000000","submitted":"2020-04-15 14:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692450-1586959377460-79213a0a","meta_rev_id":"1baa13a94291639678a806b64ed2e5a46c125ba3","_number":692450,"virtual_id_number":692450,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 14:02:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-15 10:08:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-15 09:32:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-15 09:56:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-15 09:56:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-04-15 14:05:05.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-15 09:32:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 17:44:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-14 20:46:33.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-15 09:32:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-15 09:56:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-15 10:08:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-15 14:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a9f2fa6d58ee4abacb412aa1582c70a86b32be7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-31 20:43:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b98219987b6dd59323a815295e99d90f0e71ed3d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-31 20:53:10.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"30e09af568057ae9659145b8d40d96dc823109f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-31 21:22:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/50/692450/1/openstack-check/tripleo-build-containers-rhel-8-master/9bc0e3d/ : FAILURE in 38m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e604973a4e426be968f1cc6d29df75a136e6733a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 01:53:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f3d80c2cc6104969bb6954c5db85f39b : TIMED_OUT in 3h 01m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a310b0c5a5d245ba972c1cde6f1bc08d : TIMED_OUT in 3h 01m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/78f414934b7c40339d9ee06fd01a9630 : FAILURE in 10m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/427dab344a764114900c4543809957d5 : FAILURE in 11m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/20b255dbfc1d4615ab341e8d66ff6b3d : FAILURE in 10m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/402348298c2e45459331b2358b7caf04 : FAILURE in 16m 29s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/bdd143f636b0426a960e45b9798f23fe : FAILURE in 11m 14s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/e8e014dcc690450d88350ae29ee3f281 : FAILURE in 18m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f84dde965b5e4ed28c8268acabdc7a8a : FAILURE in 17m 10s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d102da0adb19436ebf0bbcbcbfc9ede6 : FAILURE in 18m 31s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5541ba004aac4e71b1e18ab0267681c9 : FAILURE in 20m 06s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/6ec4d81296ce463b9bc85c28c0e518f4 : FAILURE in 14m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ae664fbd59e4e17be4d6140fd1fe40e : SUCCESS in 26m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d9a2898ea7341fc9e21d50ac5df11e3 : SUCCESS in 25m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8b01f23d73f34b4e9871b6a8fdac7303 : SUCCESS in 58m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4551d200fd994faeae417966a595ecf1 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea9fabfd7523427bb1ea81464e662b48 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe5daa093d142309701491fbb45e6e1 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68c1715b51f047268036275f36cf674e : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54d1b5f215c541cba874648544487b52 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028c1cbaec8b4334b5b5898180610bd9 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af5b1ff35e6d41028bad5c4b0f941d0b : SUCCESS in 47m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/138c7016d18f4fe688d5d5bda90ac68e : SUCCESS in 48m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/38d8da826f9841c58059f79e3de54c05 : FAILURE in 22m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df761fe6e08f438f8dde9e8a4b32d00f : FAILURE in 24m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a23781b1a2548979efe3adc7c00f6b0 : FAILURE in 46m 46s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0de2096b71584be4874bc44b1e149025 : FAILURE in 35m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/4fb9c00fb1de4f0c9b04e423a606de48 : FAILURE in 38m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8b66ab95262345808ff547b421b5a913 : FAILURE in 1h 14m 24s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e983b1220edd4c2ab613ba0025c210b5 : FAILURE in 30m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/43b46ba9d033418d8953c96631d50636 : FAILURE in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09137e0da003a03989b58f60351eeeb46d3865ee","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-01 09:11:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"85fd7e880722fc380fa4f8b597e8fdf501cd6cc5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-01 09:51:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"da19f44ec461b4cb883fe7cb1750dd32f2582b3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-01 10:29:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/50/692450/3/openstack-check/tripleo-build-containers-rhel-8-master/decd33e/ : FAILURE in 37m 04s","accounts_in_message":[],"_revision_number":3},{"id":"a6fdabe988447fbee04c3bdb92a351c1730b95e0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-01 11:41:25.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5d5c0b2b11d37778e6f5db710b513e4826c0b38a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-01 12:20:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/50/692450/4/openstack-check/tripleo-build-containers-rhel-8-master/36aeb20/ : FAILURE in 37m 54s","accounts_in_message":[],"_revision_number":4},{"id":"19efcea77f4612add3c3148badb07b7749c97d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 14:46:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a9b2eef0329548788afdd5d47cab04d0 : TIMED_OUT in 3h 01m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd7271ddb80249ad8b9296bf3bc1141a : TIMED_OUT in 3h 01m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5281fb072a2f4ac39c06a9f9a44984ad : FAILURE in 10m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6319cf43240149bfba1eb546ccf0ece3 : FAILURE in 9m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c32411ac5d2480bb4d1b14fcdf9646b : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48448c1fb7004d23963d693db678460c : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa021b67023b4602a2e64de09d982d2e : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f112a1991ef440c18ce1260ebaf798a0 : SUCCESS in 2m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b6bdcc05dff491ba00a4295dd4baca2 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa97111278f345d099d63e7725e4963c : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":4},{"id":"29577517d988ed1fbac4038438b093908cafb93d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-07 17:09:43.000000000","message":"Patch Set 4:\n\nI think we can remove distro_python_version in this patch, right?","accounts_in_message":[],"_revision_number":4},{"id":"53dc20a29e050509c8d480e5b55a68b080b160c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-08 16:04:45.000000000","message":"Patch Set 4: Workflow-1\n\nmarking as WIP as it will take time before it will be mergeable","accounts_in_message":[],"_revision_number":4},{"id":"a903707bc0f0734cdc6b176b570bbdb5ffbf5179","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 13:44:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6d6d431a160f2b77a8a71e047c07c1838aa73f0b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 14:12:20.000000000","message":"Patch Set 5: Code-Review+1\n\n(11 comments)\n\nrebase on top of master will require changing enable_repos macro and it\u0027s tests","accounts_in_message":[],"_revision_number":5},{"id":"a55a7f711f8037d5a868f7aa4d89298160eaf183","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 14:18:41.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/3c87bfda06d845ca9b585a6339036740 : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":5},{"id":"f031fab94b8f53978a6951e220d5c378914f7ebd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 14:48:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"33607a9d5af3d0bf3f47c3cd209769cd4be88351","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 14:48:40.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nAlso removed support for yum as a value for distro_package_manager.","accounts_in_message":[],"_revision_number":5},{"id":"91ad72f8cffd1131ac9aa778f275840d4327080d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 15:09:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/5eb177fd799d45d2ae286e8e44715980 : FAILURE in 8m 18s","accounts_in_message":[],"_revision_number":6},{"id":"7dd1ebe1e00e3d8f51e7d3e090c1001551408316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 16:43:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/578daa1e44e9439aa5e13df7d68f411c : FAILURE in 29m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/00452aee76d04c4c95af59ede9f0be2c : FAILURE in 38m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0871933afa144784ad47672a61e8a287 : FAILURE in 28m 51s (non-voting)\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/472f36a33aaf484c99ee3b17f689e346 : FAILURE in 32m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b99009246f11410eb1026080b89d848d : SUCCESS in 39m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e249d3c5b52b42408e3fc228dfda68ca : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f399e06334da4c6ab54f2474fdb38e40 : FAILURE in 27m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84e9adfec065401d85003d89bfdee3a4 : FAILURE in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc183c642a84988955f8b5fa7e8d66c : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e21143701fbc47adbbe4a0b43725a2f5 : FAILURE in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25872c6704cb4f94bea8fba2eb029d29 : FAILURE in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9033b1f2841d4566bef36c50d06fa472 : SUCCESS in 8m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4309c4ff00be4d7189b0e430e78093f8 : FAILURE in 50m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7329fa0f84194828ac305fb2fb73b4ad : FAILURE in 53m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3876bbcb3e054e6d98aba09e894c30ef : FAILURE in 27m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3abdba45664c4319ab57c1c870cb9bad : FAILURE in 22m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5965cd402e4e45c4bd223e8bc4787c6f : FAILURE in 52m 09s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b3570638499744389b61ce92f9634c7d : FAILURE in 31m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/f00fe80c374e427a967171b45e313983 : FAILURE in 34m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/15a8bfc5f3d348de8345d499cde4e751 : FAILURE in 1h 27m 52s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/42eab61c3a584c0c81521234c176ddd5 : FAILURE in 36m 39s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3115d867847e4d3b95e649b8bcfb7328 : FAILURE in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"acfa4b61fd20ab71f8a765f825aadd855935f353","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 17:16:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f478b5927fc4d29acedbd9eda0775a2 : SUCCESS in 2h 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f66eae04dfa79fcba2e8c643b3add7a6f8202e97","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-21 14:39:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b86c9a50192eb5d32c63669dfea51414bb570eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 14:40:15.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"1ad96aeab0a43c6010e43d3ee128565ca4ee7e9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 14:40:16.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"e2410ce62c35a7504a396aef936ba8467f270f98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-21 14:40:19.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"b88aa6a2da0ab4dae877caa1ca53649c52707569","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 17:17:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"83123bae083f75b17251883c300f99a9322ad2a3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 17:40:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3189f7cf8a17c44a9eeed0f6cb0e3e20a69a3f18","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 17:50:54.000000000","message":"Patch Set 9:\n\nCI jobs need renaming to drop \u00278\u0027 from names","accounts_in_message":[],"_revision_number":9},{"id":"961a9a5dce273f9205f66239dd583092a9836c43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-09 18:10:39.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f92a8e3e5d7f4fd7bba829562d4df63f : FAILURE in 6m 42s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7fcbabd37bc64df3b71cf142c5bf5caa : FAILURE in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"443e8f15eff1567a0fdef6d218704ee41d389077","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 18:17:41.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"08094d774bd3b670262ec19e8762362e00b50e1f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 18:19:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"67cdbb9fa7dba59a7fec7bec6e2004ad72c929ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-09 19:02:03.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/87c0031a811341f8b9bc9eab2b241f0a : FAILURE in 6m 06s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/fbc6162df4314da2989a9614c5273deb : SUCCESS in 40m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d2e2195f2b1498da70a2c557ffbaed260a619c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 20:41:58.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7a8b150077074d01a030dab8021422ce : SUCCESS in 22m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6fa08e095cf5491397fa2a7a8de1cd9f : SUCCESS in 29m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1d83121d74084a1fa50484f38971053e : SUCCESS in 1h 16m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b243a941b9fa48f090cd2764c9da9109 : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9525261fb4f049b09137c9d52f6d0a4e : SUCCESS in 1h 26m 20s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b73fa092a4184f7795246c6b4b4d0ef0 : FAILURE in 25m 25s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8a8bcce8fc8c4115a46cbc2d5ef6d1dc : SUCCESS in 1h 56m 21s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/df2e74d9faf94086a3505667a68d871f : SUCCESS in 2h 02m 39s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/abba39645b4842248e735719bde55d74 : SUCCESS in 1h 16m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/086dd63c0234435c9a1cf2e119ae888b : SUCCESS in 19m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/efde5895d61f4c8bbdd0fe5478852edd : SUCCESS in 21m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/74ca69f8c3ae42a78acebf77f357c7c2 : SUCCESS in 56m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55167c6abbae45e68b7ef2dcbafcf76c : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd91bddc31544fbab0b41030c5c6f3d : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88895fcdf9554cc8b57f34520208c737 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37665d6f9c7349ba8e8388a795e7d1d6 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5f28bd2fa644a32a67268fb35f07e36 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b379581b59441658e979f1ce5c8231a : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba7ef2cfaafb4cb3b118b0c0984a81e5 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2558773f63247029e4043c1104f59ea : SUCCESS in 44m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6750fa7928ee4777959f87c1f3612a20 : SUCCESS in 45m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba501313a7604e32b2ad4f1640b1e567 : SUCCESS in 1h 28m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a3f27896caa4e04b317f9175a027132 : SUCCESS in 1h 37m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/425b3eec453f4b67a9eeff0c28e9a999 : SUCCESS in 1h 41m 32s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ec6f8c9126c9477a92a68bb2390bdcc7 : FAILURE in 22m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9bd86c712b4a4cdeb0c78d611e1bd1f3 : FAILURE in 27m 04s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6b1fc376976a4d65bc42f55f1b591743 : TIMED_OUT in 2h 11m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ade10faa743d4fc9b1e8237d59a2457a : SUCCESS in 1h 07m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f8d6be284b48704ef3d01fb0bae87754d439c2ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 21:24:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55e46191330946faa67aba80fb1e867d : TIMED_OUT in 3h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c16ee72d89d408fc2110bf9be701014d40dd2a07","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-16 15:17:26.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"03343566398912efa204e0b099171a6f12f78deb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 15:24:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f44d93783dc4d1489b2d89e4eeb92f3 : FAILURE in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2277f32eb12a9b4c2d5572cf26c656b8ccde5633","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-16 15:44:40.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/0121302933a5402780465cce791bbfed : FAILURE in 7m 42s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ee0a03fd0fb54ae6a6d9616bceec62c3 : SUCCESS in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"177c46b6369a3cc20f72367df62135814be5fb5f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-16 17:24:44.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"ed2492603e7c34f7338b2b3f74d77f946ce989e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-16 17:25:28.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"337fe6f230b79173a6eb9be6603c51ed73547447","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 17:34:16.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1fe192d1d40a449e8d9d083aa6e8a5e5 : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"160f487ccc8143252ae0abbb331279bc26002870","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-16 18:21:12.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/0ba849d4654c4e6e9e8dbf5a0166178c : FAILURE in 7m 27s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/94e8d3c92585401abf7d6454097969eb : SUCCESS in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3a90250599dbc5fda6d139b85de7d731500e99d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 19:30:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/30af15894b374423b70d5e6c0d56571e : SUCCESS in 33m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8f90be395dd544d886a47d4faecad475 : SUCCESS in 31m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a359eb4739d407cbcc378f4b117d7b9 : FAILURE in 11m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7f254b8cefb84e20a25a8a45b5f34fdd : FAILURE in 10m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/512c9f190fb0449a8dc87046d2753278 : FAILURE in 11m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c218ab61d87549038f2500eed891cb23 : SUCCESS in 42m 17s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/53980497a30640cfb69d78d8a68e8371 : FAILURE in 10m 43s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2dc4f79b6b804eddb7e87ebe70726387 : FAILURE in 12m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f759c3a63d27415286a6dcf40c5661b4 : FAILURE in 11m 50s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/5b32899b732248e194e2b7b8bb3d5e92 : FAILURE in 11m 17s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/dca4e3f871724ad8b348cca122c2c77d : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/bfd948bd0dc94707a70547f39b475e0c : FAILURE in 21m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e00dc8c4ebf54ee7b284cccf7475da5f : SUCCESS in 21m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c7c09f7c51824f1ba39d85c274a3cca1 : SUCCESS in 19m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f7bd55820d074139a124be390b806dcf : SUCCESS in 1h 13m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c39940ebaa9943cdae8e239132a3550f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f99eb47229445887358c2a5d3f3ce0 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b039f15506442278b8b81093b8bcd57 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aec968c102ef4295a9158ceec9041c8a : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b94a31adbdf40ddb724ebf7cdbc6e36 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73cfec6865e4d34847a1b927ae7bfcb : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb516cac99c94afa958aed7796dc27c4 : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/706b0b5a67a744f1b30d23a565521dfa : SUCCESS in 35m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/43e35cc3023f4902958c5aa270994834 : SUCCESS in 36m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca41f983988943a586e57cd837723bd9 : SUCCESS in 1h 16m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e287a6d1ed284283ac9597d565e23fa0 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab564fa9eb27475f9068f59121cd6919 : SUCCESS in 1h 11m 54s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8da195c3912a444c968bb4e46cb14473 : SUCCESS in 1h 33m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2ceef1c55c734443a6368eac138e1f82 : SUCCESS in 1h 42m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8d0b483c62ea45868afa8df3b88b5d74 : SUCCESS in 1h 28m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ab6851ce4dab4778a3b0bfb345299810 : SUCCESS in 59m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5ecb297669604bd9642b846dc9dedad1230e11c8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-30 10:33:37.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"5ff3db3c8eb7d2ced10fc371cef98a0933724163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 10:46:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38a27d9e9a904f4b803d783e4e2591cd : FAILURE in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c8121d2ca8459daa1f5e1e7c8c5502c9a5b9c339","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-30 11:00:22.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/bf4bd8e570d74f8d85125a2bbac8a334 : SUCCESS in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e2aaf9ba5003e7c39eaae94087c3fd5a7c28290a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 12:36:15.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4c49b9d23f4e483980c975add1fa243f : SUCCESS in 24m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b20e43205e3a49fcba71d61617444240 : SUCCESS in 37m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c5be09836eed4033a5b321916cfa9f1b : SUCCESS in 1h 13m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1d50a6cb4b724e15914f8e6e289c35f2 : SUCCESS in 1h 09m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/88ae99d0c9cc4af6a17b36b3bb8a3442 : SUCCESS in 1h 19m 49s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e7dc0c97a6814ef29cd9073f7ec4c71b : SUCCESS in 43m 27s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/cd6bcb695c6f44d9b420839ea46861bb : SUCCESS in 37m 31s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8d3608c56de4468688e3f9082d20bd8e : SUCCESS in 1h 56m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/86906d16dd1c4807a2c1b0220579b11b : POST_FAILURE in 1h 01m 41s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/14a60169c23c49b5b350d22985731fac : SUCCESS in 1h 50m 23s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4ef9acddcd2e46449a0b9cf129b216c7 : SUCCESS in 1h 46m 03s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/0ef39e43ef594c0683aa00673fafb4af : SUCCESS in 1h 31m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/809f6dcf45d8495eaea2d14d6d60a226 : SUCCESS in 22m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/080b7c611b8a4e42ad33bef0ccbb63f2 : SUCCESS in 17m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/59130f7628824248bee1eb4e0fb1b517 : SUCCESS in 1h 03m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea036e8c3e1e4d389ac8f967b553b39a : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ef3c9a930b4f9e8df77f2b02a4dff0 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/397b900fa09d44a49f4920bf644c4526 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80e9535b89434dc6a98d1166994708e1 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/209a61a6724944c4a580a49b94006585 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb72e667f59541be80249eb4e401a407 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48adcaba5f144d8194bd749aeda0dada : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3f1b71a8ec440cc95078aca63c32253 : SUCCESS in 25m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16441525e27d4b79b7be77e0d44e623b : SUCCESS in 28m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aedffc375ffb4cb3b525290043f1907b : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f0cb044438143da929617ba2a1f23cd : SUCCESS in 56m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01aae87060214edca501f66060e16ec7 : SUCCESS in 1h 08m 04s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/70c2b0dadce94029a1fd170218034368 : SUCCESS in 1h 37m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/c6b0280552f04666b85fcbafa013da19 : FAILURE in 1h 22m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d7746ef54ff740f2bf273bcfb8b67fe6 : SUCCESS in 1h 31m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d5ac37c62e9540d890ec2801860ad79d : SUCCESS in 1h 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"07eac249189d36f60421d2950228ed63c25bf796","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 15:58:50.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"8c389da3e55a816e5cc6fc5173517999b5b300e2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 16:21:37.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"161fe73df8f24c47f820e35424b4f058e7081558","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 16:26:25.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"016bb32cc7df381a0562292a3ff38b25c3a65c36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:20:21.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db13fef186b84845a1580f80e1cae302 : FAILURE in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f385f6b0d5fb9c6f11cb933dd0843a7c4841bea4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:48:04.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ffd8f1b05ce34ce297fdd2a64aa6fe58 : FAILURE in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6b42ca6b87b557142652ef24d028c8e55a231072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:49:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/35be7c03ae66428ba9d8e2f2ba59a20d : FAILURE in 31m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8876f90bab7d4b3888e9cc4addbc43d4 : FAILURE in 1h 51m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/75b6f54ad1734f40bbe4fd1323776c27 : FAILURE in 27m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d8e75500936a49c58ded8e08405b6b21 : FAILURE in 33m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0bf08f2f699e48db8a0074d89fdfd691 : FAILURE in 34m 42s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f21f7d5579d34b398c416bd8dd8a13ff : FAILURE in 50m 28s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/029fe72b9bd5414991a045712ef65ed1 : SUCCESS in 32m 23s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/05332e74a67b4d91bd93ff6f1dbe36c1 : FAILURE in 25m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7dea854c7e104097875e379e3b12530f : FAILURE in 30m 48s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/43da267908304a0e91742ca9bbfa447e : FAILURE in 23m 54s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4ca524c9b1ed4c2d84a0971d2931acab : FAILURE in 28m 54s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/307fe1202b8044848d33f0b9853f6359 : FAILURE in 26m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c06d5049328c47259dc2f0f548bb3bb1 : SUCCESS in 35m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/48bdbecdbfee4d8bb64b4f972c837772 : SUCCESS in 35m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cb25bf8c86bf459cbd96f8449a9e1b2b : SUCCESS in 1h 14m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aa945ad301e4a05a3a1cf599a17d14d : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df4dbc9b57e4243811bfae5cf94634e : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fcef23c153e4bc89e6736da895e0b0b : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67486337e1ee4ea09c4ad78b543bf400 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63b50969be7046ab8ff97edb8f7bd788 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db076497669f412ea799e8f12e7a3fa7 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f165114771445f2ae0cf87eb638fd3d : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/127ed95fd6b64abb813f634d059ee722 : FAILURE in 34m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3b8881c7db4427f9d307d2bf2f2c913 : FAILURE in 41m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26a3a31fc3a64a198f35745bcdb8fd88 : SUCCESS in 1h 08m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/180c84eb390c4e97bf1ef304a0bd0e24 : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2c2740f5a4b4979b9d72bc003d06c97 : SUCCESS in 1h 24m 30s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bd8fc0c7faaf465f833e7399643d89d6 : SUCCESS in 1h 46m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/61690ab86e3d48509a777a2c93f38ac9 : SUCCESS in 2h 04m 42s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b702affbed3e477a877f1315cc503ff7 : SUCCESS in 1h 49m 34s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0be31d8d2adf457f8bb57507e11e98e7 : SUCCESS in 1h 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6f42e7a644f2c8c757852f72e3adabd83df05a14","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 08:51:25.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"0b154c40110a90058dc4f3a0ce191fd01dd1ccf1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 09:27:17.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/bfd73bbede224f2587065d79643a88ea : SUCCESS in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b4969dcdc07e90414a8eb425eefcddec9cfe56f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:38:27.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4771c013c0414001893e5eb0a42868ff : FAILURE in 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8835fdb54e4bfec9cd89563d6b7d81aaa31c427a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:36:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b2a30c78d920426ab8f1514310d1bfcc : SUCCESS in 32m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4178f5faa24e44c9a4120e3f290fc685 : SUCCESS in 32m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/94df834a1d82425aa04863f12e3eb046 : SUCCESS in 1h 24m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4ed46942c3cc4766bea3106a156f626d : FAILURE in 45m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c5b1bf84e9a84fe089fceaee3c77f2b9 : SUCCESS in 1h 27m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/467ac989480a498eb48f19a457a38b97 : SUCCESS in 50m 04s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/7447ea5915b941e1a0027c09617e16df : SUCCESS in 29m 55s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5464bc1926414588b8da63585e950c0a : SUCCESS in 1h 47m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e5cfb26d886f44fd98c298b001d712a0 : SUCCESS in 2h 41m 10s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/fb80b9d6f57e42dd9e904d0892a8765a : SUCCESS in 1h 53m 08s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/097bffb30f994ba7bdbb659ae3db5eee : SUCCESS in 1h 47m 43s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/bfbad83afb8f4a5f92c2bea2b07e9f3b : SUCCESS in 1h 16m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/577f642d3e9446ad963709bdaedd8b80 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8a68bc4ab744421b7fac28e3c7c826e : SUCCESS in 43m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6309544d43444bfb15eafe85588d3d6 : SUCCESS in 1h 05m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7e0be8f64f84fcb8f50e60f3c7c0980 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f29295b4a294c2fa24972d54d7f0383 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17508cf59b1949a290e154489cee3bdc : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31c54fd6ca0742a0852ff5472dafd1aa : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b75d0e70a2e34931a73665331cfbe0ec : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8566bc7ef84a76882864bb74d44586 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adca3ff3109e4dc2a06eb83421fb0f11 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e00d0ea907f4834b4cffdbf4898a447 : SUCCESS in 21m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/751626edc63e49bf8ea207952184b838 : SUCCESS in 26m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/32d5dc213689478e9c5157bd582b9dc5 : SUCCESS in 1h 03m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae80521bf95f45409403f3c93f4b80a4 : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/baffe44f91204e46a283740c4eea8607 : SUCCESS in 1h 16m 22s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/14e4eda628a6425bba302790d7148dfc : SUCCESS in 1h 37m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f1c0b733d29b40f7931d86ab8df97cb7 : SUCCESS in 1h 41m 37s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a347ea9574404d518213f29a75a506ff : SUCCESS in 2h 08m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4043249394704c7197a0f94891425443 : SUCCESS in 1h 01m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"08efb60c03cabe52d60790ecc55bdcc7f9e4734a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 12:24:49.000000000","message":"Patch Set 17: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"db77f2e2be08ce393b14f888cd8a826c1af7f420","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 18:44:39.000000000","message":"Patch Set 17:\n\nmore collectd conditionals incoming","accounts_in_message":[],"_revision_number":17},{"id":"9cb33c753a2cf2887b438d0f18653259fc36cd90","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-14 20:46:33.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"65e21d162a9ba63b1714d3dab27c04e56750b676","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-15 09:08:34.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"a4e8835be89756bdc1d2702e3c2f93ad01517b40","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-15 09:11:09.000000000","message":"Patch Set 18: Review-Priority+1\n\nwe should sort that out in ussuri.","accounts_in_message":[],"_revision_number":18},{"id":"97a7569e906f3797e5de5bd19a2423e2d107f009","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 09:15:21.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"487c4f1ce68aff54a329765279a9948d97276336","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 09:28:18.000000000","message":"Patch Set 19: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":19},{"id":"30f841895315b4503760d12a5a0f68e95543d05b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 09:32:09.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"a140167c26bcae5794377556b108919082d7fc21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 09:32:18.000000000","message":"Patch Set 20: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":20},{"id":"c95670f0af25feedc791ade44eed39c1642c62fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-15 09:56:07.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"3407f7adf59ac2efc97b10be4b53f638f431dd68","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-15 09:57:30.000000000","message":"Patch Set 20:\n\nwe probably need to follow up due to collectd\u0027s new conditionals (merge this and grep for more conditionals on the python version)","accounts_in_message":[],"_revision_number":20},{"id":"b1cfbd19893a6c5185bf30f60f4759dd0a1bf5b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-15 10:08:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ed9676a186864b90b8cf0cd5e608bbb1 : SUCCESS in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ecbb1587d110a2b71de2dccab71fd40e025125d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:45:02.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f300ca17e9b841fb8ad30dbcce05957a : SUCCESS in 54m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/279583c3b454468494841685eb1b0908 : SUCCESS in 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"30be23b133c77f007f904953b34e0793d5b01d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 12:25:32.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b3498365b446484a80ad9c9a4ebaa0cc : SUCCESS in 46m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2ef64b552cdc4b099e2cdcc44ce04d49 : SUCCESS in 45m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cc19353d6b494f929351864908034143 : SUCCESS in 1h 19m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1f72e5f7f44540de915e6c95bcf0d961 : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2a868a7795864e0895d2e2a2e1988356 : SUCCESS in 1h 25m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/491374cf87944330bf3c1e008edabbce : SUCCESS in 35m 38s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/c405cf99e51b4476a7f53a1d51cb5484 : SUCCESS in 28m 19s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/475c503def5b424293b5bd55a9188e34 : SUCCESS in 1h 57m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/c3a9d07b29aa4bcfb7bf8059b759670e : SUCCESS in 2h 06m 17s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/75c6b0b4ccb0450faa5ca5181a3dd7fa : SUCCESS in 1h 58m 12s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5c254498033f4930a5e7e7113c9dbf08 : TIMED_OUT in 2h 07m 31s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/bc7706ab16f5453ead43cdc923a82917 : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/28984b537b67462cbfdf98d785fda60e : SUCCESS in 21m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/42c9bb06d52a4f99a7d876e0a99ba3de : SUCCESS in 18m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/92d64d55e5744ba29672b3fce0127674 : SUCCESS in 55m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9865d95280674f5db84768ae0e82eff8 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1003ec8e30af4bb492b2fb9a91676e51 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df7c4841a3924e7bbf0834077a478833 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e098438f01e4aed88fc5f1538063f65 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2f870d0506343f3a25d9c1b6f423350 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f211abcd51f4a9a8e67a9ae7e3503ea : SUCCESS in 7m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c01eab9dd0ed452082f0027b1e231444 : SUCCESS in 20m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0bdac07e7b5f4026b240529576fc1dcc : SUCCESS in 31m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a53d31154a44ce4af95377938baab7a : SUCCESS in 58m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f0da4ac14433441d99a6dc5aa11ccc36 : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3bfd6e47065649f8924376c7b72cfcae : SUCCESS in 1h 09m 53s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a99b306718734502ad8d36de55d24f85 : SUCCESS in 1h 35m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5dd78c39921348828558d27180daa84f : TIMED_OUT in 2h 45m 50s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/505d0001e73a4d008c305cf40144615b : SUCCESS in 1h 34m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cce8962c458444ce98e5ef42615b7d36 : SUCCESS in 1h 00m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"dd5b3d4bdcfe095fe544549b0e1987d3a95e17e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 12:25:45.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"f9d1f9d582b45687b2803d1cbcf2df624971c88c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:02:57.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/242e156c9f54484f8ef68e3aa2b9f1f4 : SUCCESS in 48m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ffc08444b2c14b7ab49858090961d174 : SUCCESS in 1h 10m 02s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/61dc47776b3745d1854520e143847c02 : SUCCESS in 1h 34m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbdeb2c4cf924313bf414943ca0baded : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d31c661b83b9482598a6c7bd8cc675dd : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd733f70afa24052b53c643389493be6 : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/865501f95f2d4d6b9e132858b304d74f : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282f26cf2d0c46ea99a713efc63e55e5 : SUCCESS in 7m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2cd8701bd0b47a0bf20993c3876dd54 : SUCCESS in 48m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44ff02f21e9d4c2da09f83bf5c71d92f : SUCCESS in 1h 10m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/861b49aa3c704478b992d9974401bd67 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":20},{"id":"a12bfedc041b2176e47d84855fddf55b7c574461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:02:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"ea02ff3a6c882291be6e56d312911fca212c8273","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:05:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d27edf42f26465ca0bd47a8c028f047 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"53443c5c712f4ec23b859105e94bcd90abc27762","revisions":{"208276b5925a094643921ac956b165c9f4e36d6d":{"kind":"REWORK","_number":1,"created":"2019-10-31 20:43:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/1"}}},"commit":{"parents":[{"commit":"f9d7d420f284ac89d0b2442fe0a25e336196d549","subject":"WIP: move to CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9d7d420f284ac89d0b2442fe0a25e336196d549"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:31:30.000000000","tz":60},"subject":"WIP: drop all py2.7/py3.6 checks used for CentOS","message":"WIP: drop all py2.7/py3.6 checks used for CentOS\n\nWith move to RHEL/CentOS 8 we no longer have Python 2\nin our images so there is no need for checking which\nPython version (2.x or 3.x) is used inside of containers.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/208276b5925a094643921ac956b165c9f4e36d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/208276b5925a094643921ac956b165c9f4e36d6d"}]},"branch":"refs/heads/master"},"435d7e01973a4d6f58517d5df57611cce3fee273":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-01 09:11:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/2"}}},"commit":{"parents":[{"commit":"19af6a132b30d1adc4299f36b3107663b965416c","subject":"WIP: move to CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19af6a132b30d1adc4299f36b3107663b965416c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-01 09:11:12.000000000","tz":60},"subject":"WIP: drop all py2.7/py3.6 checks used for CentOS","message":"WIP: drop all py2.7/py3.6 checks used for CentOS\n\nWith move to RHEL/CentOS 8 we no longer have Python 2\nin our images so there is no need for checking which\nPython version (2.x or 3.x) is used inside of containers.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/435d7e01973a4d6f58517d5df57611cce3fee273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/435d7e01973a4d6f58517d5df57611cce3fee273"}]},"branch":"refs/heads/master"},"b0fab24b94b2b9ca9a7687c050c020a4c9e54812":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-01 09:51:59.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/3"}}},"commit":{"parents":[{"commit":"4c4a4846a4f23a69e5224db293ff02efb87e1c08","subject":"WIP: move to CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c4a4846a4f23a69e5224db293ff02efb87e1c08"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-01 09:51:48.000000000","tz":60},"subject":"WIP: drop all py2.7/py3.6 checks used for CentOS","message":"WIP: drop all py2.7/py3.6 checks used for CentOS\n\nWith move to RHEL/CentOS 8 we no longer have Python 2\nin our images so there is no need for checking which\nPython version (2.x or 3.x) is used inside of containers.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0fab24b94b2b9ca9a7687c050c020a4c9e54812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0fab24b94b2b9ca9a7687c050c020a4c9e54812"}]},"branch":"refs/heads/master"},"ebd35715b3bc0e1cab841ac63d27202c6e78dcf9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-01 11:41:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/4"}}},"commit":{"parents":[{"commit":"682580c0eb7bda4ce56e785512058f71b95f9abd","subject":"WIP: move to CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/682580c0eb7bda4ce56e785512058f71b95f9abd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-01 11:41:05.000000000","tz":60},"subject":"WIP: drop all py2.7/py3.6 checks used for CentOS","message":"WIP: drop all py2.7/py3.6 checks used for CentOS\n\nWith move to RHEL/CentOS 8 we no longer have Python 2\nin our images so there is no need for checking which\nPython version (2.x or 3.x) is used inside of containers.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebd35715b3bc0e1cab841ac63d27202c6e78dcf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebd35715b3bc0e1cab841ac63d27202c6e78dcf9"}]},"branch":"refs/heads/master"},"86ebda66be8410a8c368bd6249f8d4bbb4199233":{"kind":"REWORK","_number":5,"created":"2019-12-17 13:44:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/5"}}},"commit":{"parents":[{"commit":"0003d42ce07eb5301a38a271239d0edb2ea5ef6a","subject":"WIP: move to CentOS 8 in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0003d42ce07eb5301a38a271239d0edb2ea5ef6a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:42:58.000000000","tz":0},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\nRelated: blueprint centos-rhel-8\nPartially-Implements: blueprint drop-py2-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86ebda66be8410a8c368bd6249f8d4bbb4199233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86ebda66be8410a8c368bd6249f8d4bbb4199233"}]},"branch":"refs/heads/master"},"cfe38e25c44dd9f1fb46a2b7ff2b6393342d4ebf":{"kind":"REWORK","_number":6,"created":"2019-12-17 14:48:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/6"}}},"commit":{"parents":[{"commit":"0003d42ce07eb5301a38a271239d0edb2ea5ef6a","subject":"WIP: move to CentOS 8 in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0003d42ce07eb5301a38a271239d0edb2ea5ef6a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 14:47:56.000000000","tz":0},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\nRelated: blueprint centos-rhel-8\nPartially-Implements: blueprint drop-py2-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe38e25c44dd9f1fb46a2b7ff2b6393342d4ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe38e25c44dd9f1fb46a2b7ff2b6393342d4ebf"}]},"branch":"refs/heads/master"},"bbfe9dcf85c8ecacd5d9735b931ec5e06eee1070":{"kind":"REWORK","_number":7,"created":"2019-12-21 14:39:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/7"}}},"commit":{"parents":[{"commit":"710090cfbf4f045c915714aeda26c5f9e6041fa5","subject":"WIP: move to CentOS 8 in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/710090cfbf4f045c915714aeda26c5f9e6041fa5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-31 20:14:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-21 14:35:57.000000000","tz":60},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\nRelated: blueprint centos-rhel-8\nPartially-Implements: blueprint drop-py2-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbfe9dcf85c8ecacd5d9735b931ec5e06eee1070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbfe9dcf85c8ecacd5d9735b931ec5e06eee1070"}]},"branch":"refs/heads/master"},"b0ec7ad15d06f079d7fa4b3a43062be2f795541d":{"kind":"REWORK","_number":8,"created":"2020-03-09 17:17:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/8"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 17:16:39.000000000","tz":60},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0ec7ad15d06f079d7fa4b3a43062be2f795541d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0ec7ad15d06f079d7fa4b3a43062be2f795541d"}]},"branch":"refs/heads/master"},"53f0956708b344e19275c36c71b5e869da257199":{"kind":"REWORK","_number":9,"created":"2020-03-09 17:40:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/9"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 17:39:58.000000000","tz":60},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f0956708b344e19275c36c71b5e869da257199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f0956708b344e19275c36c71b5e869da257199"}]},"branch":"refs/heads/master"},"d49e9349757f77f5372a28ceb59ce75b662c9c2a":{"kind":"REWORK","_number":10,"created":"2020-03-09 18:17:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/10"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 18:15:27.000000000","tz":60},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49e9349757f77f5372a28ceb59ce75b662c9c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49e9349757f77f5372a28ceb59ce75b662c9c2a"}]},"branch":"refs/heads/master"},"ada6153f6db016f10488caf4432eb3b8bd4a7888":{"kind":"REWORK","_number":11,"created":"2020-03-09 18:19:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/11"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 18:19:42.000000000","tz":60},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ada6153f6db016f10488caf4432eb3b8bd4a7888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ada6153f6db016f10488caf4432eb3b8bd4a7888"}]},"branch":"refs/heads/master"},"75a85a3bb24e6284f0fee8976b617fcd43eabb57":{"kind":"REWORK","_number":12,"created":"2020-03-16 15:17:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/12"}}},"commit":{"parents":[{"commit":"4a4a73fd875252238016b2b312c37c114655e76f","subject":"CentOS 8: Switch default base image, tag as master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a4a73fd875252238016b2b312c37c114655e76f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 15:09:13.000000000","tz":0},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a85a3bb24e6284f0fee8976b617fcd43eabb57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a85a3bb24e6284f0fee8976b617fcd43eabb57"}]},"branch":"refs/heads/master"},"570e59fd62401774c53247d03fe6985c9b762514":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-03-16 17:24:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/13"}}},"commit":{"parents":[{"commit":"95c741f94e4cdd4ea2b75e850a5ad3a842ec07ac","subject":"CentOS 8: Switch default base image, tag as master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95c741f94e4cdd4ea2b75e850a5ad3a842ec07ac"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:24:44.000000000","tz":0},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/570e59fd62401774c53247d03fe6985c9b762514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/570e59fd62401774c53247d03fe6985c9b762514"}]},"branch":"refs/heads/master"},"2510348576b5d125ba692ad34366eb75972876c8":{"kind":"REWORK","_number":14,"created":"2020-03-16 17:25:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/14"}}},"commit":{"parents":[{"commit":"95c741f94e4cdd4ea2b75e850a5ad3a842ec07ac","subject":"CentOS 8: Switch default base image, tag as master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95c741f94e4cdd4ea2b75e850a5ad3a842ec07ac"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 17:25:24.000000000","tz":0},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2510348576b5d125ba692ad34366eb75972876c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2510348576b5d125ba692ad34366eb75972876c8"}]},"branch":"refs/heads/master"},"4df9e085b00e3074ee7df489503fdbeba0b7953b":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-03-30 10:33:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/15"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-30 10:33:09.000000000","tz":120},"subject":"WIP: Remove support for CentOS 7","message":"WIP: Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df9e085b00e3074ee7df489503fdbeba0b7953b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df9e085b00e3074ee7df489503fdbeba0b7953b"}]},"branch":"refs/heads/master"},"b3a2447911228269e4f259e18800972689f92fb3":{"kind":"REWORK","_number":16,"created":"2020-04-08 16:21:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/16"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:21:25.000000000","tz":120},"subject":"Remove support for CentOS 7","message":"Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3a2447911228269e4f259e18800972689f92fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3a2447911228269e4f259e18800972689f92fb3"}]},"branch":"refs/heads/master"},"c99179d7355a5aec628a5f26ee9176a229932a9a":{"kind":"REWORK","_number":17,"created":"2020-04-09 08:51:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/17"}}},"commit":{"parents":[{"commit":"a176c6e080be2e71829e8ca5e8b330fdda70038a","subject":"Merge \"Add docs and release note for CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a176c6e080be2e71829e8ca5e8b330fdda70038a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-09 08:51:12.000000000","tz":60},"subject":"Remove support for CentOS 7","message":"Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c99179d7355a5aec628a5f26ee9176a229932a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c99179d7355a5aec628a5f26ee9176a229932a9a"}]},"branch":"refs/heads/master"},"c82d1114a34a950c96a8f0c516e3e37719419529":{"kind":"REWORK","_number":18,"created":"2020-04-15 09:08:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/692450/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/18"}}},"commit":{"parents":[{"commit":"a176c6e080be2e71829e8ca5e8b330fdda70038a","subject":"Merge \"Add docs and release note for CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a176c6e080be2e71829e8ca5e8b330fdda70038a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-15 09:06:39.000000000","tz":0},"subject":"Remove support for CentOS 7","message":"Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c82d1114a34a950c96a8f0c516e3e37719419529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c82d1114a34a950c96a8f0c516e3e37719419529"}]},"branch":"refs/heads/master"},"08f6da4a03474072cb6495fc47b1bbf1bb53b621":{"kind":"REWORK","_number":19,"created":"2020-04-15 09:15:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/19"}}},"commit":{"parents":[{"commit":"a176c6e080be2e71829e8ca5e8b330fdda70038a","subject":"Merge \"Add docs and release note for CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a176c6e080be2e71829e8ca5e8b330fdda70038a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-15 09:15:18.000000000","tz":0},"subject":"Remove support for CentOS 7","message":"Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f6da4a03474072cb6495fc47b1bbf1bb53b621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f6da4a03474072cb6495fc47b1bbf1bb53b621"}]},"branch":"refs/heads/master"},"53443c5c712f4ec23b859105e94bcd90abc27762":{"kind":"REWORK","_number":20,"created":"2020-04-15 09:32:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/692450/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/692450/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/692450/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/692450/20"}}},"commit":{"parents":[{"commit":"a176c6e080be2e71829e8ca5e8b330fdda70038a","subject":"Merge \"Add docs and release note for CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a176c6e080be2e71829e8ca5e8b330fdda70038a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:44:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-15 09:32:06.000000000","tz":0},"subject":"Remove support for CentOS 7","message":"Remove support for CentOS 7\n\nWith the move to RHEL/CentOS 8 we no longer have Python 2 in our images\nso there is no need for checking which Python version (2.x or 3.x) is\nused inside of containers.\n\nWe also no longer have to support yum as a value for\ndistro_package_manager.\n\nPartially-Implements: blueprint centos-rhel-8\n\nChange-Id: Ie45cf3465fedddbde7856961527421883ba3d5c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53443c5c712f4ec23b859105e94bcd90abc27762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53443c5c712f4ec23b859105e94bcd90abc27762"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
