)]}'
{"id":"openstack%2Fkolla~692530","triplet_id":"openstack%2Fkolla~stable%2Fpike~I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b","project":"openstack/kolla","branch":"stable/pike","hashtags":[],"change_id":"I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b","subject":"Remove Zuul configuration","status":"MERGED","created":"2019-11-01 11:22:29.000000000","updated":"2019-11-04 12:23:17.000000000","submitted":"2019-11-04 12:23:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692530-1572870197795-c91a571c","meta_rev_id":"e95dddee1b0fcf8a8fd6c776e53542dbe8a55c6b","_number":692530,"virtual_id_number":692530,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-04 12:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-04 12:23:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-04 12:23:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-04 12:23:17.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 11:23:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-04 12:23:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-04 12:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c42e7b51b5b200d1ef02d8d044002547a090894","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 11:22:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c09117b4fe8496171cc25ef2a437e3bc943e1248","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-01 11:23:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69290894df693e23472c111e90fa074262762b91","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 11:23:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10bb50cbdc8b09feae360d6d71d15dce8d79432e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-01 11:24:15.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"052b645dc2132c1b2bfabde5604c2ef74633cea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 12:07:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa9e6f57cd448f6a8558d28f0360a19 : FAILURE in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e3cd722abdf42358e31073ea12a5413 : SUCCESS in 4m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ea68cd35a9c4430a47eb7930a7620cd : SUCCESS in 3m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ecbb03d73c004beaaa2437c02be509b2 : SUCCESS in 4m 10s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3dd0fcbf771846488a49a476343aa29f : FAILURE in 7m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2a0da69305d04a86a09213eb098c2c10 : FAILURE in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f4d7c43f4394b7f8d787589579a41e8 : SUCCESS in 25m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/553c8440ebe04bcc8917f7afe63c9a8f : SUCCESS in 29m 07s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ac3aabb35992423988a29fa536cacd46 : FAILURE in 7m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f41205503c0849e08e0ec98da498cbba : FAILURE in 7m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/035f8232b63d453db8cb6277fe1c623f : FAILURE in 12m 23s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b452a7400e324f9ca79219fd4e322165 : FAILURE in 10m 12s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7ec3db63c384e829a20ad9df9487317 : FAILURE in 25m 09s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c85ee1de9994acca7fb34c093559bfe : FAILURE in 24m 11s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/471241f5b3d34c77a15ba7e0372079c0 : FAILURE in 9m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/faba5e86c93b41999166b0d2f335953e : FAILURE in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e1dccf6748fed981e09d80950596bf29f9937f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:32:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"011f6af008b1756450d3c66fa6626f5de7cf0f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:04:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d41d53fe0948ba8069877c4cc3269d : FAILURE in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/183ccd71eae943b7935c4290abb003af : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/35800880b78d4f26b45edae7d0fd3550 : SUCCESS in 5m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9651eba6ee7e4691bc0ec8a71f5faced : SUCCESS in 4m 03s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2379650774a2464db972c42b4964a3aa : SUCCESS in 33m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2d7152e5376244f199db4b1d541e62b6 : SUCCESS in 54m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc5c16ca79f34872b17470d5f440d147 : SUCCESS in 1h 04m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5769680f30c1435c9718aa8ed2ac1c7f : SUCCESS in 23m 14s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/70c9dca00c7f43f39ac58cd5b2f7e90b : FAILURE in 17m 53s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7d754ec1429449f0ac0573846caa1396 : FAILURE in 17m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3e41265e20504b90a05f5acff2af5419 : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ec0e00f11e894ec7810ce23c483d4f03 : FAILURE in 24m 15s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c19c41d711be4f5698bebd0c11784486 : FAILURE in 12m 15s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aad6a8acf0664b3d9e0a1ab8455e5fdb : FAILURE in 10m 02s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a36b169ecbcf46dcbaa118114ec4bd7a : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2e01094048bc413395cbd6e0adcb4af2 : SUCCESS in 1h 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29df1192afb6f06c86c99931d953d1add5d1f7fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 11:18:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b4332d58da624d9823f67b9941a268b566c66189","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-04 11:23:15.000000000","message":"Patch Set 4: Code-Review+2\n\nbye","accounts_in_message":[],"_revision_number":4},{"id":"a0da3054b2405c320874b9e9563dbfe8d156dcda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 11:26:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nbye pike","accounts_in_message":[],"_revision_number":4},{"id":"b1a5ace5623e2e27788875a150b0f4ed33595c86","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 11:27:49.000000000","message":"Patch Set 4:\n\nI don\u0027t think it is being picked up actually","accounts_in_message":[],"_revision_number":4},{"id":"03bd814297bce91712d3c5f07c1a1d849ab504e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 12:06:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ccee8f18e1d4c1b36ab639100033c57c97294c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:06:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/4175a80377814360867d466e8ef0e100 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":5},{"id":"a7ae8345c8eccca20956f9b47e0f2aecf49fd1ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 12:15:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c26f97ebc7f1605055a3e46af280c3f6699197fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 12:16:19.000000000","message":"Patch Set 5:\n\nand noop gate?","accounts_in_message":[],"_revision_number":5},{"id":"0fa390daf50b1e10b3703012db7647473f93a941","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 12:22:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"19be053a582cb5be3c9e34ebcdfc5f5af370d07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:22:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/6ece93a0bffd4c4385b4b84e2af642b9 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":6},{"id":"51765892db3ee63ff5b2d125a52bcfaf696abb95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 12:23:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1567dc48976c56f13a1f757038d189ebd9373b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:23:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5cde8c9259a403d23be4e94e6b303664c8b857e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:23:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"529894f29e4ddeb73261bc87ad7fa619a1d6b330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:23:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/96f6c3939dac41b6bbb9c43268332493 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6149a454b5a37fc9819b553ae2bc5bd29930dd6a","revisions":{"028fc9b74ed704f2b6d63b003571c204f9923ebd":{"kind":"REWORK","_number":1,"created":"2019-11-01 11:22:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/692530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/692530/1"}}},"commit":{"parents":[{"commit":"093fc1dda0788223a91c08e65ce8623789cebe8f","subject":"Fix fluentd image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"subject":"Disable pike publishing jobs","message":"Disable pike publishing jobs\n\nPike is now end of life.\n\nChange-Id: I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/028fc9b74ed704f2b6d63b003571c204f9923ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/028fc9b74ed704f2b6d63b003571c204f9923ebd"}]},"branch":"refs/heads/stable/pike"},"f7c4b502fe311cec0457699631e207904348ef4e":{"kind":"REWORK","_number":2,"created":"2019-11-01 11:23:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/692530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/692530/2"}}},"commit":{"parents":[{"commit":"093fc1dda0788223a91c08e65ce8623789cebe8f","subject":"Fix fluentd image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:23:47.000000000","tz":0},"subject":"Disable pike publishing jobs","message":"Disable pike publishing jobs\n\nPike is now end of life.\n\nChange-Id: I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7c4b502fe311cec0457699631e207904348ef4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7c4b502fe311cec0457699631e207904348ef4e"}]},"branch":"refs/heads/stable/pike"},"0d7a9c74248adb934e6e9399901babc3bad0122b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-01 14:32:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/692530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/692530/3"}}},"commit":{"parents":[{"commit":"59185141703d459acbd7392c54437b5f5e843729","subject":"Change ODL repo to nexus.opendaylight.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59185141703d459acbd7392c54437b5f5e843729"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 14:32:16.000000000","tz":0},"subject":"Disable pike publishing jobs","message":"Disable pike publishing jobs\n\nPike is now end of life.\n\nChange-Id: I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d7a9c74248adb934e6e9399901babc3bad0122b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d7a9c74248adb934e6e9399901babc3bad0122b"}]},"branch":"refs/heads/stable/pike"},"2e58d04084c71e75ee230b4db3351e5878f253da":{"kind":"REWORK","_number":4,"created":"2019-11-04 11:18:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692530/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/692530/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/692530/4"}}},"commit":{"parents":[{"commit":"093fc1dda0788223a91c08e65ce8623789cebe8f","subject":"Fix fluentd image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 11:18:39.000000000","tz":0},"subject":"Remove Zuul configuration","message":"Remove Zuul configuration\n\nPike is now end of life.\n\nThe goal here is to disable pike publishing jobs, and the CI would need\nsome work get it passing just to remove those jobs, so it\u0027s simpler to\nremove it entirely.\n\nChange-Id: I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e58d04084c71e75ee230b4db3351e5878f253da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e58d04084c71e75ee230b4db3351e5878f253da"}]},"branch":"refs/heads/stable/pike"},"bc00b474ad1ed8b51c2f0697fa30d5591560636f":{"kind":"REWORK","_number":5,"created":"2019-11-04 12:06:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692530/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/692530/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/692530/5"}}},"commit":{"parents":[{"commit":"093fc1dda0788223a91c08e65ce8623789cebe8f","subject":"Fix fluentd image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 12:06:00.000000000","tz":0},"subject":"Remove Zuul configuration","message":"Remove Zuul configuration\n\nPike is now end of life.\n\nThe goal here is to disable pike publishing jobs, and the CI would need\nsome work get it passing just to remove those jobs, so it\u0027s simpler to\nremove it entirely.\n\nChange-Id: I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc00b474ad1ed8b51c2f0697fa30d5591560636f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc00b474ad1ed8b51c2f0697fa30d5591560636f"}]},"branch":"refs/heads/stable/pike"},"6149a454b5a37fc9819b553ae2bc5bd29930dd6a":{"kind":"REWORK","_number":6,"created":"2019-11-04 12:22:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692530/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/692530/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/692530/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/692530/6"}}},"commit":{"parents":[{"commit":"093fc1dda0788223a91c08e65ce8623789cebe8f","subject":"Fix fluentd image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-01 11:21:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 12:22:02.000000000","tz":0},"subject":"Remove Zuul configuration","message":"Remove Zuul configuration\n\nPike is now end of life.\n\nThe goal here is to disable pike publishing jobs, and the CI would need\nsome work get it passing just to remove those jobs, so it\u0027s simpler to\nremove it entirely.\n\nChange-Id: I806b55302e0da4ebaeb5610ad2ae9bfe1138e79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6149a454b5a37fc9819b553ae2bc5bd29930dd6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6149a454b5a37fc9819b553ae2bc5bd29930dd6a"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
