)]}'
{"id":"openstack%2Fkolla~692794","triplet_id":"openstack%2Fkolla~master~Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49","subject":"CI: Move to CentOS 8 nodepool images","status":"MERGED","created":"2019-11-04 10:31:10.000000000","updated":"2019-11-27 13:50:17.000000000","submitted":"2019-11-27 13:47:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692794-1574862429894-599b186f","meta_rev_id":"ace064578b47729167beed96b8e49532f1b70409","_number":692794,"virtual_id_number":692794,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 13:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-11-26 13:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-26 13:18:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-11-26 13:34:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-26 10:22:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-06 08:43:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 09:04:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-26 13:18:04.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-26 13:34:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-27 13:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba8f377ffc423d73b0cbfdb5a84c42504fc73472","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 10:31:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9910972ad683634de694213632d446e759a6fa65","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-04 10:33:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6eef60f2d60bfbf8017e5e4c043a3a51a466157c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 10:37:24.000000000","message":"Patch Set 1:\n\ntotal move? don\u0027t we want to test upgrades on centos7?","accounts_in_message":[],"_revision_number":1},{"id":"aa45ca58a36857ede7f9c2ca37437d96b11062b1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 10:37:51.000000000","message":"Patch Set 1:\n\nah, this is kolla, ignore me","accounts_in_message":[],"_revision_number":1},{"id":"01c24bcef78af6a79d330349908edb23ddcfe449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 13:36:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8c83dd8dd4db47098fccfb4b313811f5 : TIMED_OUT in 3h 00m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d6b02d0478034dd08e2cc5a92b19d15a : TIMED_OUT in 3h 00m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/353262df616f47588e9603e3f533a28f : SUCCESS in 32m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c36c133cd514bc089b5541eabc11958 : SUCCESS in 29m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a7a4265b757f4863b3ab41b11dfbe0f3 : FAILURE in 30m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d263d86dc554837856e1b283a9c9369 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99baca6d67aa465e88bf708a9da3f357 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/29345c8aec48480ebb7dfede42bb29c3 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ad5b54d32734174ad4df9ef71c952a3 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/663317d71b094943921b4a0aaa215b4c : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed25e44145141f8bc989d822c575c90 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/debcc45c16744bfd8173351e01125510 : SUCCESS in 26m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/339c07b0460941a38322ef6fc6658d0c : SUCCESS in 28m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87e2cbf2741142b7afdfdf9faf000487 : FAILURE in 21m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/350a36480e694490a2466f491a851ccb : FAILURE in 25m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b548253ea2c9404ea3355ebfb40bd3ca : FAILURE in 47m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9d555e38bc051309fac60da980e01fed4861b80b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-06 08:42:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"27e5084b11f857ecf3858c477c3046307d3ba391","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-06 08:43:13.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"5bde5bed1813db0c552ea5dc1653894db518150e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 08:43:15.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"7a556cf8e9c770a4beb69b564eea6e29cff436a5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-08 16:04:54.000000000","message":"Patch Set 2: Workflow-1\n\nmarking as WIP as it will take time before it will be mergeable","accounts_in_message":[],"_revision_number":2},{"id":"110cca4d30816d90cfa29b5345d000a7ace00d78","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-11 09:04:40.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6c8b04704605f11ae13ba731d0775fb6a1413012","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 10:43:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8ec3b346f76c9d743a20aa5ea5a60451089836dd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 10:43:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c2f02c033a3aad8e7131462c7000840bc63154de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:47:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a640c7492daf48038aaf486ef45799f0 : FAILURE in 20m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/82d8d749927a4226a354a9844358fa77 : FAILURE in 32m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0dd8ff63ac243fdab4f0a4045f9f87c : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854111b7f05f4db1a8b19946df3cd660 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90b811978da54b41bc1266b9b9e76c08 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88cb79bcdd2f431fa5922becee250fb7 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8dcb594e198444984876e6ec6e94573 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f0cce6ab924ac78f386ae42e1fd4df : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"45970e0c5f0f39b190dc54720b43c24002916cfd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 15:12:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"519ceaa66ba49e479934bda242acb6d90b2489a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:52:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ef050e992f804bde80aeaf95d102cdcd : SUCCESS in 32m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/11d851eb1d404f5481a2dde2b40cef95 : SUCCESS in 47m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8abad517b3634d85b3d03d5533533b29 : SUCCESS in 1h 35m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/88a7c60ec7cd42bb85157433fd16bb51 : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d78652acecb344a8a3b6372ce47ce69c : SUCCESS in 1h 31m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/196e3318d84e42d3bd56694b01e94297 : SUCCESS in 58m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/795740deab084f61b42540918e6ddb78 : SUCCESS in 25m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e4087f69c6b447d1afde37d74cdfd273 : SUCCESS in 22m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/633c53fbfa5143fa8190de20a4cc2129 : SUCCESS in 54m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5782a7c492bf451e931a3f18d00dbaa5 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb66abc1591441d0bd0d9c148b5a7e99 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7efbbb5011b4a24824806d0262c1687 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed3052e5974848d0a7ec158e4fa87428 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07ce056da324505b77a1d5e3c43d463 : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c86564b8115464aa0248e94272b1e7a : SUCCESS in 29m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed26988f5d8146b1a0b2af2998022f9a : SUCCESS in 35m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c948cad2282461a9932c2cf7a3acaec : SUCCESS in 58m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24d3d412764848d790892220cda598c3 : SUCCESS in 55m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5e841d4df1ec4c68b710fe7f24a8200f : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f2f2141ed55afa03eb4580632da2ef721a3d9960","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 10:20:25.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"93fe7f5df716ea2ccae1338f78cbb40906ca67fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 10:20:34.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"b8b732868a70bcd48f30edf4eb2f20c97b21b8f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 10:22:22.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f6b35a2279ca23b929679ae870fe346ae44f2e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 12:06:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/797d16c7162e4f63801008d587ee1cd7 : SUCCESS in 37m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a040619acf9d43f386c5afa83e5fbf0c : SUCCESS in 1h 01m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bec71248b5184dacac8cbc73f18927ab : SUCCESS in 1h 22m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a42eb284c2e7419bba02998a85239edc : SUCCESS in 1h 32m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d54d6942fc1749bfb19522cb6b004f77 : SUCCESS in 1h 38m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d39b402267e344faa7b29782bba380a6 : SUCCESS in 55m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7353daf0c7814dc09f01c7708f7c56bb : SUCCESS in 27m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6a296c65da50486285addd0d6d674746 : SUCCESS in 43m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8e95b09baa0e4bfb8c28411bb96b8f06 : SUCCESS in 1h 00m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/433211cf3e514cdba9fcd8fda898664c : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735404cfea9e42e8aa2d1b116d551ea7 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac54f41c45dc4091b536b067baec1287 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0a1cfa7e84a4bebb784b7d2c45a6349 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12c2ddb9b2f482396694fc59e85c037 : SUCCESS in 7m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55988c6430794e3b891300244503edca : SUCCESS in 48m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a929187cafb847ba872d2c1f3a4f9be7 : SUCCESS in 35m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/42bd6ffcde4c4e16b07008d8928d5e13 : SUCCESS in 54m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d40815af5254e1c89490717d86c2e99 : SUCCESS in 58m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c3e719dfb7e48d586beadd7137f9fe5 : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":6},{"id":"3b341d7c98f1795ec64e79467e1691b0b0580f88","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-26 13:18:04.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2234230acdb9245d5ff32be2055a00906673ed6f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-26 13:34:11.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d0be7c9ebd00673dc18bf8897d4e1c8781ea4973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:34:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f25e1b10ee4f7133c32003d47a9fcc7f6ddd9b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:25:24.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5f3a780244dd4ad78df99b6fb403af90 : SUCCESS in 41m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/35e1a6f58e954b7eb22bc2f48a3ce2a0 : FAILURE in 1h 23m 28s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1239fafb2070414888ae795f066cb17c : SUCCESS in 1h 49m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/31b530594bf74f20aaeaa36d91e9bfa8 : SUCCESS in 1h 06m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4382745e4274d688744025700128c95 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513ba65b32d84ce3ac582556132cae6c : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5e1f3fd068463a98421a64f9f5bff7 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a6dccfbfdb244dca403cfbdab3fd0d3 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7df21554ad5419bb1159ee7a7f08fc1 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/075edd8b90d14f008b5eacdb3e6b615e : SUCCESS in 26m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d439531cf8514b1b9e30a2de3c4b3890 : SUCCESS in 53m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6b39d5b8f2124322a012c241690449c7 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":6},{"id":"1a63091266722d7f0fb26307f607ef25d98138e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 16:09:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7b89c8c9083e3499db55fd27fb0414327599b46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 17:52:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4c676daeb7594e6bbb765a873b4264a7 : FAILURE in 36m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a769a0d1e11645b8b397b95f894f98de : FAILURE in 41m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9cb80d2b3a974f418f2ce623cf1d898e : SUCCESS in 1h 17m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/492aa0a4beb340669db1e507bbab97e1 : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b189315abd59435d8e1933380cfaf675 : SUCCESS in 1h 37m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/dc746a215cc743288f5735fff3c898bb : SUCCESS in 57m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fa85f82733f9413b9b03f1df50388600 : SUCCESS in 41m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e86df34bb9424f228fb131a37f6dbe60 : SUCCESS in 38m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/586950bcf59b4d30b7070fad57d92e52 : SUCCESS in 59m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d51e3cdf4c84438936d2a5176a12101 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfea788927b64b0aafbed1cfe802afa7 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4c5b116699244279ccf3b7825623bf2 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dced82dce3204cecba29ad973b8cf522 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3011b272e02c4060bd3ca6fe928839ac : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd0f09c431e244e4b869952cffac9bab : SUCCESS in 27m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2fd3726a69bc4d04a83b268d0ab8f1b1 : SUCCESS in 29m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed5abfcd55a649fdb46d918819bf2d13 : SUCCESS in 58m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a91f57224dd048f0b0a6f203f5c9a42b : SUCCESS in 58m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2282173cc7a43f98541b9807cfe8c70 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":6},{"id":"10a9f99783219d928a5c87ecd32ea06938a1e30d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-27 10:34:04.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f0cfd679df128e6fb906dbf53f0efb1c6c3f564f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:07:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1627fa7987ec4a6e9d6e0900a9f26fe8 : SUCCESS in 30m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2162887feba74f378c4ded8d36d97bd4 : SUCCESS in 38m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dbe0a0d3e2ed404aaf5442bf141eec4e : SUCCESS in 1h 20m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/58757f528a33465a84b037192309ae6b : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3915e431ea714a6da102a1d2e78a518e : SUCCESS in 1h 21m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/030b15e3b7bc4e1ab616171ccc0925e2 : SUCCESS in 50m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a1620a4ce0d94a158546d17842dd36a3 : SUCCESS in 37m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/94d7a85e17cb4485b8950d83f7d887f1 : SUCCESS in 17m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5297c8743dce43508031dece80f349ab : SUCCESS in 56m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e498f250e46b4ab1aa7a57fa22db13cb : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f40022c75123437f8cae96d4168296a9 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aec23520b1e426d863986933c0b0be9 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c7918eec67844d9a875aa93905156c1 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19349d1d6cd64d36b103bc70af02acec : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd09b490e90d4e04b69fb942302f9ca7 : SUCCESS in 27m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7930e769d454c5582e7ca717ed461b7 : SUCCESS in 30m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06dd0e01a96b42db8d20d52a7bda0d5d : SUCCESS in 59m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e117fdc799de4889b323450f1882759f : SUCCESS in 54m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/615a8f63a1344921b59b6663ca683a83 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":6},{"id":"d7650e9f68549a106f958167fac98ae436a0cd60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:07:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"534089e74c9c9ae664d36cd8e7dcfb76357c969f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 13:47:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"fbc69881876ce2abb472c4583c4982acf86fe780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 13:47:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fb072f090ef842d18b0c7637f17428e4 : SUCCESS in 40m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a1def3063a774f78a0d4210f61c6cb29 : SUCCESS in 1h 19m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1acf3b96842d440882713afe96efee8b : SUCCESS in 1h 37m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ac7dd8ee42224438a99b488dacf089ee : SUCCESS in 57m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c716ee9007ba427dae91c159dbf2daf6 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc64ec27f95048deafc8706d883854b0 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dcab6772bdc4a7ba2ffb0d87cb9166d : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ffc1c6fa5db4f93ae62dfa01e65c91b : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd3c9923daf4d82a630dc635ef28a0f : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e41ff80aeb1440b8572c4691e4f9277 : SUCCESS in 27m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76cc2c795b904687b72e143d05cb0fc6 : SUCCESS in 55m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eadf5e6fa9544c15bf1a795da4a9c292 : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":6},{"id":"357817e3ff89788d59a1fd5ed2dff9b68ac6bd1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 13:50:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4580123f946471ebc024ccc3ffc9e41 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6c86915597c00da845bbfa433e897b1573f50c70","revisions":{"c2f74479de6c1b00e4faf9273717837bb962515f":{"kind":"REWORK","_number":1,"created":"2019-11-04 10:31:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/94/692794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/692794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/692794/1"}}},"commit":{"parents":[{"commit":"a95d65337353e776ec269015d194b887cc80fbf4","subject":"WIP: move to CentOS 8 in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a95d65337353e776ec269015d194b887cc80fbf4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:20:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:30:00.000000000","tz":0},"subject":"WIP: CI: Move to CentOS 8 nodepool images","message":"WIP: CI: Move to CentOS 8 nodepool images\n\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python3\n\nChange-Id: Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2f74479de6c1b00e4faf9273717837bb962515f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2f74479de6c1b00e4faf9273717837bb962515f"}]},"branch":"refs/heads/master"},"72e4cf806f78cdeb0cd18b40ff677e772f0c067b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-06 08:42:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/692794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/692794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/692794/2"}}},"commit":{"parents":[{"commit":"42643e211eb256d55e54cb5ea2704fb8b637a8a9","subject":"WIP: move to CentOS 8 in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42643e211eb256d55e54cb5ea2704fb8b637a8a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:20:45.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-11-06 08:40:55.000000000","tz":60},"subject":"WIP: CI: Move to CentOS 8 nodepool images","message":"WIP: CI: Move to CentOS 8 nodepool images\n\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python3\n\nChange-Id: Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e4cf806f78cdeb0cd18b40ff677e772f0c067b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e4cf806f78cdeb0cd18b40ff677e772f0c067b"}]},"branch":"refs/heads/master"},"e6ec6277e3254e717113ccec5c158947f7fdccb1":{"kind":"REWORK","_number":3,"created":"2019-11-14 10:43:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/692794/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/692794/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/692794/3"}}},"commit":{"parents":[{"commit":"eab03abfe22b1d97eefd037f6a53053b4e347480","subject":"WIP: move to CentOS 8 in containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab03abfe22b1d97eefd037f6a53053b4e347480"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:20:45.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 10:41:18.000000000","tz":60},"subject":"WIP: CI: Move to CentOS 8 nodepool images","message":"WIP: CI: Move to CentOS 8 nodepool images\n\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python3\n\nChange-Id: Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6ec6277e3254e717113ccec5c158947f7fdccb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6ec6277e3254e717113ccec5c158947f7fdccb1"}]},"branch":"refs/heads/master"},"f67ff19e1d72dec4bf2c66a08c419e6c500865bf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-25 15:12:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/94/692794/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/692794/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/692794/4"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:20:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-25 15:11:39.000000000","tz":0},"subject":"WIP: CI: Move to CentOS 8 nodepool images","message":"WIP: CI: Move to CentOS 8 nodepool images\n\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python3\n\nChange-Id: Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f67ff19e1d72dec4bf2c66a08c419e6c500865bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f67ff19e1d72dec4bf2c66a08c419e6c500865bf"}]},"branch":"refs/heads/master"},"9bfd38dd731c0f5ed8f027794fd46775cdee6748":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-11-26 10:20:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/94/692794/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/692794/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/692794/5"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:20:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 10:20:20.000000000","tz":0},"subject":"CI: Move to CentOS 8 nodepool images","message":"CI: Move to CentOS 8 nodepool images\n\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python3\n\nChange-Id: Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bfd38dd731c0f5ed8f027794fd46775cdee6748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bfd38dd731c0f5ed8f027794fd46775cdee6748"}]},"branch":"refs/heads/master"},"6c86915597c00da845bbfa433e897b1573f50c70":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-11-26 10:22:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/94/692794/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/692794/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/692794/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/692794/6"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 10:20:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 10:22:06.000000000","tz":0},"subject":"CI: Move to CentOS 8 nodepool images","message":"CI: Move to CentOS 8 nodepool images\n\nThis change modifies the CentOS build CI jobs to use a CentOS 8 host\nimage.\n\nNOTE: This does not affect the version of CentOS used as the base image\nin containers built by kolla.\n\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python3\n\nChange-Id: Ie25b54a4e6116ed2bcb9c3b5c44eb6edbd2e6b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c86915597c00da845bbfa433e897b1573f50c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c86915597c00da845bbfa433e897b1573f50c70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
