)]}'
{"id":"openstack%2Fkolla~692796","triplet_id":"openstack%2Fkolla~master~I2ccbfe4ae55c6a3653d8c03a12e6bca64e58d396","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I2ccbfe4ae55c6a3653d8c03a12e6bca64e58d396","subject":"Fix centos qdrouterd for centos \u003e 7","status":"MERGED","created":"2019-11-04 10:37:30.000000000","updated":"2019-11-06 10:15:47.000000000","submitted":"2019-11-04 14:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692796-1572879173657-9a54f4b4","meta_rev_id":"3d87524203a6dd6be975ea472ac4cccef14d4c48","_number":692796,"virtual_id_number":692796,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-04 11:41:44.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-04 14:52:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-04 11:37:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-04 11:00:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-04 12:45:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2019-11-04 13:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2019-11-04 12:44:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-11-04 11:01:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-04 11:02:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-11-04 13:28:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 11:00:57.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-04 11:02:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-04 11:37:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-04 11:41:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-04 12:45:03.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-11-04 13:25:38.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2019-11-04 13:28:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-04 14:52:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"598d85b7502990dcac211bca1919b58a8ab1cbc5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-04 10:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbd2f6c9bf0030a52efb4358a3ba431cee0f4d60","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-04 10:39:12.000000000","message":"Patch Set 1:\n\npart of https://review.opendev.org/#/c/692368/ already","accounts_in_message":[],"_revision_number":1},{"id":"6818e159e337ff2bac80a48ac2918374a7dc6867","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-04 10:40:53.000000000","message":"Patch Set 1:\n\n@Marcin Juszkiewicz this issue is critical as it broke our promotions for centos/rhel8 this weekend. Can we proceed with this part at least as soon as possible?","accounts_in_message":[],"_revision_number":1},{"id":"41a14358024a9f2d668f75e126b5fafcca32223d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-04 10:46:49.000000000","message":"Patch Set 1:\n\n@Marcin Juszkiewicz also that patch breaks building for centos7, which is not acceptable, it\u0027s still widely used.","accounts_in_message":[],"_revision_number":1},{"id":"66f2fb569bcaf7458a992573e09ac6dc89d50ec1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-04 10:48:34.000000000","message":"Patch Set 1: Code-Review+1\n\nI am not a fan of \"cmd1||cmd2\" approach but let\u0027s see what zuul says.","accounts_in_message":[],"_revision_number":1},{"id":"9ad0ba40d217f8a24b082fdf2c9f0a39e59b3d9c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 10:54:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9362d71e02cc79ad4ae5c2363a8fa2d73fb34a42","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-04 10:56:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43fda2e73df666570c3f1eb2eb2dcc02a7097293","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 10:57:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89e2851fa1a13f94c818a929a139a629752c18e9","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-04 11:00:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a13ec02e89486689c19b632a8ef185927c5bc42","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-04 11:00:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"535409db8ed25ba1f84ce6897cae97807097e0ba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 11:01:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd8878e6ac095f8a793c82c5c476ffd23105afe3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 11:02:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"24ed00a2a82555d6efa0403b1656d6d029e2e0ee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-04 11:37:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/96/692796/2/openstack-check/tripleo-build-containers-rhel-8-master/51df0b8/ : FAILURE in 36m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3d192bd551076c1c2b486f2fe80d3950f6ac1065","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-04 11:41:44.000000000","message":"Patch Set 2:\n\nI don\u0027t think it works: http://logs.rdoproject.org/96/692796/2/openstack-check/tripleo-build-containers-rhel-8-master/51df0b8/logs/buildah-builds/kolla-ccdp3mm0/docker/qdrouterd/qdrouterd-build.log.txt.gz","accounts_in_message":[],"_revision_number":2},{"id":"d985d8422f1fe3eaaab3f0df8dac7a8d93af15ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 12:08:11.000000000","message":"Patch Set 2:\n\n\u003e I don\u0027t think it works: http://logs.rdoproject.org/96/692796/2/openstack-check/tripleo-build-containers-rhel-8-master/51df0b8/logs/buildah-builds/kolla-ccdp3mm0/docker/qdrouterd/qdrouterd-build.log.txt.gz\n\nThat patch has not picked up the change: http://logs.rdoproject.org/96/692796/2/openstack-check/tripleo-build-containers-rhel-8-master/51df0b8/logs/buildah-builds/kolla-ccdp3mm0/docker/qdrouterd/Dockerfile.j2.txt.gz","accounts_in_message":[],"_revision_number":2},{"id":"8f883645dd946e4ebe309f189edc14d7c7bbb0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 12:43:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1b88a72039eb42afbc572d6aa4fc63bb : SUCCESS in 37m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a7099d77b4984fa2b320e6fdb7f25e57 : SUCCESS in 49m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1eefaa5ffa9e437a8eefa46ebc481a09 : SUCCESS in 1h 35m 53s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4790b26e477645cab8b36034cf39b41b : SUCCESS in 1h 25m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/21b68179a0114305ac8663f87ea099a4 : SUCCESS in 1h 37m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fa77ead513184eb7834eee7914d089ed : SUCCESS in 42m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/300f914a1e4b407f900b32d71e04fecb : SUCCESS in 25m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec0789786b354ba888d01d734c4543f6 : SUCCESS in 17m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3bad569fc3eb49b0af856601b6d88726 : SUCCESS in 1h 04m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce3e1802d38a45fabcdb1c53e0e9d553 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc0162aab294a7eb7e07ecb6d5e4f6b : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce1923f5f64d46fa9e4286311c8fb14b : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6544852160d4ac592455ed714338b4f : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efb4f32e3507428d9d5ca409985b5115 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb89499c2c64cf1a9edc3bc9ed4c04a : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d644429fcc3a4579a24b127d0ea4fc0b : SUCCESS in 24m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/18d5deab2cc8426dbf337ac7e1103f1e : SUCCESS in 37m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/909a046e034245fba645f918fb2b786c : SUCCESS in 1h 31m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/585bbbbd296d4a4db44822c137a8f2b6 : SUCCESS in 52m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96c7fc285ac84d6db2707a26bffb4edb : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f45e576a8ab65a0665f8a497dab1f3c4d0d7ca90","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-04 12:44:13.000000000","message":"Patch Set 2: Code-Review+1\n\nright, rhel8 job didn\u0027t pick it up, I\u0027ll work on it to fix so that we can\u0027t prevent such breakages, would be also useful for future centos8 jobs","accounts_in_message":[],"_revision_number":2},{"id":"db32856eee1d1c392cc9d424efded61ecc20a5fc","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-11-04 12:45:03.000000000","message":"Patch Set 2: Code-Review+1\n\nbut rhel-8 job is still failing here","accounts_in_message":[],"_revision_number":2},{"id":"1b1a29e1b94978d33d62edd1e1d4fd47755b9661","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-11-04 13:25:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1319ad53947a7fbdc0a3bbb7e01dba71f6bbf553","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 13:28:51.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4ef396df30a45b708125641e77d4de2ed1c0a41e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 13:29:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8d31ddcc5a2bf7aea20fc4585422ef1cc547c6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 14:52:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"83eb713fcffad471ad8fc258f59d82b2ef0212c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 14:52:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6253982a966044519ce5d4de965f9e76 : SUCCESS in 46m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8c0a915a80484ee197817f026f6b2253 : SUCCESS in 1h 22m 22s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1b808df47e27448eb1b7d699d59ecd91 : SUCCESS in 1h 14m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/536d3ebf01184609a8cd7345e8eb3dba : SUCCESS in 43m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d0a91807a8445e99d2823461185531a : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a559c390174fa19af9c0a57703ec47 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/580d87cd4b2e4abf8332ee5cd464c79b : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74b11eef3b5c41568b85dafa303f95be : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad889b825733468a9bd1c8aa2af151bc : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64064f5a71d6485f989f40316998790f : SUCCESS in 4m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/935dcb83c73942a684faab50c477278b : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d0a201e51274c2f9cf84df38c2999ab : SUCCESS in 56m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ccb2ffb3fe048e08599557595ff284d : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2fc60ba79a2e428e64fede3bc5ef1d4d8c9bfac5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 14:54:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7063ce3f5edc43cdb5edba86011a867a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2},{"id":"3d87524203a6dd6be975ea472ac4cccef14d4c48","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-06 10:15:47.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6b17bb2d3658ee5fccfc9e69f1ab1278a2452a29","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31a061099b0c518fc43ded1e1a675ad257b6d57d","revisions":{"0ca8c8e677fe6d563dde07b9461ae7b5facba0cf":{"kind":"REWORK","_number":1,"created":"2019-11-04 10:37:30.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/96/692796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/692796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/692796/1"}}},"commit":{"parents":[{"commit":"3ef9bf828ed225270c668d0ab396f4ecefdf9e31","subject":"Merge \"Bump OpenStack versions for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ef9bf828ed225270c668d0ab396f4ecefdf9e31"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-04 10:35:46.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-04 10:37:28.000000000","tz":120},"subject":"Fix centos qdrouterd for centos \u003e 7","message":"Fix centos qdrouterd for centos \u003e 7\n\nYum doesn\u0027t present in centos/rhel version 8, we\u0027ll need to\nuse dnf in that case. Make dnf first option with fallback to yum.\n\nChange-Id: I2ccbfe4ae55c6a3653d8c03a12e6bca64e58d396\nCloses-Bug: #1851135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca8c8e677fe6d563dde07b9461ae7b5facba0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca8c8e677fe6d563dde07b9461ae7b5facba0cf"}]},"branch":"refs/heads/master"},"31a061099b0c518fc43ded1e1a675ad257b6d57d":{"kind":"REWORK","_number":2,"created":"2019-11-04 11:00:17.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/96/692796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/692796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/692796/2"}}},"commit":{"parents":[{"commit":"3ef9bf828ed225270c668d0ab396f4ecefdf9e31","subject":"Merge \"Bump OpenStack versions for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ef9bf828ed225270c668d0ab396f4ecefdf9e31"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-04 10:35:46.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-04 10:59:57.000000000","tz":120},"subject":"Fix centos qdrouterd for centos \u003e 7","message":"Fix centos qdrouterd for centos \u003e 7\n\nYum doesn\u0027t present in centos/rhel version 8, we\u0027ll need to\nuse dnf in that case.\n\nChange-Id: I2ccbfe4ae55c6a3653d8c03a12e6bca64e58d396\nCloses-Bug: #1851135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
