)]}'
{"id":"openstack%2Fkolla~692870","triplet_id":"openstack%2Fkolla~master~I8fd93fa77ba9fc63910360300b8783a34bd4e6d5","project":"openstack/kolla","branch":"master","topic":"bug/1851169","hashtags":[],"change_id":"I8fd93fa77ba9fc63910360300b8783a34bd4e6d5","subject":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup","status":"MERGED","created":"2019-11-04 18:54:16.000000000","updated":"2019-11-11 09:38:17.000000000","submitted":"2019-11-07 10:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692870-1573122477137-7361ecdf","meta_rev_id":"ce7ea433b76cdb0e6e4373e53739cd913b2bb90e","_number":692870,"virtual_id_number":692870,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 10:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 15:00:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":2,"date":"2019-11-07 08:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-05 15:51:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":1,"date":"2019-11-07 08:47:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 15:00:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 15:51:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-07 08:47:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-07 10:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75b2d0be85c348f9a7aa99554b882861d5fd32e1","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2019-11-04 18:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6554ba7e3c4d2b751966adc44dd8a3cd76aa6e0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-04 19:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/70/692870/1/openstack-check/tripleo-build-containers-rhel-8-master/b431490/ : FAILURE in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff10b5caa0b08baee3def80ea9438c3a51919a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 20:30:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/caf08fe0ad4c41f78ebd3055e805256d : SUCCESS in 55m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7aca2e5a94db4521ae18bfd4a5c71d5d : SUCCESS in 52m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a8708400ec62422f8c7fc8eda7b475d0 : SUCCESS in 1h 30m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/60a9bd374ae341d4bf4652dd0a4f52be : SUCCESS in 1h 27m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8a6d3fe42b424a9eafbf32e511fb9651 : SUCCESS in 1h 30m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/84469a2e96534da4b6e45b5f6b47cf9b : SUCCESS in 51m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f54b2285d3f34ff68ba92952e60c1d0a : SUCCESS in 20m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e91b37f774da4be5a1d8ac7cd2a600c4 : SUCCESS in 18m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/07fbf7efa9534c0f846d097e0f68285b : SUCCESS in 48m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b648747c91ca4aad8fd650bd1d75d48a : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9cd7e3aa604c7cb75b58737a54b170 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f1525e6078c425fb168d8775ed414a6 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/568fb8e499d1437fba0417654cffe5a3 : SUCCESS in 2m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d2afe37219748b4b1b23b7ecaa62884 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd748d5ad1934dfbb82999a5efcbf0e5 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e099117945b74bcbb273a4ea3da64bc9 : SUCCESS in 23m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3951917fc3c4c10a39bed7d4ac9f29a : SUCCESS in 29m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ca56bc91745442eab8768de592e7457 : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f7c0492575848c385f848243f9d329f : SUCCESS in 54m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e03c11c19b98409e9c256a0123dcc7f7 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b050c148e6bdccc4894ef8e70972169ef9d735bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 10:21:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29912892352824d86309d48e3ef5216f1f74ed18","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2019-11-05 12:12:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1029e3e74f9ea83443d7df9a09985403663a45da","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2019-11-05 12:14:53.000000000","message":"Patch Set 2:\n\nGood catch, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"e1da672db5403b15fa3eb4590c30284711532a3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 12:51:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/70/692870/2/openstack-check/tripleo-build-containers-rhel-8-master/322ba24/ : FAILURE in 37m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1dd1b19d8642b149be55d11b44e5df54e2a050c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 12:52:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9abc2032c33a59152ba0e26f9a88efb00087ee94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:54:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/33bfee993b42420582f57e1077ec0a8b : SUCCESS in 48m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/479d5137340b408985ba32775a969b3d : SUCCESS in 53m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fa894119eb9b4940b796f830655dc5f8 : SUCCESS in 1h 25m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0d09fc6afc704b5e9beb5a279047a024 : SUCCESS in 1h 21m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7cf3d5a96100448b9bf43f701a89ceed : SUCCESS in 1h 37m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/090c2f4258094be2aec04539c41506e4 : SUCCESS in 59m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/58ac933d27234f478511b575a9204454 : SUCCESS in 24m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bfcd430ae84d414fad00678f2b5ac90e : SUCCESS in 18m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49e9b3bb610047b58b0e942d9b21347c : SUCCESS in 1h 06m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c39e7ff0f4948ff9562d26b5136612d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ba996a359d4537b8b0e42dbea54e0c : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f86a3b2d394a4c6996c25d932aed350e : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ca98ca54de4255a495ed4a84952de6 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f2a27ece8934c70867bcb9b349d5505 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5532eface7eb4babb329ce1149fe60d4 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2fec22bacc44fb49a081db0e6d687b2 : SUCCESS in 28m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc4453a2889a44ce9a4d0cf2ec554735 : SUCCESS in 29m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/47c2281db89540c984b47a6caff033c3 : SUCCESS in 57m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a457a96004244689b0abb9b114ff9c8b : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2f90c56a7bea44e9a7b974f8ef74f627 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c62a118337a3080d1db63e2b31076323e4109d6b","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2019-11-05 14:15:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a15c88d0c2946b88358584972a647f679a830a08","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 15:00:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/70/692870/3/openstack-check/tripleo-build-containers-rhel-8-master/e5d5b7e/ : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":3},{"id":"23706f6dd93243f0df9f557da1b7bba25df228a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 15:44:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2157971aac2143fcacdc7c824cda246f : SUCCESS in 59m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4299b790baab4c51a0aa4948e6f7dc64 : SUCCESS in 1h 02m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b47f58e34bc84f1899d5fc28dcfe373e : SUCCESS in 1h 27m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b95448fb9f0149cfb904a95556f49497 : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1d616dd1b0334e3aa3a3bd95255578ad : SUCCESS in 1h 23m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/53acdebab3ef46d7af9f0eb45143d6e6 : SUCCESS in 51m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9e46c58a8594462b2450f7512cff63b : SUCCESS in 28m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e000409d5e44777ad51f46f5bd62e12 : SUCCESS in 20m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/598e0607d330455ebeb2e4ba3c51b5ba : SUCCESS in 53m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68cd7be2a6964b119d75cc3709f6690e : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0791cbf15ee46b4a88e8d208e32df44 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e415873c4db42d886a0981c0edd1841 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f69c62a72b2f47f6a660ab475820fa25 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e49269864b474819bcbc3139a4995a10 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d381a5c94a53435e8fdb058ce89bf170 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c21bbd3b7f884b42821230a3863c9a3d : SUCCESS in 25m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d08ca3737491432c97cd2235401b6aab : SUCCESS in 31m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b68ae95484944f5a8cc31033bfa4c1d : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afb4d7e1a2ff491fbff8a8ee4c1c0770 : SUCCESS in 57m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d67f68d24611479e864e0840ee199a7e : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0d7738c652345d968ce8af9e35606feccc932b20","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 15:51:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"547a98b165c44db626b89d9aa37faaeecd61fc40","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-07 08:47:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b15bd3bf707a826969d68d201ce9f4850fd1557b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:47:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f64cf16febf09e6c92796b312a98a0494a6a1f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 10:27:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8cbae70506354fdcb2fa60b2f6275986 : SUCCESS in 45m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/df9cf710f4844dc780a63688fbdd3bae : SUCCESS in 1h 28m 15s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0e78ed38e7a74e94b6b80838fadaa171 : SUCCESS in 1h 34m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f0877ab001614b3d99bdcf5b9e3729fa : SUCCESS in 54m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d470bae748d46cfa1a0765967dc826d : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f308febe7d40108e0f5e964bffe149 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e948dbd314f4ad5b8196409ca9c81cb : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0748f7445e20407ba56d13882893a479 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44fea134769443deb1b1d9f2a8740874 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67da775a260f458cbca9748eb3d2f8f8 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1cf27705a8c4d12bd3e4c0539dd5d24 : SUCCESS in 27m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/033a69365b3b430f95cbfe455d1ef721 : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b9241e3ed08482986484379b050f941 : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":3},{"id":"3f3eea9fd28f327cbadbb26cb94a5f231f9c19e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 10:27:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7e5ed39f53f85811ed252093bccb12936c7c7d55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 10:29:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d1ef6e8fa04857999013d327aaac3c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3},{"id":"aa9428472cfa66fdbb9e6cc3ecedee0fe163c822","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2019-11-08 18:49:57.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 78762fabbd7440d699b6cb52c1da94a7529eb39c","accounts_in_message":[],"_revision_number":3},{"id":"45d2e53fa8c353dc479adbbedfab57a76c8289a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 09:38:17.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit daafc9380b326cb04ced3033dd8686673df3dc0d","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2d122afaf9817cadc668b308c4f736e47c09a55f","revisions":{"df9b4f1fd10eda140380fdf442e357564d332fe5":{"kind":"REWORK","_number":1,"created":"2019-11-04 18:54:16.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/70/692870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/692870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/692870/1"}}},"commit":{"parents":[{"commit":"31a061099b0c518fc43ded1e1a675ad257b6d57d","subject":"Fix centos qdrouterd for centos \u003e 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2019-11-04 18:46:49.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2019-11-04 18:52:48.000000000","tz":-300},"subject":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup","message":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup\n\nInstead of doing a recursive chown on /srv/node, chown /srv/node and\nthen chown anything under /srv/node non-recursively.\n\nChange-Id: I8fd93fa77ba9fc63910360300b8783a34bd4e6d5\nCloses-Bug: #1851169\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df9b4f1fd10eda140380fdf442e357564d332fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df9b4f1fd10eda140380fdf442e357564d332fe5"}]},"branch":"refs/heads/master"},"b8406a2daf2a598ecef19beef703a8202ad4d812":{"kind":"REWORK","_number":2,"created":"2019-11-05 12:12:45.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/70/692870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/692870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/692870/2"}}},"commit":{"parents":[{"commit":"31a061099b0c518fc43ded1e1a675ad257b6d57d","subject":"Fix centos qdrouterd for centos \u003e 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2019-11-04 18:46:49.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2019-11-05 12:10:57.000000000","tz":-300},"subject":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup","message":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup\n\nInstead of doing a recursive chown on /srv/node, chown /srv/node and\nthen chown anything under /srv/node non-recursively.\n\nChange-Id: I8fd93fa77ba9fc63910360300b8783a34bd4e6d5\nCloses-Bug: #1851169\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8406a2daf2a598ecef19beef703a8202ad4d812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8406a2daf2a598ecef19beef703a8202ad4d812"}]},"branch":"refs/heads/master"},"2d122afaf9817cadc668b308c4f736e47c09a55f":{"kind":"REWORK","_number":3,"created":"2019-11-05 14:15:40.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/70/692870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/692870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/692870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/692870/3"}}},"commit":{"parents":[{"commit":"31a061099b0c518fc43ded1e1a675ad257b6d57d","subject":"Fix centos qdrouterd for centos \u003e 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2019-11-04 18:46:49.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2019-11-05 14:12:46.000000000","tz":-300},"subject":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup","message":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup\n\nInstead of doing a recursive chown on /srv/node, chown only /srv/node\nand any immediate subdirectory under /srv/node.\n\nChange-Id: I8fd93fa77ba9fc63910360300b8783a34bd4e6d5\nCloses-Bug: #1851169\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d122afaf9817cadc668b308c4f736e47c09a55f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d122afaf9817cadc668b308c4f736e47c09a55f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
