)]}'
{"id":"openstack%2Fkolla~692978","triplet_id":"openstack%2Fkolla~master~If6f05237c2512c643857765eecf452f5416b3d60","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"If6f05237c2512c643857765eecf452f5416b3d60","subject":"Limit centos repo disable to centos hosts only","status":"MERGED","created":"2019-11-05 10:50:39.000000000","updated":"2019-11-05 14:01:02.000000000","submitted":"2019-11-05 13:58:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692978-1572962314493-aeac7fa8","meta_rev_id":"8ead586c2bd89a0b342b9b08d697f45e832ca6ad","_number":692978,"virtual_id_number":692978,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 13:58:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 11:29:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-11-05 11:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-11-05 10:54:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-05 11:52:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-05 11:52:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 10:54:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-05 11:29:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 11:32:13.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-11-05 11:52:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-05 13:58:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c1ca938f5dbefe2ca85a4d1fff29320f3696980","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-05 10:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46ed3c9477d955563cbe5f9c4fcbc0775d71aeda","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 10:53:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfe129230bea30a308f09f03dc515916079745d5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-05 10:53:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9c398eee6d1215699a6a39ff21d4647291329e42","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 10:54:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bbdb277ab30358a45514209be759019b867ac2d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 11:29:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/78/692978/2/openstack-check/tripleo-build-containers-rhel-8-master/3935cb8/ : FAILURE in 35m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a17644b959dce46ed388edad81f3ffd606defd59","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-11-05 11:32:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f6b52e8c36babde03b46f4bfcddac04789d12724","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-05 11:52:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca4b96a05407d90af3ff420c7a223376ab3478f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 12:24:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c648751537584635accb85ddf81b20ac : SUCCESS in 36m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a58cc21555864ebab5b9dbaed61235bf : SUCCESS in 48m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d8049e9896a6403eab4a5bd950ae15de : SUCCESS in 1h 19m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2d3766aa8a848658f435b91f2feaa13 : SUCCESS in 1h 11m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8e6995d5055f4fc8af2c465126a5d2a2 : SUCCESS in 1h 22m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3f44e443769e487e884713c7690073e0 : SUCCESS in 46m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e0091e67ec554508b56445530c2d4895 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9c21d929b071495986cb5e0b82c43f18 : SUCCESS in 22m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/172b5a189a0049b68cd9d4b97cf5048e : SUCCESS in 51m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ce7d0ebe7534cebac7aea3e0b3baf58 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d996aa88674209b3bf3a91f7f38e8e : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18b32bee5e7948b1be9acc2f986158c1 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d89c5cde3f2c45d3a21f5edb1cc6a234 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff22741b5a934f0684ab787e6c76144b : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e74481378814df9af1b53638b0a3b56 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05418f6171bf413ab63f2684cbf9ce96 : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9730716d0a7c47138d23cdeef377b45f : SUCCESS in 31m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48c1ea4af0dc4b5bb252a854b5f82d96 : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e868c8d437f14517b2ed9c77e09a6a00 : SUCCESS in 54m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de647df48597469eb9b796310b8367d4 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":2},{"id":"048e911e37e482d543d5f9b6c7b886ed5a226dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 12:25:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"466f501f7fb6fb28f33ad92ce12f3ff68aa90ef0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:58:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b948e06eaf68789dd9273e6fe1f0a37b6ce7bac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:58:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/be7112ebe42f499392c4a1cd2478959c : SUCCESS in 47m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bb834e331b8a4d0892618adab9efd1b9 : SUCCESS in 1h 19m 16s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e198d95f9a5b466a8af565be5865da56 : SUCCESS in 1h 21m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/626366977cb64d2b873df7e5cc55b51e : SUCCESS in 1h 00m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a848442eae774fec901e567f017f1885 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e078cb25b044bbaaf81e6976a56399b : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a230dea21d44b98b73c09fbc8538c84 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38400e2951242e9be310e6b65b648ff : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89eb12de56bd44a09f4198942188b526 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd4c78db27044c0890edc13f2b329d6 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f410f93348ea4c3db30e5c22a076b211 : SUCCESS in 40m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4e13f275c7b42d4baa296996ea661f4 : SUCCESS in 1h 08m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a1d347c3e22647e8b2238c13b7583122 : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8ead586c2bd89a0b342b9b08d697f45e832ca6ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 14:01:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0120f34eec2c4a7d83af24ff3464ad9b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4b37201a596dbe242887ba51d7092eaed11a0cf","revisions":{"342dce9a745876f396ad46571919045077b9fdc6":{"kind":"REWORK","_number":1,"created":"2019-11-05 10:50:39.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/78/692978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/692978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/692978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/692978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/692978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/692978/1"}}},"commit":{"parents":[{"commit":"31a061099b0c518fc43ded1e1a675ad257b6d57d","subject":"Fix centos qdrouterd for centos \u003e 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 10:49:42.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 10:49:42.000000000","tz":120},"subject":"Limit centos repo disable to centos hosts only","message":"Limit centos repo disable to centos hosts only\n\nRun dnf or yum to disable centos repos only if used on centos\nmachine.\n\nChange-Id: If6f05237c2512c643857765eecf452f5416b3d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/342dce9a745876f396ad46571919045077b9fdc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/342dce9a745876f396ad46571919045077b9fdc6"}]},"branch":"refs/heads/master"},"e4b37201a596dbe242887ba51d7092eaed11a0cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-05 10:53:48.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/78/692978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/692978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/692978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/692978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/692978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/692978/2"}}},"commit":{"parents":[{"commit":"31a061099b0c518fc43ded1e1a675ad257b6d57d","subject":"Fix centos qdrouterd for centos \u003e 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a061099b0c518fc43ded1e1a675ad257b6d57d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 10:49:42.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 10:53:28.000000000","tz":120},"subject":"Limit centos repo disable to centos hosts only","message":"Limit centos repo disable to centos hosts only\n\nRun dnf or yum to disable centos repos only if used on centos\nmachine.\n\nCloses-Bug: #1851135\nChange-Id: If6f05237c2512c643857765eecf452f5416b3d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4b37201a596dbe242887ba51d7092eaed11a0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4b37201a596dbe242887ba51d7092eaed11a0cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
