)]}'
{"id":"openstack%2Fkolla~692988","triplet_id":"openstack%2Fkolla~stable%2Fstein~I1c3b817712ae10edd45de01382b044af2fb728bd","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1840631-stable/stein","hashtags":[],"change_id":"I1c3b817712ae10edd45de01382b044af2fb728bd","subject":"Add effective and permitted capability to blackbox exporter","status":"MERGED","created":"2019-11-05 12:03:44.000000000","updated":"2019-11-07 06:18:18.000000000","submitted":"2019-11-07 06:16:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692988-1573107389407-2bb4bfd8","meta_rev_id":"3d277a5db6b32de94261e453a16ae463f93c4e04","_number":692988,"virtual_id_number":692988,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 06:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":2,"date":"2019-11-07 05:13:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-05 14:00:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":1,"date":"2019-11-07 05:13:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-05 12:03:44.000000000","_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 12:03:44.000000000","updated_by":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"reviewer":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"state":"REVIEWER"},{"updated":"2019-11-07 05:13:41.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-07 06:16:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9c500f77a69fe2da14dfab4a24b7254287c6bb3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 12:03:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0ef9505a5f24f61b2ae4523c6d275accf9b8ad07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4f8a80d8470d44af9327f73140165326 : SUCCESS in 39m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/938e828a34b24b249bab8725803a0cbb : SUCCESS in 54m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3a7a6f717e17421a9731a2a6b1b83a96 : SUCCESS in 1h 03m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ea38366faa3e4764afd813ccb3435bba : SUCCESS in 1h 03m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e6148b791a5e43aabe1bc1976cc335a8 : SUCCESS in 1h 42m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d61d3708a56a4d3f8a3a33a2aae88bfe : SUCCESS in 45m 13s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b6b3d936e27b450183e77eff2171a83f : FAILURE in 41m 08s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5ad8ed79ec7248a1ae46b3aa1f88b67d : SUCCESS in 57m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5ce5111be5141ea8598cf1e6a014393 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45aa41a78a54a8fbb5507cd4b81abd1 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4fe174a14174d2b9dad29b003e19b1f : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e422ace7c0fa49e587ee4f42707aef0d : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/decd1ef62d4a421593793d0aa2e53c6a : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730ada470b3a4bff954d9f19b9ac5b57 : SUCCESS in 2m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a454c54ef7b34723857553edb4425a59 : SUCCESS in 38m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f25d7e2d72844f8898079ab99ec95cbc : SUCCESS in 32m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e19cf1fa3977411fadab28a2f9539504 : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/238c22dada3145c1b8fde41ae20446ef : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dfd7db1fcd9f4163abaca00d3bef2c9e : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"80167cabf6d6fac7a6571981250cd3a04e5c2981","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 14:00:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59c1484819e0f59670a790d8a4cc12ce33de2033","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-07 05:13:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5eeb7cfe6c3bb89c4f691f35298973f17792e523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 05:13:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcc1ca115568aaf3073b5acd0bdadb852be0b49f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 06:16:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"399b5d824dcf398e3d0ac5b2f13d5236cece7ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 06:16:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b40015d33e1b4a3baf1728d83f94fa31 : SUCCESS in 50m 10s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a44677872d144202b892cd11473a2d8c : SUCCESS in 57m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2eb7a2b0779c4401a50997b738810e5c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972c7f9e68c1487890a659e2bcddfd6d : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/750765ee88854d819b4d7794e9a20240 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3d5e4a769fe4e33abfa04d4412e6cf7 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac886446af634cc2a9d383b4493008d9 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60759346fed64f7196134330f47e1d0c : SUCCESS in 3m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a344a7aaef44d2ab5a8b021fded0baa : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"441772b83d4d9edad764787e7be2e8abb0e481de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 06:18:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4995d6a98ff467098a8f4d28d4237b5 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e712fadd3d7e12163455a74a8076791d925de11","revisions":{"8e712fadd3d7e12163455a74a8076791d925de11":{"kind":"REWORK","_number":1,"created":"2019-11-05 12:03:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/88/692988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/692988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/692988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/692988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/692988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/692988/1"}}},"commit":{"parents":[{"commit":"77028db9077a7162f698aec388f7489778c3398e","subject":"Merge \"Bump OpenStack versions for Stein\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77028db9077a7162f698aec388f7489778c3398e"}]}],"author":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2019-08-16 16:10:33.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-05 12:03:44.000000000","tz":0},"subject":"Add effective and permitted capability to blackbox exporter","message":"Add effective and permitted capability to blackbox exporter\n\nThe ICMP probe in the Prometheus blackbox exporter requires\nelevated privileges to function. Linux root user or CAP_NET_RAW\ncapability is required. Can be set by executing setcap\ncap_net_raw+ep blackbox_exporter. This change also bumps the version\nof the blackbox exporter to the next minor version to allow for this\nsupport.\n\n\u0027+ep\u0027 means we\u0027re adding the capability as effective and permitted.\n\nSee: https://github.com/prometheus/blackbox_exporter\n\nChange-Id: I1c3b817712ae10edd45de01382b044af2fb728bd\nCloses-Bug: #1840631\nCo-authored-by: Jack Heskett \u003cJack.Heskett@gresearch.co.uk\u003e\n(cherry picked from commit b5a4478b0fa9b775a9b5c45545e78e9215a50588)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e712fadd3d7e12163455a74a8076791d925de11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e712fadd3d7e12163455a74a8076791d925de11"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
