)]}'
{"id":"openstack%2Fkolla~692993","triplet_id":"openstack%2Fkolla~stable%2Fstein~I7bbdcbd5c60f654a484a872209e0537e3950346c","project":"openstack/kolla","branch":"stable/stein","topic":"686963-stable/stein","hashtags":[],"change_id":"I7bbdcbd5c60f654a484a872209e0537e3950346c","subject":"get \u0027apt_*\u0027 options of kolla-build.conf working again","status":"MERGED","created":"2019-11-05 12:20:02.000000000","updated":"2019-11-07 09:47:44.000000000","submitted":"2019-11-07 09:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692993-1573119974915-dfbf8c2d","meta_rev_id":"8bfaf1867b679700aa72c7e733982297c4def5fe","_number":692993,"virtual_id_number":692993,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 09:46:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-11-07 08:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-05 13:59:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-11-07 08:45:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 08:45:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-07 09:46:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1572d5b1d0158ffa93c714ecc25f4d65389ad6bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 12:20:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"419bdc959fd541ef85a9300552c26565935026aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:57:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3405726336054ded93fd60c3a50691ff : SUCCESS in 50m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2af387606d4a4f05842aa430731358bd : SUCCESS in 48m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/828d8b7879a44a4391f7a5dbaddb0abd : SUCCESS in 1h 22m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f15d8425104846fb975c894525193bb7 : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8100f795425445999ef5b7d4a6a51d83 : SUCCESS in 1h 32m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b4e4e3c5abfd499c955175a422b9c8ed : SUCCESS in 58m 14s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5ecab2978ee046b3899fffc3339d2b43 : FAILURE in 52m 53s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f7509e0ba89941fa9bd81c5a05adb7bf : SUCCESS in 57m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd0447841eef484b94dc413fde3677d1 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5834e2ac56944c2a007963b6eee7e01 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ff250db18d24a39953773aedb059ee7 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fe10a83737142e5b2f71cd226914c77 : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22a9b12769534fe3b824b3e996061ce5 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115a385596f149e39f5503017d84754d : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73aae48640a343fd977516772d60e9f3 : SUCCESS in 26m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/69a3418b86bb457d947fec9311d2cd03 : SUCCESS in 33m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/492d88f2c8c74269947f2529371f7494 : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a1f986fe13d46028a20182ad8a51071 : SUCCESS in 50m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39bcca92b761471695398a7c15cc869b : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"abf639890dbd17e04fdb706845fba41a01403fe7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 13:59:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9f45efcb2a9d1c0b38cfde7e20120a6cf0b0db7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-07 08:45:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e00141752e21c23df2e2408324a00c36a9f32c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:46:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b3d95a5bb72c706d6076f35ff5593b835bd314d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:46:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24a5cd01197c2a5cfd4ee63cb0437fde58ba5635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:46:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/72c27eed857d4b078e0084f54d1cc7f8 : SUCCESS in 55m 37s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3a312176b9cb4fccac37a1f576a29e5d : SUCCESS in 43m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8a3ea85087c4598bef40f604cd1d2e8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7737bdcf428d4e11be726d4e5f8db7ec : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/278959dc08d74ba2a9c6d4efb20b583b : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5c4185ff84641f0b1634953d9905da0 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4871e251270a4c83942c66ade4c69192 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe01a2f51d847be9f060a73b5ad0c98 : SUCCESS in 3m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4d732f0b166478481999399d6b50a8e : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"016267cfbcf330702dd51ef4ec04034ca0545498","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:47:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197fa73888db42c4825b3eacadd417b3 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f320e34ea3e5f1b8dcc188a082071be7a49c1d0","revisions":{"0f320e34ea3e5f1b8dcc188a082071be7a49c1d0":{"kind":"REWORK","_number":1,"created":"2019-11-05 12:20:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/692993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/692993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/692993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/692993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/692993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/692993/1"}}},"commit":{"parents":[{"commit":"77028db9077a7162f698aec388f7489778c3398e","subject":"Merge \"Bump OpenStack versions for Stein\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77028db9077a7162f698aec388f7489778c3398e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-17 10:36:17.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-05 12:20:02.000000000","tz":0},"subject":"get \u0027apt_*\u0027 options of kolla-build.conf working again","message":"get \u0027apt_*\u0027 options of kolla-build.conf working again\n\nWe broke \u0027apt_sources_list\u0027 and \u0027apt_preferences\u0027 options quite a while\nago. Turned out that fix is quite simple.\n\nThis change adds empty sources.list and apt-preferences files which are\nalways added into images. If kolla-build.conf will use any of APT\nrelated options then those files will be replaced with provided ones.\n\nChange-Id: I7bbdcbd5c60f654a484a872209e0537e3950346c\nCloses-bug: #1706549\n(cherry picked from commit bf20b41e7dea702b693fa98ea27822c8a33a5591)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f320e34ea3e5f1b8dcc188a082071be7a49c1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f320e34ea3e5f1b8dcc188a082071be7a49c1d0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
