)]}'
{"id":"openstack%2Fkolla~692996","triplet_id":"openstack%2Fkolla~stable%2Fstein~Idf1aed905a7d9ad461dd6fe17ed199c8d60370e6","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1843043","hashtags":[],"change_id":"Idf1aed905a7d9ad461dd6fe17ed199c8d60370e6","subject":"Stein: support mariabackup for backups","status":"MERGED","created":"2019-11-05 12:33:48.000000000","updated":"2019-11-07 09:49:46.000000000","submitted":"2019-11-07 09:48:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692996-1573120119455-cd6aae0a","meta_rev_id":"e3abc726016b4d6229176794420a0c7a1affd039","_number":692996,"virtual_id_number":692996,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 09:48:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-11-07 08:46:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-07 05:14:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-11-07 08:46:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 05:14:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-07 08:46:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-07 09:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0092ed93246a462abe91d65354f8fcb5b406761","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 12:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5aa8b55d3648ec6d44789bb0a829f1e24b762b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:58:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9363a52ab9974625a1f273927d55709b : SUCCESS in 38m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b308c625682040c4b3e62f5ab97f0390 : SUCCESS in 53m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4c42d0091c764b87a57bba6c3409f527 : SUCCESS in 1h 18m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/08cd7ac8df4e4182bd7e81fbd72e57ab : SUCCESS in 1h 10m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f23ae3433933448bb1ddf17fb6d88be9 : SUCCESS in 1h 24m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c2281ed6d48d43dba9f0e9891cc1d21c : SUCCESS in 53m 25s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ba140a74448f40a19ebeadd73eb322b0 : FAILURE in 47m 06s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8f6337da05bf4aa1aa7655cd608ee6ff : SUCCESS in 1h 00m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a887bc75e342fbaae524a5960c79e9 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e30397cc84446b7a4abe9bb8bac5d69 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcfdb89c678e43209247e37ebac2a25d : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01dc2aef63b24c29a301521030dfae8c : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ad0778ca1b7426c8d1789dca115452b : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2baeb995f7748978c4cc08d0015eb0c : SUCCESS in 3m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7481c89246f4eb9810811832b094fa4 : SUCCESS in 23m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ff68347871e4078bf44d049d61cd5f5 : SUCCESS in 28m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fef1d3e3bd246deb174ea8ec6cb1ad7 : SUCCESS in 52m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d03af5c1e0a44e4a26002e641e58d73 : SUCCESS in 46m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3fd9a5a30b8e436e88792dee0b3646e0 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":1},{"id":"95e11c9e069d603926fea3b60889741fad88e72f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-07 05:14:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c02086ecbfbeb63e97e6943c6fa50126ccdde573","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-07 08:46:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3504e3ad94f26b7bdf9e8d47aab82138e52a7808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:46:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c7cdb1c37cc09eb77344db74178b05c4e89d851","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:48:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83802b1e30d437faea932989a9aafd292a101869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:48:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8987a5bc3e6243d39b4cb4ecf95819c2 : SUCCESS in 1h 01m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ec810470fc8f48968bd5bba3845b5d14 : SUCCESS in 1h 00m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eecbdb9635c043808649b1895132d3b1 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6014f14b3c474ee099f5754cfdb97bee : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02a8443133854ad194459fb5e77d1a70 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c25f3f05d8a74fdbaaa2965a9ca5e57f : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5ec186d658c4ca5a39dcd1c9ef21371 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816d7eadb7114b26b636819a87bc2f87 : SUCCESS in 4m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e6cc2d4e0e246f4900a7d3d10f2ae77 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3704df94ed04ce441b1258cd40f207a5cb2ee026","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:49:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e94d4e1d18c4aeea5c484659d5944a5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25890a485b20f9530f769c23af841d6a50593b37","revisions":{"25890a485b20f9530f769c23af841d6a50593b37":{"kind":"REWORK","_number":1,"created":"2019-11-05 12:33:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/96/692996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/692996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/692996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/692996/1"}}},"commit":{"parents":[{"commit":"77028db9077a7162f698aec388f7489778c3398e","subject":"Merge \"Bump OpenStack versions for Stein\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77028db9077a7162f698aec388f7489778c3398e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-12 09:48:07.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-05 12:32:27.000000000","tz":0},"subject":"Stein: support mariabackup for backups","message":"Stein: support mariabackup for backups\n\nThis is a combination of 3 commits from Train. Rather than dropping\nxtrabackup, we are keeping it in addition to adding support for\nbackups using mariabackup.\n\n1. Drop Percona XtraBackup in favour of MariaDB Backup\n\nXtraBackup started failing to install on Ubuntu and does not support\nMariaDB 10.3 version.\n\nRelated-Bug: #1843043\n\nI9cf62c8e03a02574d85e5349df1d23a51d4c99be\n\n2. Fix mariabackup arguments\n\nAfter switching from innobackupex to mariabackup [1] for database\nbackups, the kolla_mariadb_backup.sh script was not modified to account\nfor the difference in arguments between innobackupex and mariabackup.\nThere is a compatibility mode, the documentation [2] for which covers\nsome of the differences.\n\nThe following have been changed:\n\n- Add explicit --backup argument, now required\n- Remove \u0027./\u0027 positional argument - not required with --stream\n- Remove --no-timestamp argument - only supported in innobackupex\n  compatibility mode\n- Remove --incremental argument - implied by --incremental-history-name\n- Remove deprecated --compress argument, which requires qpress to be\n  installed (it is not). The stream is now passed through gzip instead\n  [3]\n\n[1] https://mariadb.com/kb/en/library/mariabackup-overview/\n[2] https://mariadb.com/kb/en/library/mariabackup-options/#-innobackupex\n[3] https://mariadb.com/kb/en/library/using-encryption-and-compression-tools-with-mariabackup/\n\nI67cff47cbf56570b8eaeb66092dd87c2769fc8a6\nCloses-Bug: #1843043\n(cherry picked from commit 65b27d772be32738a8508ea8e6c07f74f229cdb7)\n\n3. Set pipefail in kolla_mariadb_backup.sh\n\nWithout this, a failure in mariabackup would be ignored due to the pipe\nto gzip.\n\nI84536b302991e8fb04d5fa67ea266bc428437ce1\nRelated-Bug: #1843043\n\nChange-Id: Idf1aed905a7d9ad461dd6fe17ed199c8d60370e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25890a485b20f9530f769c23af841d6a50593b37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25890a485b20f9530f769c23af841d6a50593b37"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
