)]}'
{"id":"openstack%2Fkolla~693056","triplet_id":"openstack%2Fkolla~master~I67ebbbbcb63c8edef77a4cf38348e50feea96fbb","project":"openstack/kolla","branch":"master","topic":"test3","hashtags":[],"change_id":"I67ebbbbcb63c8edef77a4cf38348e50feea96fbb","subject":"DNM: test 3party with patch","status":"ABANDONED","created":"2019-11-05 17:02:47.000000000","updated":"2019-11-06 11:56:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a87c57c08ff6013d830b3061724c8dd1edd2913","_number":693056,"virtual_id_number":693056,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-11-06 11:24:09.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-06 11:22:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-06 11:51:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-06 11:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-06 11:51:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"163d237d65aa3c8871489d629899bdb8c43570e4","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-05 17:02:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73e4e62b956fc9aed2dda7b16a9e87d70e92d248","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 17:39:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/693056/1/openstack-check/tripleo-build-containers-rhel-8-master/774d374/ : FAILURE in 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8afe180cbe5a14c92f191d4d6e68fee8150a6f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 18:00:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0fd9c21957a94edcbb9ef855ec5764ee : SUCCESS in 52m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dc816cf600349738022d3b4f1c82295 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1e5fd5197f4a55a5738dfa2a067719 : FAILURE in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f61e907c6314d59adfad3bcacd32012 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ba3f4e103ff484e84cec857a3119ca2 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2286bb2277164c87bd3f57c83b62d482 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9905cde21804279a03357879e2e4c00 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"77a96affe2c80a6932b6820fb8213dd235d36141","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-06 10:15:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f4efacba26e351c79750c5982b1a791fef92a2b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-06 10:41:07.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"f0d2f9dc4fc66543d69a89b18723e2aae519714b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-06 10:41:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"257cc60c1c84700f1eb7885e7f5e9faf2925f95b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-06 11:10:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/693056/2/openstack-check/tripleo-build-containers-rhel-8-master/8d3b6c0/ : FAILURE in 28m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f6d200be3609e838d7a791781b654ec1275e18a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 11:22:02.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"abbf1e1139740a18e5eba91d7c2f3f65c3fa4704","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-06 11:24:09.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"ad512f591e338dcfbb22fdc4a5c8bf547d2a26d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-06 11:51:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/693056/2/openstack-check/tripleo-build-containers-rhel-8-master/00b8f23/ : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e7ff8e8a60af59b6dc072588382824061bd094a6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-06 11:56:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a4056079b14fce8229c2f10e09278b2b2ad019f","revisions":{"5eb3a7b3490acf05aa282a66eec7ecf6e67942c8":{"kind":"REWORK","_number":1,"created":"2019-11-05 17:02:47.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/56/693056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/693056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/693056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/693056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/693056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/693056/1"}}},"commit":{"parents":[{"commit":"a7e8360f53d3daebc4991970898eaa3cfdd4f1e5","subject":"Merge \"Limit centos repo disable to centos hosts only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7e8360f53d3daebc4991970898eaa3cfdd4f1e5"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 11:32:15.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 17:02:11.000000000","tz":120},"subject":"DNM: test 3party with patch","message":"DNM: test 3party with patch\n\nDepends-On: https://review.opendev.org/693048\nChange-Id: I67ebbbbcb63c8edef77a4cf38348e50feea96fbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5eb3a7b3490acf05aa282a66eec7ecf6e67942c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5eb3a7b3490acf05aa282a66eec7ecf6e67942c8"}]},"branch":"refs/heads/master"},"3a4056079b14fce8229c2f10e09278b2b2ad019f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-06 10:41:17.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/56/693056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/693056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/693056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/693056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/693056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/693056/2"}}},"commit":{"parents":[{"commit":"2a17ec861adc94ef6cc2c5192a18ebdf3317ff3e","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a17ec861adc94ef6cc2c5192a18ebdf3317ff3e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-05 11:32:15.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-11-06 10:41:17.000000000","tz":0},"subject":"DNM: test 3party with patch","message":"DNM: test 3party with patch\n\nDepends-On: https://review.opendev.org/693048\nChange-Id: I67ebbbbcb63c8edef77a4cf38348e50feea96fbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a4056079b14fce8229c2f10e09278b2b2ad019f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a4056079b14fce8229c2f10e09278b2b2ad019f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
