)]}'
{"id":"openstack%2Fkolla~693400","triplet_id":"openstack%2Fkolla~master~Ia592dda4ee20979c701efe6007f236dba9848997","project":"openstack/kolla","branch":"master","topic":"bug/1851692","hashtags":[],"change_id":"Ia592dda4ee20979c701efe6007f236dba9848997","subject":"[manila] Fixes manila-share on debian/ubuntu","status":"MERGED","created":"2019-11-07 18:04:51.000000000","updated":"2019-11-11 08:39:21.000000000","submitted":"2019-11-08 13:53:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693400-1573221192485-8bdcbf74","meta_rev_id":"20bb8d0a84247e31b73e925b9a2492a103b98da1","_number":693400,"virtual_id_number":693400,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2019-11-08 12:34:19.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 13:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-08 11:13:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-11-08 11:20:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-08 12:33:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-08 12:33:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 11:13:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-08 11:20:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-08 12:33:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-08 13:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55e845c206498b76f1a09c68dc7105580d951531","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-07 18:04:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"120a409557529df3de1e640f73763a7d6846c57e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-07 18:51:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/00/693400/1/openstack-check/tripleo-build-containers-rhel-8-master/9e1886e/ : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2073554a70723f1ea543232341ee9eb817566056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 19:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0aeec89708914e03baccba8c73d75675 : SUCCESS in 45m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d415f2926c964225993bbff569a16f27 : SUCCESS in 53m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6276d216df974c2f90170058eebe9337 : SUCCESS in 1h 25m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ad2a655adbb549e09c9958ca22db42af : SUCCESS in 1h 23m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e944b620b2b6489bb6858af0bbf7ae33 : SUCCESS in 1h 48m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/648e38213bb1433cbfdad79e0e88e689 : SUCCESS in 1h 25m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b6e73d31e4d4b78b880406f8b55197e : SUCCESS in 48m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e4989fa521541389d429994c7ce7064 : SUCCESS in 46m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/28fcc2edf5994d3f9c72ef331f9b5b34 : SUCCESS in 1h 19m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0a1547c4e2340d4b7f7bc302847ee1e : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/182a9d31d0554919bbcb464928258b11 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/670511691d0c4ba481106cc2ed990ebf : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41f47796ddf84cd5866075d282946a23 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7b6dcc2ef3748bea577332c9888fa0c : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b6ea609b1a400cb758be02889b0e01 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a36c7baa483444e0b6aff1f8320130aa : SUCCESS in 52m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e08b68d5cc348fa8be604bc8f77994b : SUCCESS in 39m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8466520ba5fb485ea50edcfd8243c793 : SUCCESS in 58m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cae5b66608dd437fb22941a713226d90 : SUCCESS in 1h 22m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9b6c5aa58ad469bb7087b2f6edd0211 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":1},{"id":"93b0152b8234af6de836ae345512075db5175066","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-08 10:25:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"994cc3237ead41f5f21472e45cd1309f1372bc14","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-08 10:36:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"68b43fd3d6e2471c85f6b59d91776256fdaafcb1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-08 11:13:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/00/693400/2/openstack-check/tripleo-build-containers-rhel-8-master/b1ed1b7/ : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d5084a70a1af597967c297011f07aa2d210e8ef8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-08 11:20:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6ce08faecb542397a5f4aec784774ff0d4f62034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 12:30:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ff323f76ed6d49b2b5dcd051682a070d : SUCCESS in 44m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c61dac7bd568483c9cb0e793a76cb3a2 : SUCCESS in 54m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/deea17d404d8456e9b0648f4f9eccc33 : SUCCESS in 1h 30m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cdfb1cd45dc2495f94438a5661ef622a : SUCCESS in 1h 46m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a09baea9e7c34904b24f280371f59a40 : SUCCESS in 1h 25m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d835087bce964fbbaa19842f88997938 : SUCCESS in 53m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/da937419f1e64a158e795e9d218a25d7 : SUCCESS in 23m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f7a5545e1554b0a817eda42f4034797 : SUCCESS in 19m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/48d62ff44e964db8b7f3b27d2bedc359 : FAILURE in 27m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e67692b869a047cfa071a4ff875d7cd5 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389c415a503b43a8b0ad88c34f25f1eb : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0953444fd0249e9b269b366c6e5e128 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a3f2175807f41a39faccedde5eb40ca : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5ffe1e464e34914a89062421b15b7a9 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab3263b88a84ee28bbb4e7462f9ea5f : SUCCESS in 5m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37f05acb1fb64a86b1f828db4565c6b1 : SUCCESS in 26m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22f2b2e259da470281d1b19b93af368a : SUCCESS in 44m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/56133bbe32b7417492b7b378eec4ac9b : SUCCESS in 55m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3fd672d7f744616bebbae6dde60fa23 : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/83e467e92e2a4c6187e78a138a3eed0c : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c1345963414397ef37a965707ea2c61774fd8aa5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-08 12:33:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e905e590d3bf141014aa15c0654e9a891831f4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 12:33:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2b08780c63378965a3dccd59f7bdcf33d46cf5f","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-08 12:34:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0e1e29aad22bf49299e57bc6c15781dfe37b6a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 13:53:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7d9ee18022df9e77212f0ec2cc50508de889f496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 13:53:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f88f8a1948f14770b05bf60fdd9cbb64 : SUCCESS in 52m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/184c4bcf51fb46849973801cf60f6055 : SUCCESS in 1h 17m 06s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/089c1fbdb1e147c380e89007dcf426cf : SUCCESS in 1h 18m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9114e152c0e746709a62887f3127bd23 : SUCCESS in 47m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6637d9f5981a463bb4ec6b161d64a6a6 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaab3e5ba39e44fb8c288b7197bbd311 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b0da07650314305bce2cdab9d217679 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40124523991c409dbc34574150253a70 : SUCCESS in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7727125bd1840fdbb7459c63a9fc942 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0936dccbb9b24c51a0a98ac859c7f45d : SUCCESS in 5m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/91738b15c8f5421b8c17e4ea3c859383 : SUCCESS in 29m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b3689d1dbbc44839da4c0602cac0b83 : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d5158003c04d4d15a9a51501ce1fed77 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a361b8dbe42a946d9aebe559443332f32eae11d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 13:54:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29f92f88fa14aa0ae96de97cd366579 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"17c0c8c48bed2d039d8c1b093aa6034770203ee3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-11 08:39:21.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit cc615dd041e9e6b82bd8b34fc87d34fe6878386f","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d5a208c2e5c81ed4169c63d1d5ea0ad31b78f2f","revisions":{"b7b2771661303b390f26a3d35ef44729c261f67d":{"kind":"REWORK","_number":1,"created":"2019-11-07 18:04:51.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/00/693400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/693400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/693400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/693400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/693400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/693400/1"}}},"commit":{"parents":[{"commit":"3e8bef4e40e0e00bc27533259843a564d7ecd852","subject":"Merge \"swift-rsyncd: don\u0027t recursively chown /srv/node at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e8bef4e40e0e00bc27533259843a564d7ecd852"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-07 18:02:53.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-07 18:02:53.000000000","tz":180},"subject":"[manila] Fixes manila-share on debian/ubuntu","message":"[manila] Fixes manila-share on debian/ubuntu\n\nmanila-share cannot start on debian/ubuntu because python3-cephfs is missing.\n\nChange-Id: Ia592dda4ee20979c701efe6007f236dba9848997\nCloses-Bug: #1851692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7b2771661303b390f26a3d35ef44729c261f67d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7b2771661303b390f26a3d35ef44729c261f67d"}]},"branch":"refs/heads/master"},"1d5a208c2e5c81ed4169c63d1d5ea0ad31b78f2f":{"kind":"REWORK","_number":2,"created":"2019-11-08 10:36:33.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/00/693400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/693400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/693400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/693400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/693400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/693400/2"}}},"commit":{"parents":[{"commit":"3e8bef4e40e0e00bc27533259843a564d7ecd852","subject":"Merge \"swift-rsyncd: don\u0027t recursively chown /srv/node at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e8bef4e40e0e00bc27533259843a564d7ecd852"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-07 18:02:53.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-08 10:36:15.000000000","tz":180},"subject":"[manila] Fixes manila-share on debian/ubuntu","message":"[manila] Fixes manila-share on debian/ubuntu\n\nmanila-share cannot start on debian/ubuntu because python3-cephfs is missing.\n\nChange-Id: Ia592dda4ee20979c701efe6007f236dba9848997\nCloses-Bug: #1851692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d5a208c2e5c81ed4169c63d1d5ea0ad31b78f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d5a208c2e5c81ed4169c63d1d5ea0ad31b78f2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
