)]}'
{"id":"openstack%2Fkolla~693830","triplet_id":"openstack%2Fkolla~master~I2fe57f7cd79724fa8a36bd702d75b013a68ecd1b","project":"openstack/kolla","branch":"master","topic":"remove-deprecated-images-ussuri","hashtags":[],"change_id":"I2fe57f7cd79724fa8a36bd702d75b013a68ecd1b","subject":"Remove deprecated images","status":"MERGED","created":"2019-11-12 12:15:26.000000000","updated":"2019-11-13 21:39:25.000000000","submitted":"2019-11-13 21:37:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693830-1573681029107-99007b45","meta_rev_id":"d09add802e065cd8dea1deab3d219620db2ee9b2","_number":693830,"virtual_id_number":693830,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-11-13 08:27:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 21:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-13 08:57:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-11-13 19:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-13 08:37:03.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-11-13 19:59:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 16:34:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-13 08:37:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-13 08:57:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-13 19:59:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-13 21:37:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f54781c81ddef4146d2e2902108753fc6da3819f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-12 12:15:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f70115417f875a3b045bc200bcb5e3d1a498f4d2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-12 12:19:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a950767d6518c1d2e9960c83af0264e20afb7819","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-12 12:20:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90851181607efd8000200beeefe28bfee84eb0d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-12 12:22:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6a80774b4353ad6628369b93661a5e6da15e3ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-12 12:22:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"16f6c7907c8cbe861b35bd6782bb4e7dffe5b6dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-12 12:54:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/30/693830/3/openstack-check/tripleo-build-containers-rhel-8-master/94d72b6/ : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ae9013ee68aba4e48b51216101722a2cb5f2b4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:57:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0eb0d95e49524c3495d6718c43bcca22 : SUCCESS in 44m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fc544f769398411789538ce4ecad7973 : SUCCESS in 42m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/378a219bcc4947ae9ebcd6ff16dbd957 : SUCCESS in 1h 33m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f63dd108f6174155aa2fe640683b1281 : SUCCESS in 1h 21m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/eea5f67564e64aab87cbd6a8a6dd8fb0 : SUCCESS in 1h 16m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/738eae037bb1462f94e255c303b8c48f : SUCCESS in 49m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df6bcfb10fe048c2b088e3107899786a : SUCCESS in 21m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4fabd653523e42b4aea6de8c87a5879a : SUCCESS in 26m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f49f9e6af1c24e7f8b3cba628b52a59c : SUCCESS in 57m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/511f90d7510d4116a12e260948107882 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9faff5f902004477bd833c698a7d75fe : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf70a351c62048738e82d96f5c72ed99 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c60e0d835624eb989b3ad620526dbf2 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d9e4d64dd244113b64945e24cf81fcb : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288e34e562f24a11a2ceb62b06bf11cd : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c18522e0d9b4d22a8f9b0893e0f4251 : SUCCESS in 7m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7245adb75194819a8f93f7ace5316c7 : SUCCESS in 28m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd899516fcc749bab8f466a85877c825 : SUCCESS in 49m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9955f3ece6414da8b24ea10d6a97b341 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/653504a5297a4e5ba6303b510c05b19a : SUCCESS in 1h 19m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ac9052b37744045b2780c38387fff63 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":3},{"id":"0bc4bff4c6c7627b6f6d773e81dd97e1d8abe4a3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-12 16:33:05.000000000","message":"Patch Set 3: Code-Review+2\n\nrm -rf *","accounts_in_message":[],"_revision_number":3},{"id":"1a4fdca8eadc308b617cda973ba1dab79f98ef4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-12 16:34:03.000000000","message":"Patch Set 3: Code-Review-1\n\nActually, need to update SKIPPED_IMAGES.","accounts_in_message":[],"_revision_number":3},{"id":"e35dc30991a70b928c163d579772feec2edc4718","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 07:26:15.000000000","message":"Patch Set 3: Workflow-1\n\nyou right, me fix","accounts_in_message":[],"_revision_number":3},{"id":"a7bd06cd027560571882fef9978a8acace3233ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 08:17:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"32df74328c66902fd3ddfa8a4698beed4e8f1469","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 08:27:39.000000000","message":"Patch Set 4:\n\nah, clean air :-)","accounts_in_message":[],"_revision_number":4},{"id":"9186a79d4c112f5bf82bef7430c948cf7bf1b980","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 08:37:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"11cda0f7b5353ddbc4506502c8397f4c4acf50dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 08:57:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/30/693830/4/openstack-check/tripleo-build-containers-rhel-8-master/6121580/ : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":4},{"id":"08317a4ad178bfb1c55d68f189d914b247bbcde5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 09:48:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/361f3077d279495ea9e7af39fc91cb9a : SUCCESS in 38m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6f820a19575b4378b3d532a6be17d1ac : SUCCESS in 48m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4307dca805c74dd7965ffceb200e9a9f : SUCCESS in 1h 24m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/320bd8b13f73483f8f60120862d02fc9 : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/31b58df80a6e453ba0315c620f83e120 : SUCCESS in 1h 23m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e49fb68a62484e3788e6212570a40ae2 : SUCCESS in 42m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0ac273e543f42089acc3c4956fc59a7 : SUCCESS in 23m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8cdeacc76fe6488d9811e0a43bf7e2f0 : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6d459b0139114d759dab975cbbd90645 : FAILURE in 20m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/038fdb41dc0f453f8fe78adc4fc67c06 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3cc3f4a42e5422081c8cf1e98b3cee7 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe0aca8ce4a64940b3af1e03b9be5d37 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3328bb0c91634eed867d03408f6ed137 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b950d5d63e58445381b1e398f73edeed : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0f2577c5184c309194d999e97f5782 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29f2c9835f974d9fa3e88acbeba1f0dc : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1ab1ab2df214d269652bc5f1540b3bc : SUCCESS in 30m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14ab3ef2fb1641bbbbeee4bf13959986 : SUCCESS in 29m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/793987e40531497da3b700d5b847c5b5 : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9fb47a39d90d4cb5ba624e95ab8bf7d3 : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09315e5142f844ba9acb2c8e717de0f9 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":4},{"id":"160d797c1cdaa6a0409b320ffe18fe587dd7f194","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-13 19:59:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"886d849c63ea48f392a165ab31ae47072fa8cf3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 19:59:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a14ffe39c29afdcbacf276f04bbed66b6b981934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 21:37:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0c207778f3f644098d866265c571b0d6 : SUCCESS in 53m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/22b709d84ac34446a39708736611210b : SUCCESS in 1h 32m 21s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2d12189f4438496facefb7ae9db1fe7d : SUCCESS in 1h 28m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/853c3fe2442a4300ac9dddc1d77bb2c5 : SUCCESS in 47m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5e6a5915a874eb7a6ab6f9b49311122 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2adf35118542bb918ba372c98c3e35 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/055a043cea6e4c36831c05e8e5e3f655 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fb3da5ae8dd429e915de3dd69396dce : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39478446babc418896114015824fcb6a : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac0317a40c340da9bdca9802817f4e5 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e491c75c2f81484eb52497314e84c811 : SUCCESS in 4m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/50e15a2a87ac4e65bd08c8b894146dd4 : SUCCESS in 35m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/68efac7512c342a78c291864fb8261ee : SUCCESS in 59m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd678d0cf2904afba443eda4dc9fecf2 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":4},{"id":"423ed08d98c17bd929753b34fa0c6e717cc92dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 21:37:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"efeaa84b33421d54f92ed785bbb43dbe93e65020","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 21:39:25.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57d645d11a342ea96fe146951e48519 : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b09345c5fbf04a099d4286506c5c2ecb : FAILURE in 1m 30s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a20fa84e29a1f3e7afb15d008c05f37aa58ee3c0","revisions":{"e7eb9ba400f3590b1bd9c731d23d7c197245e866":{"kind":"REWORK","_number":1,"created":"2019-11-12 12:15:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/693830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/693830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/693830/1"}}},"commit":{"parents":[{"commit":"dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:14:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:14:22.000000000","tz":60},"subject":"[Ussuri] Remove deprecated images","message":"[Ussuri] Remove deprecated images\n\nChange-Id: I2fe57f7cd79724fa8a36bd702d75b013a68ecd1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7eb9ba400f3590b1bd9c731d23d7c197245e866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7eb9ba400f3590b1bd9c731d23d7c197245e866"}]},"branch":"refs/heads/master"},"18116a3fe98de1f598a6b393120e619de3d33d95":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-12 12:19:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/693830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/693830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/693830/2"}}},"commit":{"parents":[{"commit":"dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:14:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:19:30.000000000","tz":0},"subject":"Remove deprecated images","message":"Remove deprecated images\n\nChange-Id: I2fe57f7cd79724fa8a36bd702d75b013a68ecd1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18116a3fe98de1f598a6b393120e619de3d33d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18116a3fe98de1f598a6b393120e619de3d33d95"}]},"branch":"refs/heads/master"},"622f8b35f807ac8cfe1087a162071f86e7add024":{"kind":"REWORK","_number":3,"created":"2019-11-12 12:22:17.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/693830/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/693830/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/693830/3"}}},"commit":{"parents":[{"commit":"dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:14:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:22:04.000000000","tz":60},"subject":"Remove deprecated images","message":"Remove deprecated images\n\nChange-Id: I2fe57f7cd79724fa8a36bd702d75b013a68ecd1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/622f8b35f807ac8cfe1087a162071f86e7add024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/622f8b35f807ac8cfe1087a162071f86e7add024"}]},"branch":"refs/heads/master"},"a20fa84e29a1f3e7afb15d008c05f37aa58ee3c0":{"kind":"REWORK","_number":4,"created":"2019-11-13 08:17:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/693830/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/693830/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/693830/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/693830/4"}}},"commit":{"parents":[{"commit":"dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-12 12:14:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 08:17:34.000000000","tz":60},"subject":"Remove deprecated images","message":"Remove deprecated images\n\nChange-Id: I2fe57f7cd79724fa8a36bd702d75b013a68ecd1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a20fa84e29a1f3e7afb15d008c05f37aa58ee3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a20fa84e29a1f3e7afb15d008c05f37aa58ee3c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
