)]}'
{"id":"openstack%2Fkolla~694054","triplet_id":"openstack%2Fkolla~master~I61c68bf738673a96dd5ba256e40e8532364ff281","project":"openstack/kolla","branch":"master","topic":"feature/rdo_train","hashtags":[],"change_id":"I61c68bf738673a96dd5ba256e40e8532364ff281","subject":"Revert \"Move to RDO Train packages\"","status":"MERGED","created":"2019-11-13 12:08:49.000000000","updated":"2019-11-14 21:44:10.000000000","submitted":"2019-11-14 21:42:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694054-1573767767107-2b47506d","meta_rev_id":"cd991489bcfa716502c35d6f50521b2c81132da6","_number":694054,"virtual_id_number":694054,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 21:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-11-13 19:58:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-14 09:44:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-14 02:27:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-11-14 03:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-11-14 09:44:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2019-11-13 12:08:49.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"date":"2019-11-14 17:33:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-13 14:54:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-14 19:01:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 12:08:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-11-13 19:58:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-14 02:27:03.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-14 03:02:34.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-14 09:44:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-14 17:33:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-14 19:01:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-14 21:42:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0991a9a5f39bf8f356a27f8ea27b76365077e8c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50fd7bcdc9d902c486605f59e391d8718d02c054","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 12:39:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/694054/1/openstack-check/tripleo-build-containers-rhel-8-master/cfe7c9b/ : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e0a036ec60469632054d9b792840c613782f3d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 14:42:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/41aac4b05caa47bfacc25958d4946f79 : SUCCESS in 54m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/929b7268bda94b26ad77c2525834e690 : SUCCESS in 1h 15m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a03ad87615ec41fa8729b7462a454653 : SUCCESS in 1h 22m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/905c4b5a01324e148d40af994b94c841 : SUCCESS in 1h 17m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5df2160f5b334b48a184745c97d374f1 : SUCCESS in 1h 34m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9dcd52fa2870425ea08b0100b34d0888 : SUCCESS in 58m 21s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d5a189230f23451288e4095573eacdd8 : SUCCESS in 58m 51s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0b2a95cfa11f4f91a5ffe4e011868d16 : SUCCESS in 2h 01m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/634183ab5f134b8190c4ba94344ca929 : FAILURE in 2h 17m 21s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/17aeafa17ffd4f59a20b68d9f02eff7b : TIMED_OUT in 2h 03m 37s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4f17678f5b8b4c06bc595ef5e2a1f13a : TIMED_OUT in 2h 04m 14s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f22a68182e4f40b08e58ba9a8a150af9 : SUCCESS in 1h 33m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6629b7491add42fb8c650f18d54dfb61 : SUCCESS in 25m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/42a4fc2b0bf04e379491e2bf364cf892 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/89255a3da51c435e881aca52b0455037 : SUCCESS in 52m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6add6c9d937749e1873c6f475f54b370 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda4ac6f9a5246989889710c434274e6 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/341a97a3e6d34aa9a9239fc97d489536 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3729fdde0c594723a073d65987f08b67 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/007ef6b2f5514ea194e48735ae108b0f : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e3e0771863492ea34a493d21cbaed7 : SUCCESS in 7m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72357b0c567d4a78bf37ff414163c9ba : SUCCESS in 41m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c97bff405594861bf15fdeea3262bce : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6dcb95075ee2416bae04d2fd04ebc225 : SUCCESS in 54m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/339fc8b2835c4e93a7ea2e128918b415 : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b345d0cba7a04acf845ba63ce36eed19 : SUCCESS in 1h 05m 52s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2525c297400b452d9254d3623fbee6e8 : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/1062bb624a164301acad143a21d869a8 : SUCCESS in 1h 37m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/d0699a6b34f541818e4e6f97313f479e : FAILURE in 2h 02m 48s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e50eb23fa5e44047bdc849227d8ee08e : SUCCESS in 1h 19m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/dc19c9118dd64a4b9b0c163ff74b3fb3 : SUCCESS in 1h 07m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dd1dccfcfe2cb39756f989cfade10016a58487b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 14:54:54.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"17ebabb52f83fc32af8b36634102431a983d965a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-13 19:58:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7010011c90838a07d345f931c6798bf941c9792b","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-14 02:27:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c6a429a0d5d29dc8ffaf7851c5b0c948237621e4","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-14 03:02:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d86656df93bba04673f31267d3b027e33aafbe4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 09:44:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa7ae11e1ecacb966de50de02c19ecc742641caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 09:44:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aae9d9cfd7028d01a691c02dd7e5cd5a335917e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:21:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c37066bef29d442fa3bd05aff22916e5 : SUCCESS in 1h 06m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7b54b1faa9b846aaac60487803a0a3d4 : SUCCESS in 1h 18m 43s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ed4bdee231f4d2f8508b32acb6e2178 : SUCCESS in 1h 32m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/feb241c5d0cb4708b0296e58591d2f29 : SUCCESS in 1h 02m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f24f164616c4b08a986aee5ae06af7a : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0abb4104578347069bad2f778108b25d : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96c69a9c18de4f6aafd52feb7d029a2d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a36435e7479246818401783d07578a6e : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4802d51919384cb5a61bf41ae9da8258 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0f85db7bf74f19bac2c0e5afd74995 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6b0298f3bfc4e16b58dbdaba8744721 : SUCCESS in 30m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a230c8a1b624af9b9c4cd7602a9f328 : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2a2efc6b802945139495231a1d28ebdf : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"dec8f98df77baab2e1faec921c9c36ef0fdf47a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 11:23:17.000000000","message":"Patch Set 1:\n\nrecheck\n\ndocker 500 when connecting to run container","accounts_in_message":[],"_revision_number":1},{"id":"1bf82946bfd549fb945b8460ab72d36610031909","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 11:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/694054/1/openstack-check/tripleo-build-containers-rhel-8-master/b570a41/ : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c697fd08c195727dc3a083cfe791765c06024041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:00:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6dfb4e31ff5e4716bfdd7f3bea289d51 : SUCCESS in 40m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/73ce37f812124adf89703bf340b2673d : SUCCESS in 48m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/32951133b21b45d7b217ce2ba9de43c8 : SUCCESS in 1h 18m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ee9ff0373ebb43abb5745f03963044f7 : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fc15ec0434b64d76ba092aee79da7a79 : SUCCESS in 1h 21m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/be25db82f7184a7d85617036bbda3f94 : SUCCESS in 1h 03m 37s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/eebc7c4414a141e8820b45f78a998f26 : SUCCESS in 54m 36s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/efb85a4bb3a046bb89f14a0f3c4ea6b1 : SUCCESS in 2h 27m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8451e7ff1fed414597ccf5e3a9b92d09 : FAILURE in 2h 27m 23s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c95241d3233f4cf1b6403f42ec2e903d : SUCCESS in 1h 52m 13s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1c6369520b4149ff82de84a9d2de298f : SUCCESS in 1h 49m 10s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f3f33c8dceda4630b39e3bad3ac7e9e7 : SUCCESS in 1h 32m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/65df124018e64f2aa62bc18183e1d8c6 : SUCCESS in 31m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d5c88494ce3e4609bcda2c2bd99b2722 : SUCCESS in 28m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f0f4de44f1d45e2902269f2913431df : FAILURE in 20m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d16c45021572424985d288c5713c4eea : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97286230e8b04828b480ed6165ebcf41 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0fff4a4b1144783833215c868f53a44 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a8cb79bed164dd39bdf3506aeb61127 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67a789129a0042bb94b2dc5b81eb3a3d : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737b437179bf4d3cb79745ca4b220d61 : SUCCESS in 7m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64ad71a517c64d8eb68cc6fc96ebda0c : SUCCESS in 27m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60d4f7c706c047aa88374e229697dfdc : SUCCESS in 27m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cc11d6589d2426a9ec2aa83dbb2801c : SUCCESS in 53m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9fad832294b4912b03b092aad13aab0 : SUCCESS in 1h 06m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41f715fc29804c88bc6189995a1d108d : SUCCESS in 1h 06m 08s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e361270b067143cebf36f97ed88d4cfa : SUCCESS in 1h 20m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/31bb3b29fedc458da9f6c58730d21cd7 : SUCCESS in 1h 23m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/46cd8754d59b4e88aab124732636e7c9 : FAILURE in 1h 44m 08s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7b79fa5f0b7b42d8a8f7b37230ed8c8e : SUCCESS in 1h 24m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/01c658c8b71642e2ae96c69d9ba37b0c : SUCCESS in 1h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b342ef9bdb7c35926c31922fe6a96b8d0e87d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c88597187e8634b7ade2305451eb2578c9a02656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 15:46:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/79673b8d48844aabbd76c8dad1cc35c6 : SUCCESS in 57m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fe486287fd0b42e59d1d7cb6a97c2ab1 : SUCCESS in 1h 29m 57s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b21f7a97d9f34aeb84eb4d0a3fb68fcd : SUCCESS in 1h 36m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d7880b351ca44604b712f5c96e58f582 : SUCCESS in 54m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1892f276336e4432857b805cd6c2b525 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ca7b5d25af4f1db0d0e7a672b5239d : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d54099e529e4bceb95220ed02041654 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f9346353c8453aa7313e74f8b95270 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04bf46dbd7df4e258aea3242a1f90c79 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ceda82c01e4fef968e360c9ac6fefe : SUCCESS in 5m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/592eb603faf24da59fbab5f95e08758a : FAILURE in 10m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8cba61a260a4a63b8ea3073abb042c4 : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e4f35420c6342e9aa615369500631d4 : FAILURE in 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"63839b1c4e5338ddb8b1369c31f002362abc3c32","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 17:33:01.000000000","message":"Patch Set 1:\n\nrecheck mirror syncing","accounts_in_message":[],"_revision_number":1},{"id":"fe5b635b323fbf077c992af2582a25644b5be4c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 19:01:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/694054/1/openstack-check/tripleo-build-containers-rhel-8-master/dd08cbc/ : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b8365f405f75c00d86848904ea80e1d17c6eac89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 20:10:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b46fd41940bc40e3a590b5ae2429d3c4 : SUCCESS in 47m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0616c89db9a24cbdad990ca3d2e5bc03 : SUCCESS in 52m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bf70a3df70ca488b80411fbbbc6cd278 : SUCCESS in 1h 30m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b1e9132db51f40ccba30d7053e1b152a : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4003518d37bc4b1b87721eb889189adc : SUCCESS in 1h 37m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c85d4fe79af4a4c889dcd4e531b07bb : SUCCESS in 57m 21s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5d16cba8d8ad42e08db4e36457890692 : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/09c76741c6234962a0375fa4c216ed86 : SUCCESS in 1h 55m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/791efee0f4a6438fa5782bc638514b1f : SUCCESS in 2h 23m 31s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c989a97c655c477cbae7df656ba795d2 : SUCCESS in 1h 39m 14s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fc5d226d7abd4e12901b8b89242f7f73 : SUCCESS in 1h 48m 40s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/dcfc00a5bc8c473690fc33c7d0a45906 : SUCCESS in 1h 24m 53s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23cf94cf63444fdeb49ae4fbc8566ef1 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d9a0130c45a14012958e38405b6a523b : SUCCESS in 19m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/00dea27cc90f48dd98439e27981e63fb : FAILURE in 26m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f6fd9c6528f4f278b8636cbdc4d16a2 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc4591e679946f78075bad9627d129d : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad2ca3481ffe49b996c441a8c625c1a5 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7896244e1f594c86b2197bccb58cd1a6 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/908cb9f7da9c451ab3465318bc2d9aa8 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f58087e1434c7e9f8ce77192cafac5 : SUCCESS in 8m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b6c67150c9d4b149c1c07a778f8c9ba : SUCCESS in 34m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a5dc82290db491693c08bf5bea6848d : SUCCESS in 39m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/640d293cf2b84edbbfb528136204cf66 : SUCCESS in 1h 10m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9704835ffc1b431d89319f800bc89662 : SUCCESS in 1h 02m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c550349c8404d779a7cb5ea7f4f7fbd : SUCCESS in 59m 27s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/3f39856eb0174690928f6f39c95f8aad : SUCCESS in 1h 15m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/a5597024c1ab41e88159ea9971eff7d6 : SUCCESS in 1h 28m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/170adfa98b184c218cfca0da57690c23 : SUCCESS in 1h 58m 15s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/77bbc93c0e914d2694885239cdfeee07 : SUCCESS in 1h 20m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/414c82dba5b84435a078fdee16bd786b : SUCCESS in 1h 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d42cdc1b39ca6686e61c077f6e03434c3489c6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 20:10:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51bc7c99c5f9135992f22c6f054050fd1dfcff9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:42:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9587ec1aa9ad456cbc77b8e1d01fc296 : SUCCESS in 53m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/72874859ff064acda9ee3fbf63f9e872 : SUCCESS in 1h 31m 11s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a2686cb53a0640699bd055034c703dd1 : SUCCESS in 1h 18m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/46a935d84be243589f3bebb56e344618 : SUCCESS in 51m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/382f54a020e7413bbc10a2fc96afa6e5 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf8b130fbb74630b153309933d49ae0 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88fc0a3a352048e9ac82093ade758514 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5dd8e3ceaba4589bde4d8ddfd464470 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6692a96c8b064273a08cfa50dc3ef557 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cbc561b92540aeaaa2b54246add8b5 : SUCCESS in 7m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8073f09403a42578cfc7e6cddbe00d4 : SUCCESS in 29m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ae4bd978e0f4f24a5d60b592ef83e42 : SUCCESS in 1h 04m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fdf4f3f09f1d4266aa01fa422cf054af : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f805639b8b27a4b82ab5a9e80f736155d15da7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:42:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8298fc7ab6ab1ef5fde4fa40291ca98c751c9618","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:44:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b7dc42852841ecbc90be265f32ff1c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8abbf34c0373f24d6417f19a2ba27f3d017e10f","revisions":{"a8abbf34c0373f24d6417f19a2ba27f3d017e10f":{"kind":"REWORK","_number":1,"created":"2019-11-13 12:08:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/694054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/694054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/694054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/694054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/694054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/694054/1"}}},"commit":{"parents":[{"commit":"c78c82c9630b334cbaef040d68bce7e5ae8cc118","subject":"Move to RDO Train packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c78c82c9630b334cbaef040d68bce7e5ae8cc118"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:08:49.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:08:49.000000000","tz":0},"subject":"Revert \"Move to RDO Train packages\"","message":"Revert \"Move to RDO Train packages\"\n\nThis reverts commit c78c82c9630b334cbaef040d68bce7e5ae8cc118.\n\nChange-Id: I61c68bf738673a96dd5ba256e40e8532364ff281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8abbf34c0373f24d6417f19a2ba27f3d017e10f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8abbf34c0373f24d6417f19a2ba27f3d017e10f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
