)]}'
{"id":"openstack%2Fkolla~694056","triplet_id":"openstack%2Fkolla~master~I95da4731650cf6a19f369493f530aa5ba227167a","project":"openstack/kolla","branch":"master","topic":"kolla-train-release","hashtags":[],"change_id":"I95da4731650cf6a19f369493f530aa5ba227167a","subject":"Revert \"Use Train release for all OpenStack projects\"","status":"MERGED","created":"2019-11-13 12:13:24.000000000","updated":"2019-11-21 16:09:39.000000000","submitted":"2019-11-21 16:08:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694056-1574352492280-8eeca06d","meta_rev_id":"8607d328b1c97dc1b59e21ebb81546f95c751aa3","_number":694056,"virtual_id_number":694056,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 16:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-21 13:15:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-14 09:47:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-15 07:30:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-15 07:30:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-14 17:34:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 12:13:24.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-11-13 12:26:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-13 13:02:56.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-11-13 13:18:04.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-14 17:34:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-21 13:15:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-21 16:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d870141609c047aa71e1d07c44e749b3a7ad3ec2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f4f4dd2e6e1018e8ed78f00218ca749f1b3011f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 12:13:37.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"66ab68f3718146f3c2226e974bc975f38476c666","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 12:13:38.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7fd6d19727685d10fe8ad80a4a996ba2d34ebc1e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:14:07.000000000","message":"Abandoned\n\nweird","accounts_in_message":[],"_revision_number":1},{"id":"c55f6ccf7ba3cfbc97943451731a223597e31258","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:17:11.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"d04dbbf51b3fa07d383039d4f5ad67a8c7a4f649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 12:17:24.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"77255ab5c55bd7964481f206e404c36149405f71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 12:17:25.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"2d17a599defd2a924f9679f984488ac9a7115d1e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:18:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b4e551bd41cf163d377e11ad3dab489fd8d120d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 12:18:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fe415c8931f6977c6ff292e26aebb9b347b69677","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 12:26:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1cfc4a3692c543b470771a79aff7ac47f3e30bc4","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-11-13 13:02:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6cdf90431f4861f0afc56e2e36bc1478495d97de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 13:12:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/694056/3/openstack-check/tripleo-build-containers-rhel-8-master/f898c79/ : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e840e1cb3a77c17ab5bf1a3e0ce57941011adfb0","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 13:18:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f66a6e5c5a12d78c0d030649f162626880362ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 13:42:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c0affd2df1d7429cbad21ed3597d9408 : SUCCESS in 39m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/96d9a0f6fb1e429a977029d25620d18a : SUCCESS in 49m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a397d5cb7ea64a9a859067adab3ea34b : SUCCESS in 1h 20m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/999279f07caf45a89534f94337d79c77 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/47347b0b09794aa7b5d441f328ceae6f : SUCCESS in 1h 21m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b480c534c4a14c3cb7c4f9b7c3817d3d : SUCCESS in 55m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee72b5272d9b4a45a67d83f77bbb18a5 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3efb2dea975f4f24a50e40f537f4c965 : SUCCESS in 26m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4e6c8bc339e6493c986c58e5299fad36 : FAILURE in 21m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd0fdc83a4844052bd8918fbbbb42f12 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3429fc19de4c0ca8316684be3e66e0 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/281fdff1c3df4b3ca1a8e409e3724c0f : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d392dad477a4661b0863bb9bc9f3564 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d6ca9e76c46430681ebe53626ce4627 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b204174474947868b79fb69548ee7f4 : SUCCESS in 8m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a4fec06427243ae8b2c04abb6e1da5d : SUCCESS in 41m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/afcddac52ce6447ab42cfe6cecb75acc : FAILURE in 46m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16402894033a4c6fa431ffc970efe7b7 : SUCCESS in 1h 04m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47cdb18ce17143ca8d4acaa893954fb6 : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c2dcb65e806647a5bd2c1400527c5b54 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0f5fb587732cc94144ac3a17f4632bfa1f4a28cf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 14:56:37.000000000","message":"Patch Set 3:\n\nubuntu source on congress: UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xef in position 1645: ordinal not in range(128)\n\nWHY?!","accounts_in_message":[],"_revision_number":3},{"id":"c3333b506f84d000948710c02c39b9cdd9f9b389","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 07:48:06.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a0240eb48a16db57f3ef45abbf1be53dac914ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 07:48:20.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"9610fba2df67236e2f271d11bc03cfa03f8bcb06","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 07:48:21.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"99768b8a867cd2c84eb925cc649bc811fd887e1a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 09:43:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"63009d3666677d61270efc566e79b21609995ab4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 09:47:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b6da305dbe7946839ca126a2f2152bbf1e68fafa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 10:32:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/694056/5/openstack-check/tripleo-build-containers-rhel-8-master/7bce600/ : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":5},{"id":"eb05b94253b86c70eefdf65942d823ddcd33b9c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:28:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9835dea53c6240dbaae0f766114dd2d4 : SUCCESS in 40m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4d1599f294a145e2bd20a88fdc4b5bd9 : SUCCESS in 1h 01m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9e6d4459440140dea335de9c154b531f : SUCCESS in 1h 41m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2b4064dccd8b47fead1bb8ff971c90ab : SUCCESS in 1h 40m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/35c5baf67faa43f68d3564d407e984d4 : SUCCESS in 1h 33m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2b5d7b8534c243a5b1f73cbfa9bedf1a : SUCCESS in 1h 03m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0685e77d2c24991bf281836717557b0 : SUCCESS in 29m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/19ddf56da188449ca71298ee3fd276f4 : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3e9d673ea4dc447fb35942cfa30d8bc6 : FAILURE in 18m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8599e5bcfa04b6fb80a847896647688 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3700460252454f4e8e2e9c6255218e1e : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56ccf46825374a1c8e7d5138174d7f52 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb4b98d53074d4db40d0129a20bc968 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1391683d914438187d6b445c7289584 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae7cd6831ba477f9d11d9060cc29031 : SUCCESS in 7m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9eabb9cfe90043b8b32845eb9a3bbfee : SUCCESS in 40m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/585a9326f166420785be092c7115294d : SUCCESS in 44m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4172932e0b3a4da0bf0d053aeeb38c76 : SUCCESS in 1h 01m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32c690610c6d4385a9fae13315637096 : SUCCESS in 59m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d92bcad08d3c46e790612bb6a200f81c : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":5},{"id":"e059067c32909b36952663a33737d4fa4e7e3fe8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 17:34:27.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"35c8e285294f70e16b2449f09fce8854337bf675","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-15 07:30:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nI guess I can self-accept as this is straightforward","accounts_in_message":[],"_revision_number":5},{"id":"8e14f27b467386349e18bffa613077cb0ecfc99d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 17:55:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d34a51169a7f2e3aad7946d8b41b4c370532b908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:36:52.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"35b141f2b5f58f31d9932daabac628dd1caf622b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 12:41:13.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ed7ec90c595f046147e329a1900aa5f6a56ad125","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 13:15:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/694056/5/openstack-check/tripleo-build-containers-rhel-8-master/743da14/ : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":5},{"id":"351035d162fd6a97a38d901c2afdf6c4d04fd668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 14:31:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ec073c3dc47d41bbb4a70f9d3b94594b : SUCCESS in 53m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8cc14fdda6924a36906133696587cf46 : SUCCESS in 45m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dfa04a901d26407ca839a40b7c5e5e08 : SUCCESS in 1h 35m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1c66fc547403453f8ad96fbb798f898b : SUCCESS in 1h 27m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9a1556b39c594f749daedba037cfd8d9 : SUCCESS in 1h 39m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/399cb14997274f19b8404d6194196fad : SUCCESS in 50m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f597b4dfddfe4ebe8423a4736638d430 : SUCCESS in 24m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a70336e7ab1c41fe968b0534282012d9 : SUCCESS in 21m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6533817c1c704a308a81c5d2e40c8840 : SUCCESS in 53m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1de7c56a3d1d494c87ca064e03e5334f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72812dac52404722972ad486732f95e2 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f38abceebf0e4c959882c726b6f0b9fe : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac5b197785bb46df9c5758ea9ee5d2fa : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e5b8507e6f48fa848c78ca96473dc9 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8322f95ad2964c7295ba1a323c01411c : SUCCESS in 28m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc4f72c896a0431988205a83d0dd3a2a : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e411082a49c940008a59d71000a701b1 : SUCCESS in 1h 06m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/468e2c52bb344c1693e02c9c7facbf30 : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75541e6fc7894c4db21b67ecc3ed10d8 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":5},{"id":"5c12ef00945a7ac90d81891d23c799dda3c5df95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 14:31:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6b1b51349657ff878f4ee0f7cf37fc47e3d4a4d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:08:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e9037cc33007a229cce8d335bd9a35e1e51a0286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:08:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c7e804cdf6c945428d666ba58feb5822 : SUCCESS in 57m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1e996619da9f4460a2c8a08323bc0ac3 : SUCCESS in 1h 19m 12s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9243ce5e74504cd6b42b21d580d463b7 : SUCCESS in 1h 33m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/af4ae876fbf2477ba46afeca70566914 : SUCCESS in 1h 01m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58c42bcbca6b4c6188fc418b3a20d3a4 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43839778e0a349889efce0376cd97bdf : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3c2f9d014954aafaa5394bdd3742f9b : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f62b9678aa346e6b91986746ca38a27 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf8583057f0417b89c77ef8372c285c : SUCCESS in 5m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/390ed28875b84df8846ca8c6a2462358 : SUCCESS in 38m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/064f2a81304a403a96d4f4ac1fcafa13 : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c45ea1378ffc44cb8782223107a2b207 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":5},{"id":"0d48ec3441c2b2ccc75962e89fb184babbcdc96c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:09:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479b296e26244d80ac2f72190170d4ff : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"269e5f5ee149d1bc9c61bf4c28d2eb1aa6f792b5","revisions":{"b00cf81f1026b8110ee9f973b850c7674058e367":{"kind":"REWORK","_number":1,"created":"2019-11-13 12:13:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/694056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/694056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/694056/1"}}},"commit":{"parents":[{"commit":"97979578506497ce0b466d7621f83cd9a9338ed0","subject":"Use Train release for all OpenStack projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97979578506497ce0b466d7621f83cd9a9338ed0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:13:24.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:13:24.000000000","tz":0},"subject":"Revert \"Use Train release for all OpenStack projects\"","message":"Revert \"Use Train release for all OpenStack projects\"\n\nThis reverts commit 97979578506497ce0b466d7621f83cd9a9338ed0.\n\nChange-Id: I95da4731650cf6a19f369493f530aa5ba227167a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b00cf81f1026b8110ee9f973b850c7674058e367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b00cf81f1026b8110ee9f973b850c7674058e367"}]},"branch":"refs/heads/master"},"a4a4d0f5f5a378e7c5ad1ac275833800eb87c60b":{"kind":"REWORK","_number":2,"created":"2019-11-13 12:18:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/694056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/694056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/694056/2"}}},"commit":{"parents":[{"commit":"dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcbc0bd89a24cd93aa8fc46a644c9d4c19dad4d5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:13:24.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:18:17.000000000","tz":60},"subject":"Revert \"Use Train release for all OpenStack projects\"","message":"Revert \"Use Train release for all OpenStack projects\"\n\nThis reverts commit 97979578506497ce0b466d7621f83cd9a9338ed0.\n\nChange-Id: I95da4731650cf6a19f369493f530aa5ba227167a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4a4d0f5f5a378e7c5ad1ac275833800eb87c60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4a4d0f5f5a378e7c5ad1ac275833800eb87c60b"}]},"branch":"refs/heads/master"},"fb9aa5d44b3fabefd7e5e0d756d954d37acfb97e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-13 12:18:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/694056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/694056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/694056/3"}}},"commit":{"parents":[{"commit":"6b40b655f124d0a95f1b7d89ce04236606a02010","subject":"Disable Searchlight for CentOS 7/source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b40b655f124d0a95f1b7d89ce04236606a02010"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:13:24.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:18:50.000000000","tz":0},"subject":"Revert \"Use Train release for all OpenStack projects\"","message":"Revert \"Use Train release for all OpenStack projects\"\n\nThis reverts commit 97979578506497ce0b466d7621f83cd9a9338ed0.\n\nChange-Id: I95da4731650cf6a19f369493f530aa5ba227167a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb9aa5d44b3fabefd7e5e0d756d954d37acfb97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb9aa5d44b3fabefd7e5e0d756d954d37acfb97e"}]},"branch":"refs/heads/master"},"df2d58d0b6e3ec6605c63cd3ed3a36bf6c2f27bd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-11-14 07:48:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/694056/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/694056/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/694056/4"}}},"commit":{"parents":[{"commit":"6b40b655f124d0a95f1b7d89ce04236606a02010","subject":"Disable Searchlight for CentOS 7/source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b40b655f124d0a95f1b7d89ce04236606a02010"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:13:24.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 07:48:01.000000000","tz":0},"subject":"Revert \"Use Train release for all OpenStack projects\"","message":"Revert \"Use Train release for all OpenStack projects\"\n\nThis reverts commit 97979578506497ce0b466d7621f83cd9a9338ed0.\n\nDepends-On: https://review.opendev.org/694099\nChange-Id: I95da4731650cf6a19f369493f530aa5ba227167a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df2d58d0b6e3ec6605c63cd3ed3a36bf6c2f27bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df2d58d0b6e3ec6605c63cd3ed3a36bf6c2f27bd"}]},"branch":"refs/heads/master"},"269e5f5ee149d1bc9c61bf4c28d2eb1aa6f792b5":{"kind":"REWORK","_number":5,"created":"2019-11-14 09:43:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/694056/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/694056/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/694056/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/694056/5"}}},"commit":{"parents":[{"commit":"414987b15f7d79cf4c51f305589b5b6d2d8fe1b4","subject":"Merge \"Remove deprecated images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/414987b15f7d79cf4c51f305589b5b6d2d8fe1b4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 12:13:24.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 09:43:22.000000000","tz":60},"subject":"Revert \"Use Train release for all OpenStack projects\"","message":"Revert \"Use Train release for all OpenStack projects\"\n\nThis reverts commit 97979578506497ce0b466d7621f83cd9a9338ed0.\n\nDepends-On: https://review.opendev.org/694099\nChange-Id: I95da4731650cf6a19f369493f530aa5ba227167a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/269e5f5ee149d1bc9c61bf4c28d2eb1aa6f792b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/269e5f5ee149d1bc9c61bf4c28d2eb1aa6f792b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
