)]}'
{"id":"openstack%2Fkolla~694099","triplet_id":"openstack%2Fkolla~master~Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d","project":"openstack/kolla","branch":"master","topic":"bug/1852452","hashtags":[],"change_id":"Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d","subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","status":"MERGED","created":"2019-11-13 15:39:30.000000000","updated":"2019-11-21 12:39:14.000000000","submitted":"2019-11-21 12:29:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694099-1574339372829-9f2e12eb","meta_rev_id":"fa93ef675025ced76797ef8b8eb8ea485596cae4","_number":694099,"virtual_id_number":694099,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 12:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-21 09:11:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-11-14 08:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-14 08:27:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-20 17:55:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-11-14 08:45:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 08:27:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-14 08:47:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-20 17:55:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-21 09:11:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-21 12:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3acdfeab91833e3c0c7fb785d8ad60ab4d68bc5c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 15:39:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30f793c65b7197607d37adf135bf46aefac54bf7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 15:40:59.000000000","message":"Patch Set 1:\n\n(1 comment)\n\ncloses: https://bugs.launchpad.net/kolla/+bug/1852452","accounts_in_message":[],"_revision_number":1},{"id":"977a8c964ef5d0743d92d4684ea3ef12ef56ee0e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 16:02:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"482c163aa597d48c08b5a8e3476dd85acdbd9901","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 16:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/99/694099/1/openstack-check/tripleo-build-containers-rhel-8-master/8267608/ : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"34d39217482b4b3f6cfeb1f66fd889d6c7237639","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 16:13:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f2934ea60a9612fbc7cc3a6dc33642f5da90765","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 16:14:15.000000000","message":"Topic changed from fix-locale to bug/1852452","accounts_in_message":[],"_revision_number":2},{"id":"242e1664c500e2ee6024f465aef1a22486204564","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 16:14:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"016400b473f431edd37cb103859ae817a69a0a60","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 16:14:44.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"593b6d8182cbb027edbdac3008fe2c6902c262b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 16:15:02.000000000","message":"Patch Set 3:\n\nadded closes-bug for you...","accounts_in_message":[],"_revision_number":3},{"id":"aa46335b61f02d6f232ae223604cc81d90c127b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 16:46:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/99/694099/3/openstack-check/tripleo-build-containers-rhel-8-master/88d81c0/ : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":3},{"id":"4069f2d9b296adc2089c36ad43ea8ac4cbd5d54c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 18:27:49.000000000","message":"Patch Set 3:\n\nit fixed Ubuntu but not Debian - to check (looks like empty run, locales moved to a separate package?)","accounts_in_message":[],"_revision_number":3},{"id":"a28c5482227edde4db39ab81195d53dba4cab97e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-13 18:27:57.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ef065a868cdae56f8c84d5ba579318cc840aee46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 18:55:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7b68162d7bdf46d79956e06fd2ae8ecc : SUCCESS in 47m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/973c5e5948154491bc10422927a2d776 : SUCCESS in 55m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/67d70766b3524cf799ea822f29e2670f : SUCCESS in 1h 28m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1618a84322fd4061abbb2649fcdca955 : SUCCESS in 1h 20m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3f3b0a25396c44078d0c0871d965b96e : SUCCESS in 1h 38m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/02cef17397ab4b5187be95d3430f2ad0 : SUCCESS in 50m 16s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/457901ffe450421e9999cb6785bada19 : SUCCESS in 47m 21s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c975e66a2ae148e89b38b71d6638dd5e : SUCCESS in 2h 14m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/91ebf1ba1a6342c5a80e1549a3137403 : SUCCESS in 2h 20m 35s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/093f0d0c610746e6a4e26671d3722a08 : SUCCESS in 1h 45m 25s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/43b4cf89665f4583bab399d1aefcd88d : TIMED_OUT in 2h 03m 03s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/986b8ed1d1574947880280fa7f97da22 : SUCCESS in 1h 32m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/36aed46412524a338c4374e5e44586e6 : SUCCESS in 34m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3cd74a883b2c4743a6bc0b9ef9421dad : SUCCESS in 31m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/12b32f87ec0a4df3bb40bacbc083bd3c : SUCCESS in 49m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1705b9744aa7410c91c1beb7d9d70562 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af937b654e3f4f69838caff924e6e74e : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e320a81ff0f04debba33f1e93f37ed3d : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fec7195e0184d98a6f5c25ac8281c94 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6393ab683614f02aee54a7d3589c827 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb724d6338174116b88c5e68a874dc66 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b62e2d488f347c795168cb04dafcee9 : SUCCESS in 32m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7936fafff69e4ce19328ec913660a293 : SUCCESS in 37m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0d9b129ec3f44c2b73dc334fed6200c : SUCCESS in 1h 07m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/499eab0f0e504b91ad7e121c70f5efef : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb1158fe5eeb49538c0cf201d2ac9ea9 : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7f35d124aff04ae19ff19d0234656cb5 : SUCCESS in 1h 15m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/55195b53c5f34130aee082902c7a93ae : SUCCESS in 1h 27m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/d8e70d7d1f9c4076b45432abd01aac1c : SUCCESS in 1h 51m 57s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/50ca27f160de49c2a9c9216f118ee8b4 : SUCCESS in 1h 23m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f28161edf22346e7a2c022d811195d5c : SUCCESS in 1h 09m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9ef6c14e4eba7a24e8e925d953a5d3f754336d96","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 07:36:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ac2618ce1a27c19495b03824c647a0b528d5d873","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 07:39:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7686e505f349cdb8190f6bd1db881036385ca86f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 08:14:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/99/694099/5/openstack-check/tripleo-build-containers-rhel-8-master/303a2cb/ : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":5},{"id":"3c139a75d1f6b6e26108e78823db0dfd80663410","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 08:27:31.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"78cc590e1248b636741a7cdae06277e4d1c15fff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 08:27:38.000000000","message":"Topic changed from fix-locale to bug/1852452","accounts_in_message":[],"_revision_number":6},{"id":"8e97924b7ba559ead08099a38276557f5b7370f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 08:27:52.000000000","message":"Patch Set 6: Code-Review+2\n\nlooks fixed","accounts_in_message":[],"_revision_number":6},{"id":"89663c13152fff9115e93244fd7c3c364d4da4c8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 08:45:31.000000000","message":"Patch Set 6: Review-Priority+1\n\nbumping priority as it fixes congress for ubuntu","accounts_in_message":[],"_revision_number":6},{"id":"22c88b73e752205c97e4bbd8ee505a498129aab9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-14 08:47:15.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7546dac9c4e475f773af4d6fd9a4d6e693d12ff2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 09:03:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/99/694099/6/openstack-check/tripleo-build-containers-rhel-8-master/cdc4aa8/ : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":6},{"id":"934f6bfe7a16763ec891960fa2c18c525e191dd8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 09:36:43.000000000","message":"Patch Set 6:\n\nWow, this locale stuff really got out of hand (I said it would!).\n\nWe\u0027ve lost the ability to configure all the locale bits in the base_lang block. Adding a new RUN command to that block would increase the image size (and is before repo setup).","accounts_in_message":[],"_revision_number":6},{"id":"f57f3a284f91a9b79417ccb054a94c63cba1faa6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 09:56:38.000000000","message":"Patch Set 6:\n\n\u003e Wow, this locale stuff really got out of hand (I said it would!).\n \nIn perfect world images would have language out of distro.\n\n \u003e We\u0027ve lost the ability to configure all the locale bits in the\n \u003e base_lang block. Adding a new RUN command to that block would\n \u003e increase the image size (and is before repo setup).\n\nDoing such thing in base_lang is insanity ;D","accounts_in_message":[],"_revision_number":6},{"id":"a9142c0202c2e61b295d27238f4e6cbcbb0e2d4b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 10:20:15.000000000","message":"Patch Set 6:\n\nMark, one can override LANG and still be good as CentOS provides them all and debian/ubuntu will build. We are treating them differently indeed, just as you wished.","accounts_in_message":[],"_revision_number":6},{"id":"a3dc2e6971ba9204c16a06629ac1e30823f9b5f7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 10:39:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2c41ae5a5852e56b1610d8ec414cd8b39e875096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:47:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b0d856bf691c43de8f5226fd3ac48c8e : SUCCESS in 52m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6a6b7196b10e406e9606170f7f948510 : SUCCESS in 48m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cdca107e1aee49efbd9b140b8afbcec5 : SUCCESS in 1h 24m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ec0aea549b1b415dad6cbab40b689033 : SUCCESS in 1h 06m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2e206af926ca440da8aea1472add345d : SUCCESS in 1h 28m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/916c0b38bff548359562d2c43149c45a : SUCCESS in 48m 26s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/ccbec622cc294cc1b151e16680c8765b : SUCCESS in 45m 44s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/66125f755596432ba6fcf9479d6a5aeb : SUCCESS in 1h 56m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f44085212a6b426fbbc78ef5707e946e : SUCCESS in 2h 10m 20s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/179ef91936ad4d30bdf012f4c20e0dc2 : TIMED_OUT in 2h 03m 00s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/353a1327a6db4481b1742c800f466c53 : SUCCESS in 1h 54m 28s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f78ca699683549aab014b7cde16d1d28 : SUCCESS in 1h 32m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b60500814c174ede821d968af1117936 : SUCCESS in 33m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/98bad79b76354c6fa2cfdd0aea884c23 : SUCCESS in 23m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3f547ac2809f4afc9bb44e31db8c2ecc : FAILURE in 32m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5644fa4495640179dfeb46ccdd8f66a : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292cf0bba5a14cfca4434fd1cd00068f : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a909ff26205f4833b6920457842bd568 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fde9ae8fb2b04edca9dd271c3d475959 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa7287eeda604eb691bbe5100881902b : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b349d8b2231a449791644402bba7174e : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd3b070c991045c6aedbec2159fa7883 : SUCCESS in 36m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/858112ed2b7e420f8367e1af3be1667e : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/56ce547c95f74d81b26691e1e87f9d79 : SUCCESS in 1h 08m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbc220bb6f6d49daa198db4388592f67 : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d017ef5c22e48369ee9463384dd3af5 : SUCCESS in 1h 13m 22s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5106b87fb2934af6b478c54e41ea3526 : SUCCESS in 1h 25m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/9cb5405a1cea47a6b972fb93f002f843 : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/778641af752f43d59e1eb183a1405642 : FAILURE in 1h 40m 41s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/02c17a04c81841d582d0dc41e4b1b695 : SUCCESS in 1h 12m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c46427f0c735492d8c96f76f38a0a28b : SUCCESS in 1h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cc11a43a4e7272c895961d77e14d79c4291302d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 10:53:38.000000000","message":"Patch Set 6: Code-Review-1\n\nFor discussion","accounts_in_message":[],"_revision_number":6},{"id":"495181837e144a600764aaa8ea396498d22ce8fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 11:01:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4793d55623c3edcb2c656861137de1e2cc520686","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 11:05:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b3fb5559dc4ab7b6608676bfd94802b1634d230a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 11:09:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"37ede968cd084ca776f71ae536645255d5b6c4fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 11:13:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bb92b4f0e8c4c445354a4647d452d38420d61ddf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-18 18:03:35.000000000","message":"Patch Set 6:\n\nMark, still blocking? I see no cleaner/simpler solution.","accounts_in_message":[],"_revision_number":6},{"id":"4a6f2ca1aa96831136bb3270ffeaad746a0f8fba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-20 17:55:37.000000000","message":"Patch Set 6: -Code-Review Workflow+1\n\nok, let\u0027s go.","accounts_in_message":[],"_revision_number":6},{"id":"4087e977ec60118a40911a74b0756103542c7791","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 17:55:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cae29e08ae863a233efa300912d4978c25b9932c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:36:51.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/61af3034e4554ac886e87f1f7aedc2d9 : SUCCESS in 51m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c4c7226f851541bd9a77e00ecca79d25 : SUCCESS in 1h 35m 45s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1fabf47ac77f452b90e8ae9703495a96 : SUCCESS in 1h 17m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d987561b46864681a3418bf65efe3141 : SUCCESS in 45m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42967f7d09664958a53fbffbadd14178 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776979c161204cb0a95a399084f00e5a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14fc44e1250c4b4a9b7e24e5dc7ef7b5 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98e53eaa56e24aa6b0334a0306612a57 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8040da1f82d54c52b0ef77df0eb4959a : SUCCESS in 4m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab657c5754844318acb473c876622cfe : FAILURE in 24m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f9293abfdfc43f099cbeaeb8ca3971d : FAILURE in 12m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02a8417d51d1474c8955cfbf8fccd9dc : FAILURE in 15m 26s","accounts_in_message":[],"_revision_number":6},{"id":"7070252a9f69f18426bcee3d3b74c142cddadc4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 08:37:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0d20bca9318b985da6e4a9df773170aeeafc9333","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 09:11:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/99/694099/6/openstack-check/tripleo-build-containers-rhel-8-master/988cd28/ : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":6},{"id":"b9ff3223770bdd19a8042e486b8479c60d119eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:54:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed34ce69f69e4ebfa85c9cee332714cd : SUCCESS in 47m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b73a1b377f30441e9565472f3e913292 : SUCCESS in 58m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f858ca1f4ba642bb9fec26370825a6fd : SUCCESS in 1h 29m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/688708a3e0d048b19c44c5ccc4e54aba : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6691e8e13d134a58bd0152d6169243a9 : SUCCESS in 1h 28m 09s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bbf1b22877f54ba99af9bb31ce214649 : SUCCESS in 1h 00m 19s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5b10222256e94e8389209b672927c469 : SUCCESS in 53m 35s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/50719298dd4f4baaa88f181b912076cb : SUCCESS in 2h 11m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4afda147649a4a8982bddc3dc7c5c31f : SUCCESS in 2h 12m 09s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/99e2ae3850e5403eb610d3f2543e052e : SUCCESS in 1h 54m 52s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f8aa4299b51f49f4a0a1792eb9454a49 : SUCCESS in 1h 56m 23s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/82b10bdafe6f4b808832a8c800119ef7 : SUCCESS in 1h 28m 53s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/85a21673f28b485db5727059972e7f43 : SUCCESS in 29m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0846cd70f7e849f09c2786dd4601c7a9 : SUCCESS in 18m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b83bb23c3e0c484585565d3c69422c71 : FAILURE in 18m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b6924534ffe4e55989b61928611a24b : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140ecf5c68cb40afafc851cfbd7a90b3 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9025f1206cbe46c79492001b61ceda8b : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/203e0dd9fd6940148e4674e7bcb620da : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b45facf59b949ae9a885a3706a1f4e7 : SUCCESS in 7m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a313cde5e014105ab1babfeca8d05a1 : SUCCESS in 26m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/62670fbb726048e59703f6a3c5f251e9 : SUCCESS in 27m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ed18cceb0e84ad0933cc65a163d2142 : SUCCESS in 48m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/961df3155b764bf19a659a37bddb6561 : SUCCESS in 1h 02m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/45f6248b9a5b4d08b7c950b762b4a3cd : SUCCESS in 1h 26m 50s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/dc83d2041ece457bae929197f3ba4289 : SUCCESS in 1h 35m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ad3cab42f2af4c2c9abefd1a220bad9c : SUCCESS in 1h 20m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b1bfcdbc62184b3898cc3b159ed02a76 : SUCCESS in 1h 33m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e5e1714c9bfe484f9c8bba891096ab21 : SUCCESS in 1h 12m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/37dadbfe729f4287bba1f07497f90c4c : SUCCESS in 1h 06m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3f609028446c44a6e60ab369e5f18b22808786ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:54:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"563f307bb8e681c14fec433b1b88953f68d969ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 12:29:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5dffd32ee50e4993996abda6881ff561 : SUCCESS in 1h 05m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/77420779fbd646b9820835bd8f16e0db : SUCCESS in 1h 10m 22s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7373825737f542a9962d1dc52c201af7 : SUCCESS in 1h 30m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fa718f74d5db43c8846dc2511215d149 : SUCCESS in 45m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a91eb0927064edc95d8274a29be79de : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9974d51e05814ac88c7906c57b0509c0 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67d62762b346451bb77c61fca0d86139 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9de8944165c448996585ccbce0c5b66 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce24bede58642ae8cf24d4a97164fe4 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6832683c40d444d4bc97e51d417aefdd : SUCCESS in 29m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd5fc6d30140429397503dfaa42192e3 : SUCCESS in 1h 05m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b53c29dcd6d45a48bf9a41225e8459f : SUCCESS in 1h 17m 25s","accounts_in_message":[],"_revision_number":6},{"id":"572b059058337901f3ef2fb7be15c6e5229c0015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 12:29:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"e68049c4d68eb7fa9e8edbd7448718c6ca4b9213","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 12:31:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85bdb2bc519e4b7081a0aeadd4a350a9 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6},{"id":"fa93ef675025ced76797ef8b8eb8ea485596cae4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 12:39:14.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a68269d0555ea838d468194be56efae6abfb8895","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"19ba215f8c0de9a1a0aa8cc40d96fc841ee80df1","revisions":{"caefa9df4b89b9ef74ac46a1ea7c51fc3451c661":{"kind":"REWORK","_number":1,"created":"2019-11-13 15:39:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/694099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/694099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/694099/1"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:39:20.000000000","tz":60},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caefa9df4b89b9ef74ac46a1ea7c51fc3451c661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caefa9df4b89b9ef74ac46a1ea7c51fc3451c661"}]},"branch":"refs/heads/master"},"f285aa68144a3f51fe30210d849cad4ca41b552e":{"kind":"REWORK","_number":2,"created":"2019-11-13 16:13:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/694099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/694099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/694099/2"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 16:12:59.000000000","tz":60},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f285aa68144a3f51fe30210d849cad4ca41b552e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f285aa68144a3f51fe30210d849cad4ca41b552e"}]},"branch":"refs/heads/master"},"d06ad09111d87d2e1d6585f4d463f7923f3f5e03":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-11-13 16:14:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/694099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/694099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/694099/3"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-13 16:14:43.000000000","tz":0},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\nCloses-bug: #1852452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d06ad09111d87d2e1d6585f4d463f7923f3f5e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d06ad09111d87d2e1d6585f4d463f7923f3f5e03"}]},"branch":"refs/heads/master"},"3ce28b36edca6b217fe96facbc6c7bf5e8a440ea":{"kind":"REWORK","_number":4,"created":"2019-11-14 07:36:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/694099/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/694099/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/694099/4"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 07:36:40.000000000","tz":60},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ce28b36edca6b217fe96facbc6c7bf5e8a440ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ce28b36edca6b217fe96facbc6c7bf5e8a440ea"}]},"branch":"refs/heads/master"},"45961a9d50a97b24e0c2cae52db2f696f316bce6":{"kind":"REWORK","_number":5,"created":"2019-11-14 07:39:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/694099/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/694099/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/694099/5"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 07:39:26.000000000","tz":60},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45961a9d50a97b24e0c2cae52db2f696f316bce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45961a9d50a97b24e0c2cae52db2f696f316bce6"}]},"branch":"refs/heads/master"},"19ba215f8c0de9a1a0aa8cc40d96fc841ee80df1":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-11-14 08:27:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/694099/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/694099/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/694099/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/694099/6"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-14 08:27:26.000000000","tz":0},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\nCloses-bug: #1852452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19ba215f8c0de9a1a0aa8cc40d96fc841ee80df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19ba215f8c0de9a1a0aa8cc40d96fc841ee80df1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
