)]}'
{"id":"openstack%2Fkolla~694104","triplet_id":"openstack%2Fkolla~master~I8209770ca1353fd89aacef0859ca3a373645c6b2","project":"openstack/kolla","branch":"master","topic":"faster-ubuntu-repos","hashtags":[],"change_id":"I8209770ca1353fd89aacef0859ca3a373645c6b2","subject":"Ubuntu: use local APT mirrors if possible","status":"MERGED","created":"2019-11-13 15:54:59.000000000","updated":"2020-01-14 15:39:06.000000000","submitted":"2020-01-14 12:00:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694104-1579003245681-bec474e8","meta_rev_id":"71863273fad1eb2d05e59a7257c205f0c92463e3","_number":694104,"virtual_id_number":694104,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 12:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-14 07:41:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-14 08:17:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-01-14 07:36:51.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-01-14 08:17:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 09:28:18.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-01-14 07:36:51.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-14 07:41:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-14 08:17:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-14 12:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d5bc7c88b4644a73c8d433640533542d952a06d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 15:54:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06a9e1df097053cacfd10dba5e7831add728ff5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-13 15:56:38.000000000","message":"Patch Set 1: Workflow-1\n\nmay require adapting gate setup scripts","accounts_in_message":[],"_revision_number":1},{"id":"6ace1ad600414c5bbc44a38029dfbad2fc8b038d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 16:26:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/04/694104/1/openstack-check/tripleo-build-containers-rhel-8-master/50b8b1a/ : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"882ed86bd30ab991ab9189f927cd11afcef4e137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 18:40:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/66ea780653a54c9c9d6d5a82524c947d : SUCCESS in 46m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1864e52c669645619f8a7f29d8b19c8b : SUCCESS in 59m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/975175eed9984464a82ccb845fef3e3a : SUCCESS in 1h 33m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f6d0e2cfb61d4b549470450ce91a29d2 : SUCCESS in 1h 36m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/04d1d05180be49ddbec95b6ab7bfe4df : SUCCESS in 1h 33m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cfb22eb16c25436587d40ad05c28837c : SUCCESS in 55m 15s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/56daa3f5c03e4ad08a3812e4c534b135 : SUCCESS in 47m 44s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c036c472c773446f8434aa667d833a0e : SUCCESS in 1h 54m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6b44eedf0b9241afb8d55affc625ed5f : FAILURE in 2h 29m 16s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/20deb1f4abb04528aa95c2f85b9fbcdd : TIMED_OUT in 2h 07m 33s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9c4be77be3454bb0adb0a7866db5d804 : SUCCESS in 1h 47m 56s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f818c379afb64bc0ac5f69e0645623e5 : SUCCESS in 1h 50m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d9f13f09a6f245199c0915568ba9b82d : SUCCESS in 34m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/26e4887e34df45a997b4587b9f3963c7 : SUCCESS in 19m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6d773d75985e42f38686eec935f7f0bd : FAILURE in 29m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3112afcda2ea43689a8b66ad773aae7c : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e17f92c71449138735a3f836a621f9 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59f6b35893ac4ab5a22dc3649bf04954 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96508b3144b84ddea597fd9881f74c27 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f1e68d2d3ca44d1aef324f2b60b1d47 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba1890490174164adb112da2612d236 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3c6ba60b3e649d397a20dccef01929f : SUCCESS in 28m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/639e178749224a46b7423c43a937a64b : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0cfdb495626d4d119ad96b6f721a9cb8 : SUCCESS in 58m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5e08a1db3e746cbba09961418c5a7f5 : SUCCESS in 58m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d03c9646d65b4ae9bc6b8ab6e9ef844a : SUCCESS in 1h 02m 01s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d8cac991ccb2485982d3a29a156b40d4 : TIMED_OUT in 2h 03m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/a24cdc0db06041639f75a68d26dd7db4 : SUCCESS in 1h 36m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ce81ee19cc11418c854e26c8b150a54a : SUCCESS in 1h 42m 54s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/8817e7163fa447f0a5b92c96aab672c9 : SUCCESS in 1h 16m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9ea77732e4504c6ea1e702a4188b635b : SUCCESS in 1h 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1cf6f4f85b123b6449fcc6afba13df2bc35502f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 11:22:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49c7cb69625166db422c9fd82320ef2663f3bfb9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 11:31:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"281d8fe65c9cc7e92001803ad9c8b3eef5db6187","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 11:35:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c0622f6d53a86aff1fa70374d910da9232089879","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 11:35:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09743922b93ced593a064324f39849f847543ad4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-10 12:05:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/6d84e8cf57d64652bd75caa5268fc563 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":3},{"id":"04b14f6e75c56898201432b8707c9f6931182ca5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 13:08:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1bb46252acc588cd95203e6ca571d8bed398e86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-10 13:35:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a68a6dbdf0a4496393c06871e69725df : SUCCESS in 26m 16s","accounts_in_message":[],"_revision_number":4},{"id":"dcb93013e40dcb0d20a9cac01df6aaf8a8dfb342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 15:45:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0e307c9c1aa94becab582646c0a63d96 : SUCCESS in 37m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cadd18e0745b4f70beeda9c96266873b : SUCCESS in 42m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e3341d93c6b9448db3b4380555107be0 : SUCCESS in 6m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6ba658100356491aa3f9f86d40e665e0 : SUCCESS in 5m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e84e6c9347f9431d8e6f35a690fb6e13 : SUCCESS in 1h 35m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/17dcf54f4d7d493aa974586cbb796f1f : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/abb52d009fa146c490e11afa37b2854d : SUCCESS in 1h 31m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bdf2722780c54286b405b63b05958c06 : SUCCESS in 51m 40s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/8aea9bcd71fe46039b6f349345e792c1 : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5744a1cd06e24e45850fb541e89981f3 : SUCCESS in 2h 01m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/642c452f17ca47189d52ce1c4981cca0 : TIMED_OUT in 2h 34m 09s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/9345b594113645158007f7e09710e32f : SUCCESS in 1h 50m 14s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/868b66eea1e44c6ab333db8bf6341984 : SUCCESS in 2h 00m 47s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/b2c07f156570416eb2646b95dd2ca81f : SUCCESS in 1h 39m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/084504b72a45463590f85992e0e92596 : SUCCESS in 34m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/00387ed837d44cb89821b15a5d87feb1 : SUCCESS in 31m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aba961ac01af488fb4f5eed2da302a1e : SUCCESS in 55m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b8350e31cc4573867b5e2e0c44e36c : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f8d7c9eeea4398abf99317af992a33 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ad8e06cb4ec4509ad205c8966a9bbbc : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae4ceb90585743b7a975f893b05fd685 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc97ce137b2f4a9cb0f161fa7d68091a : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f32362add8324e4482ffc7818fb2311a : SUCCESS in 26m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c210e6c5eeb44305b4ed8fb5b4d18ac4 : SUCCESS in 33m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c4c609250f4422391faf6023ed1836c : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d27cc7c72ce440fa527d1d2e3fa2ff5 : SUCCESS in 53m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a89af48de8c740f69df70df5f3626ec5 : SUCCESS in 1h 03m 31s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d871417f3b0042e18b8876f382295b33 : SUCCESS in 1h 15m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e40e9c211ebf4aaf99d6f11ee6bf3f72 : SUCCESS in 1h 25m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9443add93145466180733e3cdf97d8d9 : FAILURE in 1h 43m 03s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/1b016c6b11f747caa686764ed895fdae : SUCCESS in 1h 12m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5588141f28da45bcb2fb7c617a3c7588 : SUCCESS in 54m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0caf8a1a2118487c33c6b17332110d5906f3d3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 17:17:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb5a7f9617f24f2eb15ea7d4dae0926c : TIMED_OUT in 3h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed13efb65df756fd12f7d82673d2378f78cb4938","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-11 10:28:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cb1be71c1f27754ed674341f78b4bf37ec038eb8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-11 10:36:33.000000000","message":"Patch Set 4:\n\nI wonder if we don\u0027t need a releasenote for our users","accounts_in_message":[],"_revision_number":4},{"id":"bb9e550140e9afe6a0cd1a7f004c57ac2e7c5813","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-11 19:19:53.000000000","message":"Patch Set 4:\n\n\u003e I wonder if we don\u0027t need a releasenote for our users\n\nIf they do not use own sources.list then it does not matter as it will still work for them. And usually better than before.\n\nIf they use own sources.list then it does not apply to them.","accounts_in_message":[],"_revision_number":4},{"id":"8450ed505f8a5377c11695e0df3fac3108229fdf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-13 08:41:44.000000000","message":"Patch Set 4:\n\n\u003e \u003e I wonder if we don\u0027t need a releasenote for our users\n \u003e \n \u003e If they do not use own sources.list then it does not matter as it\n \u003e will still work for them. And usually better than before.\n \u003e \n \u003e If they use own sources.list then it does not apply to them.\n\nnot necessarily - if they use firewalls/proxies or whatever, the will be affected by this change, not that I think they are doing it right atm","accounts_in_message":[],"_revision_number":4},{"id":"1a391c93066efedcb10c1561a17f80cfa6e1cc58","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-13 08:45:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"53621a072994b9de1122bed7e8e974abe0bda1b8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-13 08:46:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"36212d7189aeb5c2b3281e84a5f21fa3274532e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-13 08:46:18.000000000","message":"Patch Set 5:\n\nthanks, Marcin","accounts_in_message":[],"_revision_number":5},{"id":"68e443225e5e8bf10de1c7b77a678b4f1388b7f4","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-13 09:08:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"26d6dff8b6b3b38b089632956fad67b54046dfaa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-13 09:09:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/294ab736fd634e51aab38aba32378b3e : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":5},{"id":"798e5481326cccebbc732dfaf718a196f665d5d1","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2020-01-13 09:28:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"33f260bb49a9ee12606f4165732fa592b786899e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 11:12:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/84d91435a09f42769ddb17fbb6cb5d77 : SUCCESS in 28m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b17108821c814caf8165d835f01d7c41 : SUCCESS in 42m 09s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2cb95685f7634829ba392200ad0a1385 : SUCCESS in 5m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b33b4e3ce2e4335903aa57fd31ecde4 : SUCCESS in 5m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5dbffd9efeb74aac8c780b4a5ac730d6 : SUCCESS in 1h 28m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/79131c9fa47445e7a3121340f7e415ff : SUCCESS in 1h 16m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f9ed1cec13724951be9c3d863a0bd741 : SUCCESS in 1h 29m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/529d00be9ae74a61baae481fffc276b4 : SUCCESS in 51m 15s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0f606174db0e4e3bb069be1d07eda1e4 : SUCCESS in 57m 56s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/cb8cab0063304dcf9589b2ebeb9d83ab : SUCCESS in 2h 11m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/d2f71ae20a2e48e39b6656d9ee698ad7 : SUCCESS in 2h 22m 55s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/01c618592d9a464a807b379c55ad628a : SUCCESS in 2h 02m 44s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4f8e709448164404bb8af43892f42fbe : TIMED_OUT in 2h 03m 04s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/cb1edcb23f734b84a77406ff78338b6a : SUCCESS in 1h 24m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c36bfc7aff8442819e16274efe27a5e7 : SUCCESS in 24m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/651cb7d7dd8d4a8cb42ea0d1f703c9e2 : SUCCESS in 16m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4bdd9f9a8df242c3b1fdc1e5bb8ac506 : SUCCESS in 55m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f41096b4e9fe44bcaa1258a39d198b57 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649406f2fbfc4eda92a2236148b2a7a8 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a3e55abc22445dfb4ebf7ef146619d2 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d41c6b7b2644056b2e776f85f6ef515 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f53364015445e58aebca84f11c9bd8 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de9477d3748d40ff8ed69e56527808d7 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8c0d1780dbc46669b88abec7325e37f : SUCCESS in 21m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bea1e693ae044ca2ad20edb2bfe9387b : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a39d2d28f93f46898dc3fef7d10781fa : SUCCESS in 51m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34daba83d56f424caa98f1863cd46b59 : SUCCESS in 57m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c132f992d7e942d9a766ecb3592355a3 : SUCCESS in 58m 26s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6a6cb134f65a45139f1040bdfd8d7fda : SUCCESS in 1h 17m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/1bb72bbac508476dafa62e6d7638c14a : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/331078ac32154b338fac9fada24efda7 : SUCCESS in 1h 38m 09s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d8e18c78710c46ef84bfbe0bbdedb589 : SUCCESS in 1h 09m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a49cab1ad61049df9f7ef0c9116ef07f : SUCCESS in 59m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"30439dc2f0dadc1d62c2a526dfb026599ad13e70","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-14 07:15:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e9b3303ab12605d04ef5a65e5d644f0a6b571ab","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-14 07:36:51.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9c814e8f092bbdd47441fd0ea98b42880997ef62","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-14 07:41:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/377568780aaa483fbaffb3029e3bca19 : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":6},{"id":"f6ea85a20cd1ad33a0cffd5b27a01bab42e9304f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 08:17:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c37160f423268419dda51197ffc75cb7409ce99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 10:09:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0b2bc674d1c74433a8eadb0c8b1c2f14 : SUCCESS in 36m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7f7dc3ac5bab4458b59394d4c140d163 : SUCCESS in 42m 08s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/271a932c6dcb4f31ac6e6d7537e183ec : SUCCESS in 5m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3fc0833750b346b6aa743e7d2a850c53 : SUCCESS in 5m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bb5e82e2c1c34e35902ac46027b54498 : SUCCESS in 1h 30m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/be6cbedf870440be899a8cdea2e9c2dd : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5ed86efb59a042dd93248fa658014da5 : SUCCESS in 1h 24m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d7982a82176f4d259d7359d5a7cbe85d : SUCCESS in 54m 43s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/583e3078f52d4809a2158a75e7671bc9 : SUCCESS in 57m 04s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/f6ce0143699d403ea306159af75c7a72 : TIMED_OUT in 2h 34m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1b1641c123164ef6b3c03d50e85c2275 : SUCCESS in 2h 23m 10s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/85adc95ac0a44800b24126f44e0c629e : TIMED_OUT in 2h 03m 47s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/931a11564b0348ea83d440397439ce69 : SUCCESS in 1h 54m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/7eea440ecf9c44cea2e36e640b333232 : SUCCESS in 1h 32m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d6f64b0804af4378902444c7dab4b27b : SUCCESS in 24m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4abb51d3996d4640b32b7c5d9bc37fc2 : SUCCESS in 18m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dcaaf61f5445419b9e7d2da99cb0b74b : SUCCESS in 54m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1e1b1636a9c46bdabf33146bb50ebee : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2f6152c3e04f05898a08cf146c2da1 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/095eccf56892456eab16ad85a8d2eb62 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74690bce652244289d06cb6d5ed15500 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144cf20fee4f4f0abdae0e42395e45ec : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c012ddf6fd0f4a2596dc51ee409a359a : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca15e0a0d67c440282e19521c2623816 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cec778094fe4502980d8cbf382e0012 : SUCCESS in 24m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/729133527caf4778b567466df0b72b87 : SUCCESS in 56m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbefcfceb8f04816a5e2b906c92a6e1b : SUCCESS in 54m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f605c4207a14434494435e81f0abbfcc : SUCCESS in 1h 07m 50s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/fb5b82ba39e5483ca103d421c333f938 : SUCCESS in 1h 11m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/9efd1739e64742d68fdd70df058c900c : SUCCESS in 2h 04m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1f900803d8644655aeb1fbc4903aa35d : SUCCESS in 2h 00m 32s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/98a1d55eab9948c595d38055d76f82c9 : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d4e29d735f794abd867e76de7412cfc8 : SUCCESS in 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a991f9a9f2468acabcbc9cf61cff83088d984251","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 10:10:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b40b54e8385df1a14c975f51a702267f6167527f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:00:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2f52c14572df4799b2517d87bd2adecc : SUCCESS in 42m 26s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/55d0344536914392ac53d61b3d087fad : SUCCESS in 5m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d6d63f2b7b5546e78e743e876c057ba4 : SUCCESS in 1h 27m 53s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d11113f990064c09924fa229d85884d5 : SUCCESS in 1h 46m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fb3777f2181343949e868df19d902357 : SUCCESS in 48m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfbebdc6fefe404fb01b544e42f0c2b9 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205b151113134e18998e699c8f490dd4 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7feca7e66a5f44508d257cfbb1ee9ead : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9be542e7ca749439d319066f3f55c85 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd4e7a069d34d6983b5536865f77926 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc7fd9e22cce4ab98ec94fa3ad451e23 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a70172f07034b12b61fd735ccc78bad : SUCCESS in 25m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ea4b253fb264c25b83fc618ccf989e9 : SUCCESS in 1h 33m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f55a5ef62d142a78e0a3eded996bbad : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":6},{"id":"21f887e0feae244bda9f608c478ec8d30c4f3542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:00:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a77d64696c66d21d54ff273d01d7e446fb3d40cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:02:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba44c75bae744c4a1c22e4eceaa881f : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b6132b3368946dabbc97bf8603ff4e7 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":6},{"id":"ff25b283bef69d9c127c07c5d2c808866904ac7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 15:39:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bae6c574e253445080a72fb875e1b901 : TIMED_OUT in 3h 04m 05s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"02c7efd42d072107d086f3d0885e4a19073cec5a","revisions":{"694a7f226c805150338aa96d8fc9cd129fcf9a70":{"kind":"REWORK","_number":1,"created":"2019-11-13 15:54:59.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/694104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/694104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/694104/1"}}},"commit":{"parents":[{"commit":"a398310b34d510da1192e59373faffc493ec4189","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a398310b34d510da1192e59373faffc493ec4189"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"subject":"Ubuntu: use local APT mirrors if possible","message":"Ubuntu: use local APT mirrors if possible\n\nChange-Id: I8209770ca1353fd89aacef0859ca3a373645c6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/694a7f226c805150338aa96d8fc9cd129fcf9a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/694a7f226c805150338aa96d8fc9cd129fcf9a70"}]},"branch":"refs/heads/master"},"e2c1e120bc3010d7981a368866208591f9418e11":{"kind":"REWORK","_number":2,"created":"2020-01-10 11:22:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/694104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/694104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/694104/2"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 11:22:03.000000000","tz":60},"subject":"Ubuntu: use local APT mirrors if possible","message":"Ubuntu: use local APT mirrors if possible\n\nAPT has a way to fetch mirror list and autoselect best mirror.\n\nChange-Id: I8209770ca1353fd89aacef0859ca3a373645c6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c1e120bc3010d7981a368866208591f9418e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c1e120bc3010d7981a368866208591f9418e11"}]},"branch":"refs/heads/master"},"a7654cd7c832948338ec1901825d9d44d1b2d1a9":{"kind":"REWORK","_number":3,"created":"2020-01-10 11:35:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/694104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/694104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/694104/3"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 11:35:29.000000000","tz":60},"subject":"Ubuntu: use local APT mirrors if possible","message":"Ubuntu: use local APT mirrors if possible\n\nAPT has a way to fetch mirror list and autoselect best mirror.\n\nChange-Id: I8209770ca1353fd89aacef0859ca3a373645c6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7654cd7c832948338ec1901825d9d44d1b2d1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7654cd7c832948338ec1901825d9d44d1b2d1a9"}]},"branch":"refs/heads/master"},"20b2d3457a2054da26f78ba09f95ab8d7820b84c":{"kind":"REWORK","_number":4,"created":"2020-01-10 13:08:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/694104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/694104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/694104/4"}}},"commit":{"parents":[{"commit":"51e50430c6299b82fbeb76d632ef2bda6e8409aa","subject":"Merge \"Remove tgtd and scsi-target-utils support in CentOS/RHEL 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51e50430c6299b82fbeb76d632ef2bda6e8409aa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 13:08:22.000000000","tz":60},"subject":"Ubuntu: use local APT mirrors if possible","message":"Ubuntu: use local APT mirrors if possible\n\nAPT has a way to fetch mirror list and autoselect best mirror.\n\nChange-Id: I8209770ca1353fd89aacef0859ca3a373645c6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20b2d3457a2054da26f78ba09f95ab8d7820b84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20b2d3457a2054da26f78ba09f95ab8d7820b84c"}]},"branch":"refs/heads/master"},"087c3e9d3f03c04a8c505b1fe81ae783a709cabc":{"kind":"REWORK","_number":5,"created":"2020-01-13 08:45:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/694104/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/694104/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/694104/5"}}},"commit":{"parents":[{"commit":"86d616bbe738effd430fff53298a7bbde1fbaca3","subject":"Merge \"CI: Fix symlinks for failed build logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86d616bbe738effd430fff53298a7bbde1fbaca3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-13 08:44:53.000000000","tz":60},"subject":"Ubuntu: use local APT mirrors if possible","message":"Ubuntu: use local APT mirrors if possible\n\nAPT has a way to fetch mirror list and autoselect best mirror.\n\nChange-Id: I8209770ca1353fd89aacef0859ca3a373645c6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/087c3e9d3f03c04a8c505b1fe81ae783a709cabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/087c3e9d3f03c04a8c505b1fe81ae783a709cabc"}]},"branch":"refs/heads/master"},"02c7efd42d072107d086f3d0885e4a19073cec5a":{"kind":"REWORK","_number":6,"created":"2020-01-14 07:15:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/694104/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/694104/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/694104/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/694104/6"}}},"commit":{"parents":[{"commit":"e2f013357b850a433a300d4363e764ea9c3b9e5d","subject":"Merge \"Fixes rally db commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2f013357b850a433a300d4363e764ea9c3b9e5d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:54:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-14 07:15:04.000000000","tz":60},"subject":"Ubuntu: use local APT mirrors if possible","message":"Ubuntu: use local APT mirrors if possible\n\nAPT has a way to fetch mirror list and autoselect best mirror.\n\nChange-Id: I8209770ca1353fd89aacef0859ca3a373645c6b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c7efd42d072107d086f3d0885e4a19073cec5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c7efd42d072107d086f3d0885e4a19073cec5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
